)]}'
{"id":"openstack%2Floci~753781","triplet_id":"openstack%2Floci~master~I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b","project":"openstack/loci","branch":"master","topic":"refactor-setup-gate","hashtags":[],"change_id":"I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b","subject":"aarch64: enable building multi-arch images","status":"MERGED","created":"2020-09-23 16:01:29.000000000","updated":"2020-10-14 22:02:08.000000000","submitted":"2020-10-14 22:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753781-1602712928069-4f1731f4","meta_rev_id":"e693f392d16240c3395de107659696c2916dcb07","_number":753781,"virtual_id_number":753781,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-14 22:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-14 21:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-10-09 15:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-14 21:12:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 15:53:53.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2020-10-14 21:12:43.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-10-14 22:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e57f8e8f2a41b0e00144fea4f3a345461126994a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-23 16:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e9f0f2c8cdacacc2df22343de89f82a1984d979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:53:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ccd99655b2314aedad43bfb20c6f053c : FAILURE in 33m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/627d23e11dfd44848ac53cbd727ec41e : FAILURE in 33m 33s\n- loci-heat https://zuul.opendev.org/t/openstack/build/18bb14cf935f4a4b93dc5e9b44a50fb5 : FAILURE in 32m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9a125c9476794cc3a12ac8b20b765731 : FAILURE in 32m 49s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/328d8bfeeae14ac98e9f7bf9ad6c2116 : FAILURE in 30m 07s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fea291c4762443dca59f7018be99fe8b : FAILURE in 42m 33s\n- loci-manila https://zuul.opendev.org/t/openstack/build/eca2503a4f2a4b9494ef8239d6d9b9fe : FAILURE in 29m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7d432a4e24244043beb0f14a50767a7d : FAILURE in 30m 09s\n- loci-nova https://zuul.opendev.org/t/openstack/build/934f7ca817da4b92b0020e01ff6c74b8 : FAILURE in 31m 42s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7e05aae3da4f42ee92b4891603fba3dd : FAILURE in 29m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e6cc073cfc274639b737b8024e8e6769 : FAILURE in 31m 17s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/02dc5e110e784ec49df30e096ebd7b7e : SUCCESS in 23m 29s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/844615e08a06414e80cc4a0ff0a83a55 : FAILURE in 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"669fb00e0857ecf13c04b012fbf5f23618a2ecca","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-24 11:04:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"62f34cb2178ac6426873d01f34f129317f818434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:02:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/484931b901d14a72abb1af99da8a6df7 : SUCCESS in 36m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f679e5f26ddb4ef984c21a9f1935ea42 : SUCCESS in 33m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b38f78b96f1e4fefa46574c72300b8b3 : SUCCESS in 33m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/bccdaf7e03784ac4bb3b2e9da73d1c6d : FAILURE in 5m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/bf68bac2aff94e1788e09eaa119e5800 : SUCCESS in 33m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1201adf763954855aed8457b36f20573 : SUCCESS in 31m 22s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9653bc378ad944a49bd40e95cf2333b3 : SUCCESS in 33m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ea9eaacb9a20471fae585acc0186013f : SUCCESS in 39m 05s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ec2241f73ae44647a1939771287a45b3 : SUCCESS in 39m 37s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a4eb26e248cc4f0e86639af4cb6018e6 : SUCCESS in 22m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bc3f30fcd2874942a9267d76997df244 : SUCCESS in 29m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/34286dac693c4352b4fc2618d1c48413 : SUCCESS in 24m 44s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/10a1ebc315044c1e82384d7ef9ca4a05 : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":2},{"id":"cdd05860e5cecc6833f8c8c030b1bae8dab242ef","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-26 18:41:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eacc34795ed90203985fb9702fddced9d715f389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 19:37:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7dbb22dd92dc4937a8e6128b123105d8 : SUCCESS in 44m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f4214db0afe84445ae30688aab450d32 : SUCCESS in 37m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/aa68b3a1dc6d46548c8c79bc0a1830ad : SUCCESS in 54m 00s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/655d58ed559446c598d62c0b732f9843 : SUCCESS in 43m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6457f47917084835adfbf4ae7c043b2c : SUCCESS in 36m 19s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ee0b16781a50417a8aa8dbffc4e94139 : SUCCESS in 42m 47s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b407e3a7c8ae4e31be253beeda157532 : SUCCESS in 37m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9e96ed700cf24f74af07b71c5a749a6f : SUCCESS in 45m 03s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9aeeec90ee8444cbab6883ec4b1c919e : SUCCESS in 39m 49s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ebebee6f6e48404395a10b574be43815 : SUCCESS in 43m 43s\n- loci-placement https://zuul.opendev.org/t/openstack/build/098a43a0314348c588e5e126fd0330bd : SUCCESS in 32m 30s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0fd4f63809bd4868be366ad13aadffab : SUCCESS in 27m 58s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/999134164bb243648cb221432fb72a88 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8e0c6441b8284b4fed8a14bafc0bd4f05173677f","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-02 03:16:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nThat is cool. Do you think it makes sense to cover it with gates?","accounts_in_message":[],"_revision_number":2},{"id":"9220ee65faa3c6f3def11cf40aa8375e8c1f1fbc","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-10-06 19:44:23.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e0a7842cc3ab4067518aa6eff9517dad1a1b646c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-10-06 19:45:03.000000000","message":"Patch Set 3:\n\nUnfortunately, it is _quite_ slow to actually run this in the gate.  If we use buildx on amd64 systems, it takes around ~2h30m to build the requirements.  I\u0027m not sure how available the arm64 upstream nodes are so that risks blocking us from merging things...","accounts_in_message":[],"_revision_number":3},{"id":"948801289695a37803df735bdd6e1a4d8be24fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 19:59:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/bdef7bf1c06e49b6a05a13d38950f436 : FAILURE in 3m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c8f51d10413947bdb132af3b28099377 : FAILURE in 4m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7bfced31db18456a81340d197b4fc062 : FAILURE in 4m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/1212d3edcbd64f3bb7fd618b0e6fbddd : FAILURE in 5m 36s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2692824b4d274a70a65e881c8c438863 : FAILURE in 6m 28s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cc7f6e5ddc2147b48ea87c1d93678452 : FAILURE in 5m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/19b9096ca104448eb8f6f6ec6a2ab0df : FAILURE in 4m 55s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/38331b420b5d4bd2bc8035e492a8805c : FAILURE in 3m 51s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bc7ba01a5d6545debae91fecddf85ebf : FAILURE in 4m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1ab81b5c13c2414389301de2f8ad7f0f : FAILURE in 3m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9f5902573b3b4ca780e773ad9588d4ff : FAILURE in 3m 39s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e247f7ce2f5f4d558199b4b890be8805 : FAILURE in 5m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/31aedf0a41ae479ca30cb73ea796d8d8 : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9450ca26642d6f54d7ebc86a051991cc3f9dff80","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-08 16:54:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a5a84a19067d8ceaaba63f99de87c782e99d59b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 17:33:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/43a04c9a21c147e1843e577c78963178 : FAILURE in 26m 37s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a1af103d378b4eed97cd2ebba4c7e1f1 : FAILURE in 26m 53s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8dea53b79af24ebab8e6b4f2a042d494 : FAILURE in 25m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/147ef4ed864a4859a7fb467160e3a2b0 : FAILURE in 30m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d3e9f20c8c574f979b259d4f75fd5a6e : FAILURE in 25m 59s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/66329b963642433d98a5e964394ea647 : FAILURE in 25m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c9e341e6344246248ab5551c1a05b105 : FAILURE in 19m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f01c0a65edaf4c4c87c9ba3a61e367bb : FAILURE in 25m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9f21fe6427764855a99b1d2fd04443dd : FAILURE in 27m 04s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b412853e278e442f842a24e7f49db6d0 : FAILURE in 25m 02s\n- loci-placement https://zuul.opendev.org/t/openstack/build/650b3e784421419bab03613283737fab : FAILURE in 23m 37s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/10b2618339ab49099671302ae1aad458 : FAILURE in 25m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/109c682a2350487c8e5cd0c186db4e6b : FAILURE in 26m 39s","accounts_in_message":[],"_revision_number":4},{"id":"7bc6e477ac74f7dc9cc93ecad1e477f96d1ae6a0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-08 17:49:12.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f2e06b59ff6d54f1e405ab271f8945ff9c0af5f5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-08 18:13:46.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7d74ddaebd3016d9a832ac4f61158ce30c117a1c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-08 18:31:50.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cc2c0a5880fd65f90ebf39f397996cb9f17f285c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:25:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5bc037181a2646f1b8ab3a34578ca2c9 : SUCCESS in 41m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/10f83b51c26544d083d4a09f927cabcf : SUCCESS in 34m 10s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a9d1d5c513714ad29b8a554a026f8832 : SUCCESS in 37m 06s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e25a1da196834a9dbf7fa589a73d9ae9 : SUCCESS in 38m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2287d24853d1456da74e683e1826aca2 : SUCCESS in 34m 59s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/57f537a1c92245a991c597e5839fdfc7 : SUCCESS in 32m 29s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4fdcaa073ebc46d3800ead03f0a3e200 : SUCCESS in 40m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/98ffc42bbb6248859a8aac77603d6608 : SUCCESS in 36m 44s\n- loci-nova https://zuul.opendev.org/t/openstack/build/14d3302c245d420491751824e99349bb : SUCCESS in 36m 24s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/03778c0127ce4465a63b3319ee739277 : SUCCESS in 18m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bbfb076d84c74b80b11236cb5d3a94d0 : SUCCESS in 16m 55s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/12b645517f504bb1824ff14d7dd59ac2 : SUCCESS in 22m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/9ac18d8b6a284d67955ef724d6b270f4 : SUCCESS in 31m 04s","accounts_in_message":[],"_revision_number":7},{"id":"eee7d3350fdb88e7cf33e1c81124ac728963d0d5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-09 15:53:53.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8008bb6f862e098537638abdcab2f78412698cd9","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-10-14 21:12:43.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"598ea1e2ada791dfe9975473da329053014b2419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 21:12:54.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"21555fc04c63df07d6990c720305ae2ff4412c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 22:02:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/69ff656f62074dbea26780c62e72d796 : SUCCESS in 39m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a6c60e0bb9894f479a54061783c26fc5 : SUCCESS in 38m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/532854e2e6b5475199d140249a146bd6 : SUCCESS in 34m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9fa7b575b22447d4ba6a04abb137a981 : SUCCESS in 35m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/bcf390f3ba9c4266b56fca385be9b4bd : SUCCESS in 33m 08s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d634e01854f5420ba99dce9e63f328b2 : SUCCESS in 33m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/521caa2548a94235ab71344b8d89f10b : SUCCESS in 32m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/912145b35fb5474eb8a90d7bd4beaa75 : SUCCESS in 38m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3aec4cd476264b908e2192eb398b0139 : SUCCESS in 36m 42s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/48a385bf75e5422ba4f15f29f48dc497 : SUCCESS in 36m 38s\n- loci-placement https://zuul.opendev.org/t/openstack/build/df61418094e24a75b9d9fc0793da412c : SUCCESS in 27m 59s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/20bacc2a28484d4398aa8195cce4949f : SUCCESS in 29m 06s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/494b52b240094f67834a6151ea06c22c : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":7},{"id":"b4f7d7f16bc637ab703048263e6a5bc1ddbbbc8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 22:02:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"115396b0aae975d2cd0778ff47fd8c63b0b2e608","revisions":{"10a8aed984687de75e7c12dacdf8e24d6f4af01a":{"kind":"REWORK","_number":1,"created":"2020-09-23 16:01:29.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/81/753781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/1"}}},"commit":{"parents":[{"commit":"27ac34f7fc541cb7b23295a4798eda7af45f05f6","subject":"aarch64: stop trying to build confluent-kafka","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/27ac34f7fc541cb7b23295a4798eda7af45f05f6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/10a8aed984687de75e7c12dacdf8e24d6f4af01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/10a8aed984687de75e7c12dacdf8e24d6f4af01a"}]},"branch":"refs/heads/master"},"09a743b8645a3d206235ce720a9ac6b450b37ea0":{"kind":"REWORK","_number":2,"created":"2020-09-24 11:04:11.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/81/753781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/2"}}},"commit":{"parents":[{"commit":"27ac34f7fc541cb7b23295a4798eda7af45f05f6","subject":"aarch64: stop trying to build confluent-kafka","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/27ac34f7fc541cb7b23295a4798eda7af45f05f6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-24 11:04:09.000000000","tz":0},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/09a743b8645a3d206235ce720a9ac6b450b37ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/09a743b8645a3d206235ce720a9ac6b450b37ea0"}]},"branch":"refs/heads/master"},"25f11f1977d61a952f64f428dc115421a8c98b5c":{"kind":"REWORK","_number":3,"created":"2020-10-06 19:44:23.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/81/753781/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/3"}}},"commit":{"parents":[{"commit":"27ac34f7fc541cb7b23295a4798eda7af45f05f6","subject":"aarch64: stop trying to build confluent-kafka","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/27ac34f7fc541cb7b23295a4798eda7af45f05f6"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-10-06 19:44:21.000000000","tz":0},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/25f11f1977d61a952f64f428dc115421a8c98b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/25f11f1977d61a952f64f428dc115421a8c98b5c"}]},"branch":"refs/heads/master"},"f16e165db37415c1716b682b69b0a2f036aa8fb4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-10-08 16:54:12.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/81/753781/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/4"}}},"commit":{"parents":[{"commit":"29a781636e0a94df6b6ffff5c41d12f93db9070f","subject":"Update leap15 base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/29a781636e0a94df6b6ffff5c41d12f93db9070f"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-10-08 16:54:12.000000000","tz":0},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f16e165db37415c1716b682b69b0a2f036aa8fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f16e165db37415c1716b682b69b0a2f036aa8fb4"}]},"branch":"refs/heads/master"},"103ac3bd08769dcfc7926d7618a80a0193f88263":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-08 17:49:12.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/81/753781/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/5"}}},"commit":{"parents":[{"commit":"e653352bf8a9f75bd1a222a0be94b51793e58293","subject":"Update leap15 base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e653352bf8a9f75bd1a222a0be94b51793e58293"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-10-08 17:49:12.000000000","tz":0},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/103ac3bd08769dcfc7926d7618a80a0193f88263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/103ac3bd08769dcfc7926d7618a80a0193f88263"}]},"branch":"refs/heads/master"},"1c2c93c1ef48caed5082189e8d954f5565d3a7aa":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-08 18:13:46.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/81/753781/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/6"}}},"commit":{"parents":[{"commit":"5efde76d8dbdc185db594089e954bf319a523bb0","subject":"Update leap15 base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5efde76d8dbdc185db594089e954bf319a523bb0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-10-08 18:13:46.000000000","tz":0},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1c2c93c1ef48caed5082189e8d954f5565d3a7aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1c2c93c1ef48caed5082189e8d954f5565d3a7aa"}]},"branch":"refs/heads/master"},"115396b0aae975d2cd0778ff47fd8c63b0b2e608":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-10-08 18:31:50.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/81/753781/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/81/753781/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/81/753781/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/81/753781/7"}}},"commit":{"parents":[{"commit":"ce66853ab9f4d8236957707beb2ee80eee31ffc5","subject":"Update leap15 base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce66853ab9f4d8236957707beb2ee80eee31ffc5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-23 15:58:50.000000000","tz":-240},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-10-08 18:31:50.000000000","tz":0},"subject":"aarch64: enable building multi-arch images","message":"aarch64: enable building multi-arch images\n\nAt the moment, the only missing piece of being able to build\nmultiarchitecture images is the fact that loci does not support pulling\nan image that has a \u0027fat manifest\u0027 (i.e. pointing at many different\nimages of different architectures but with the same name, built using\nbuildx).\n\nThis patch adds support to fetch-wheels in order for it to be able to\nunderstand how these manifests work and download the correct one for the\narchitecture that is being built.\n\nThis has been tested locally here in our environments and successfully\nproduces multiarchitecture images using docker buildx.\n\nChange-Id: I5ea132d12fec50b4c2cddd8b8c7c1f251ee2c30b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
