)]}'
{"id":"openstack%2Floci~754228","triplet_id":"openstack%2Floci~master~I53cb32f1af566228582ed6bd69a9067b9a879ef9","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I53cb32f1af566228582ed6bd69a9067b9a879ef9","subject":"Switch to ensure-docker role for docker installation","status":"ABANDONED","created":"2020-09-24 22:54:42.000000000","updated":"2020-09-25 02:57:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9d30d9229552c0b603257ba10902ee9bd666e4f","_number":754228,"virtual_id_number":754228,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-25 00:10:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 00:10:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ad3474ec315f899ee91f289aa12e4833adba596","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 22:54:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5f65d249f48238ce9fbea2b5fac789739669285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 23:05:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f7b9f1977c8d442da652efa8b0a26ec6 : FAILURE in 3m 32s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3f8554bd954f40bebf338d4289ac8b9d : FAILURE in 3m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/67b2005a6cbe48fa890948da479dfeac : FAILURE in 5m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4da1ec04b43c49bd91463c2298dac86f : FAILURE in 3m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c6a1f69ffb55484fa05527eb70504f60 : FAILURE in 3m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7961e01ca45a4511a40a6cce87913920 : FAILURE in 4m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c4c4ac34ddf8446fab97dc4f7d0f022a : FAILURE in 4m 35s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/fb27303e9947418f8b74d944d19742dd : FAILURE in 3m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7247244400764fbca7a6ddeee4382c9a : FAILURE in 3m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/23b2a71338c74202b73b8b0390f7d707 : FAILURE in 3m 57s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2376de70e91c42b6a7e0466e796de633 : FAILURE in 3m 20s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a79a35740a23461bb8aa5e3cc47b8567 : FAILURE in 3m 21s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bcd2d6628b2b4367b087471174ab3797 : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c1a7fbf6e11da032cd62139266b01eafa88dfa98","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:12:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed44de674c0b1239f8ddccbb1abb4eda13ca24c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 23:22:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c80105084d0c459c876b9f63af2bc938 : FAILURE in 3m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/704d691dc7ff4ee8b9db2f3c12573f7f : FAILURE in 4m 37s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fefb557fadb049709dd0aac7433a94b8 : FAILURE in 3m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/06befe3230e24afca1d26fd898c05271 : FAILURE in 3m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/263122d48b5045f287b5981f77a1503f : FAILURE in 3m 28s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f169e602274d4bbaa8e0dbd43feacc4e : FAILURE in 3m 18s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8f9d68f24f4040fab4ba24fb4b0681da : FAILURE in 5m 07s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9110bf35d474476d8555b2699ceee92e : FAILURE in 4m 36s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c1c4a388685c48378148a3f68ea9de61 : FAILURE in 4m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ab1b0d2a1c6f41e097160c75c5af519e : FAILURE in 3m 52s\n- loci-placement https://zuul.opendev.org/t/openstack/build/efbbbe9a2b684fd08529b23d5aadd8bc : FAILURE in 3m 48s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ea1669cc0f3e4c91950ad84f79260f5c : FAILURE in 4m 25s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/cd85f90cdc7e4329a7c25de6c05ecd8c : FAILURE in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3400a54b1de43b8d51777bd8a84a4602e50f7d8d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:23:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"f225336c103e3634b291a3ba63ab399b9b950baa","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:25:14.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"de3ea6b0cdc01dab778264b0cf51496561844d7c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:26:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09ae555ba58fb7ba9ce2e5d1c0faa413c92c8491","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:29:08.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ccce99a5293815f66bc51d1085e7d0779d966a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:10:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/799c25be6acb47aa9df4ffd5a0824993 : FAILURE in 34m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2adc4c6146ec4f41ac257cb1b6e0494a : FAILURE in 27m 17s\n- loci-heat https://zuul.opendev.org/t/openstack/build/efa080c55d98497184d0a49cebec28db : FAILURE in 34m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/db0a5679217741d2b83c1b228d2c368d : FAILURE in 27m 14s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6f75806d868c4e78a1a08bb014bafe16 : FAILURE in 29m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/af8511f6dea04c4f8c02872c71628830 : FAILURE in 27m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8cfba9f0e68d45b38da32316a95c9566 : FAILURE in 24m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c9bc58179d0341b7bc9879df4e4082be : FAILURE in 27m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2c8a2f5e46a34067b5d246bd6ccecc76 : FAILURE in 25m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a073641684634f27916b9590f7e59df3 : FAILURE in 26m 21s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f8ce22a051f648e188369804f3c6c1d1 : FAILURE in 25m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ae9a3d6080684fc784d16be654cfe7eb : SUCCESS in 26m 53s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a4e5c7429d314b8db4523281e74d95d9 : FAILURE in 23m 51s","accounts_in_message":[],"_revision_number":4},{"id":"d9d30d9229552c0b603257ba10902ee9bd666e4f","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-25 02:57:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e947cffad00505eb4e305820628d81cf8cc9cebb","revisions":{"5f19fb0dfe944c73a1000b464ae93660cd6f030e":{"kind":"REWORK","_number":1,"created":"2020-09-24 22:54:42.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/28/754228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/754228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/754228/1"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 22:54:23.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 22:54:23.000000000","tz":-300},"subject":"Switch to ensure-docker role for docker installation","message":"Switch to ensure-docker role for docker installation\n\nChange-Id: I53cb32f1af566228582ed6bd69a9067b9a879ef9\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f19fb0dfe944c73a1000b464ae93660cd6f030e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f19fb0dfe944c73a1000b464ae93660cd6f030e"}]},"branch":"refs/heads/master"},"3b71857ea70f5323720e43e6e4b9381aa5320b1b":{"kind":"REWORK","_number":2,"created":"2020-09-24 23:12:49.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/28/754228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/754228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/754228/2"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 22:54:23.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 23:12:46.000000000","tz":-300},"subject":"Switch to ensure-docker role for docker installation","message":"Switch to ensure-docker role for docker installation\n\nChange-Id: I53cb32f1af566228582ed6bd69a9067b9a879ef9\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b71857ea70f5323720e43e6e4b9381aa5320b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b71857ea70f5323720e43e6e4b9381aa5320b1b"}]},"branch":"refs/heads/master"},"53e52ccc2f327d0f197e30cdb28f22ca8754bddb":{"kind":"REWORK","_number":3,"created":"2020-09-24 23:26:58.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/28/754228/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/754228/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/754228/3"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 22:54:23.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 23:26:54.000000000","tz":-300},"subject":"Switch to ensure-docker role for docker installation","message":"Switch to ensure-docker role for docker installation\n\nChange-Id: I53cb32f1af566228582ed6bd69a9067b9a879ef9\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/53e52ccc2f327d0f197e30cdb28f22ca8754bddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/53e52ccc2f327d0f197e30cdb28f22ca8754bddb"}]},"branch":"refs/heads/master"},"e947cffad00505eb4e305820628d81cf8cc9cebb":{"kind":"REWORK","_number":4,"created":"2020-09-24 23:29:08.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/28/754228/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/754228/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/754228/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/754228/4"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-09-24 22:54:23.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-09-24 23:29:03.000000000","tz":0},"subject":"Switch to ensure-docker role for docker installation","message":"Switch to ensure-docker role for docker installation\n\nChange-Id: I53cb32f1af566228582ed6bd69a9067b9a879ef9\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e947cffad00505eb4e305820628d81cf8cc9cebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e947cffad00505eb4e305820628d81cf8cc9cebb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
