)]}'
{"id":"openstack%2Floci~766510","triplet_id":"openstack%2Floci~master~I8b25d7ddf5432de65cfd814193e10afee516fd1a","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I8b25d7ddf5432de65cfd814193e10afee516fd1a","subject":"Fix gates","status":"MERGED","created":"2020-12-10 17:12:38.000000000","updated":"2020-12-10 20:06:20.000000000","submitted":"2020-12-10 20:06:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766510","meta_rev_id":"54e28d893341add5481e473b01479597ff26815a","_number":766510,"virtual_id_number":766510,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 20:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2020-12-10 19:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":1,"date":"2020-12-10 19:02:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":2,"date":"2020-12-10 18:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2020-12-10 19:02:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"}],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 18:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 18:55:52.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"REVIEWER"},{"updated":"2020-12-10 18:55:52.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-12-10 18:55:52.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-12-10 18:55:52.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"9d0536d3102fc448a79da0748b3f434dbc36704f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-10 17:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2c286c427d1070ca5e5a22afdc78a07ed2dbd56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 18:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/eff61bf781d347ce945073317ce66cc8 : FAILURE in 26m 54s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8e8babe6fe7c4be6ac44095390442de0 : FAILURE in 34m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7c3ddffd0dbc4114b2ea972c71c8329b : FAILURE in 32m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d05c6a46ddec4e5dbf89c887593e8f02 : FAILURE in 30m 45s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/459dc0d4f8294e68a212f603f55a8aff : FAILURE in 29m 08s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8afb5b67ed424ce7bd1b4d713af69445 : FAILURE in 29m 08s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7b3788ec53284b458af9c794d45accdc : FAILURE in 32m 40s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/bc37933f688a46eca96444c950d3c04d : FAILURE in 27m 19s\n- loci-nova https://zuul.opendev.org/t/openstack/build/498bc7aece5d43bdb734207980912712 : FAILURE in 30m 40s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4e3c78854e6d4291b1e9daaf907cf09f : FAILURE in 31m 52s\n- loci-placement https://zuul.opendev.org/t/openstack/build/818216dbe0d6454196b706868204efae : FAILURE in 26m 28s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c0795b7762cc4e8482eb432706b9d514 : FAILURE in 26m 05s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/06973e7f245e42d89901744d05c702f1 : FAILURE in 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b523477f5209efdd00554eeb5fa1678030811574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-10 18:08:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb21a56882a5e2d1f0e6f1c503a821ef3ba9187b","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-12-10 18:56:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8456ebc3f487b1a3755070d559892ace4b7575d2","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-12-10 19:02:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9a39a713c1493238eeda50dfe5027f917b37a13d","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-10 19:02:49.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"429d278549ee1b78ba2bb1b3f029f2a3d49ad80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 19:07:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e0911bbb5f1c4782b1bbd36302af5e38 : SUCCESS in 38m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/96c767e335864e1598befef46e365bec : SUCCESS in 38m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ef5289104dc8455a93cd2750513fa589 : SUCCESS in 39m 14s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a618d21202a140208c8d5da9ec417d77 : SUCCESS in 40m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/77720bde7e9645118aa67901b2e71418 : SUCCESS in 38m 45s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/84ba01073ca645998c0e852e4fd0678d : SUCCESS in 43m 44s\n- loci-manila https://zuul.opendev.org/t/openstack/build/776f720d36f642438586c2b3b2e01abe : SUCCESS in 38m 39s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0f1b937378ff4252a9d2c469f39a6020 : SUCCESS in 57m 37s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5a94f0b7fd644e7884bbd5554edb51d3 : SUCCESS in 43m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/775dd2969ef8477c8091aabb1ac35075 : SUCCESS in 52m 51s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5e18399ca70d4220ad8a66084e4fa6d1 : SUCCESS in 32m 48s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/563e2fb7b65e405dbdce67f80952b500 : SUCCESS in 26m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/203dec700b174a699a23f3669871c197 : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":2},{"id":"379526e1c88f41f9da615756e6e68e5c1705c2cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 19:07:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81917d4e52285ae6232961118a150ecf973d0e8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:06:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/25f67f9b439e4ffeacdba9d6e85d8f3d : SUCCESS in 33m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f0d0aa47417b43d7a921f04b40575ede : SUCCESS in 33m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f624812ee08846f79d1a7ec00047d17f : SUCCESS in 33m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5923153e368347fea3efcf483484c06f : SUCCESS in 36m 55s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/db6540760e3a4ee2898493a96391a107 : SUCCESS in 35m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8d829f243a2d443189d38db11939bcb3 : SUCCESS in 33m 18s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6ecf903747a046cc866389c3db4f3e4e : SUCCESS in 48m 02s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/84493734fe9c4e92bb5620f61f90608c : SUCCESS in 42m 17s\n- loci-nova https://zuul.opendev.org/t/openstack/build/89d0344e27434524b129afa7a29cdfd5 : SUCCESS in 53m 35s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7440cabc23c54bb29c47ad36b0f132fd : SUCCESS in 39m 06s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3d093971693e41bb87f483ebbc4595b9 : SUCCESS in 32m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7fd54675b79d40c5befe60df89d6a85b : SUCCESS in 35m 47s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c9a937e67dd944f98acec4fa56366e84 : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":2},{"id":"54e28d893341add5481e473b01479597ff26815a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:06:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","revisions":{"81875954f6f906753b7054de3c3d3b259cfe0754":{"kind":"REWORK","_number":1,"created":"2020-12-10 17:12:38.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/10/766510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/766510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/766510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/766510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/766510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/766510/1"}}},"commit":{"parents":[{"commit":"115396b0aae975d2cd0778ff47fd8c63b0b2e608","subject":"aarch64: enable building multi-arch images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 17:11:01.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 17:11:01.000000000","tz":-360},"subject":"Fix gates","message":"Fix gates\n\nopendev luminous ceph mirror is no longer available for xenial ocata\nbuild, thus using original location.\nSetting locale to resolve salt installation failure for master.\n\nChange-Id: I8b25d7ddf5432de65cfd814193e10afee516fd1a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/81875954f6f906753b7054de3c3d3b259cfe0754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/81875954f6f906753b7054de3c3d3b259cfe0754"}]},"branch":"refs/heads/master"},"5e3fa8dd2c2cecec46401a640e24fe1f86795597":{"kind":"REWORK","_number":2,"created":"2020-12-10 18:08:02.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/10/766510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/766510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/766510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/766510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/766510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/766510/2"}}},"commit":{"parents":[{"commit":"115396b0aae975d2cd0778ff47fd8c63b0b2e608","subject":"aarch64: enable building multi-arch images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 17:11:01.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 18:07:58.000000000","tz":-360},"subject":"Fix gates","message":"Fix gates\n\nopendev luminous ceph mirror is no longer available for xenial ocata\nbuild, thus using original location.\nSetting locale to resolve salt installation failure for master.\n\nChange-Id: I8b25d7ddf5432de65cfd814193e10afee516fd1a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
