)]}'
{"id":"openstack%2Floci~766635","triplet_id":"openstack%2Floci~master~Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","status":"MERGED","created":"2020-12-10 22:48:59.000000000","updated":"2020-12-24 04:49:43.000000000","submitted":"2020-12-24 04:49:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766635","meta_rev_id":"77f96ba675e077442f8bceebf1822b393237f083","_number":766635,"virtual_id_number":766635,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-24 04:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2020-12-23 23:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":2,"date":"2020-12-18 20:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2020-12-23 23:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 23:45:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 21:49:05.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2020-12-16 21:49:05.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-12-16 21:49:05.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-12-16 21:49:05.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-12-16 21:49:05.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-12-16 21:49:05.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-12-18 20:54:15.000000000","updated_by":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"reviewer":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"state":"REVIEWER"}],"messages":[{"id":"823fed5becd6d101b1591b104cd68beeaeec5cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-10 22:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6846d4195d9e321c43a65f2c80fc826345ad00ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-10 22:51:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53f4f930b2bef3c9e4d77ca866001c0c2511f678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 23:45:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1e17bf2dc52c4b4b86924d2699b31a33 : SUCCESS in 40m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0e8c95b3edc842e082a05686f201b024 : SUCCESS in 43m 54s\n- loci-heat https://zuul.opendev.org/t/openstack/build/07945f45224c4266b10790c62201225d : SUCCESS in 40m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d31fb86688d44bf484030fb77f33d063 : SUCCESS in 44m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5d100fe55ea54ba885b7b8976a41eb6b : SUCCESS in 40m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/072081b0de4f42728dc6765aebc0edfd : SUCCESS in 39m 07s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2f32112a97c848db9a96913cd91e8e0e : SUCCESS in 43m 15s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/04a9460e66f743b688ff8796cf649bfe : SUCCESS in 47m 39s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e345dc66f40f4d04aa6fd8cebec0be09 : SUCCESS in 36m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/97126b1bcc8d47f59d47c719a40fd7f6 : SUCCESS in 40m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a1793fde2bdb4df293ab9b7569efe938 : SUCCESS in 28m 05s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/68d50864af0c4cbc8caf705236083599 : SUCCESS in 23m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b83cbee9e824404ab19d7f6094c22123 : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3ae23c4aed6ad9e16e287fb49b2fd0d20d588b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-10 23:54:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bb4e6982599d03e5cec2fa8e5b0cbabb2d8c8dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-11 00:24:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b3e20574474ae62162accccb0c24932889503a99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 00:33:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8fa8d2c971154260afa1c918e02faffb : FAILURE in 5m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9c383986e98d47e3ba881983389413be : FAILURE in 4m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4ab3a5ecaee245c18ce751f6adb80f4b : FAILURE in 5m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2d87fff3d7e04532986a75bbab34e126 : FAILURE in 4m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/85faf5d1588b4554b0c3d88d28135e5a : FAILURE in 4m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e3bdeee13c2d4200bc1b94a2556a051a : FAILURE in 4m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a39e1936814741f98d0178a92be88c50 : FAILURE in 4m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/93d684c365d044719378f0ec760db2e5 : FAILURE in 4m 43s\n- loci-nova https://zuul.opendev.org/t/openstack/build/4465bfae46794a439a35e29752335210 : FAILURE in 4m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/92c9878436be47049e25a5163fbd8928 : FAILURE in 4m 34s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7b643367a6a648ae88cca6af7adcf480 : FAILURE in 4m 32s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e574af593caf44dd825bf2e5a897f20f : FAILURE in 4m 39s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1d729fe824094dae9a1698d25ef04099 : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":4},{"id":"5f698de831050bdf349cb98124c4354aabdf9599","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-11 00:36:32.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ebfb54284c623def707c1b6e797b9fcdfb41ba77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-11 01:14:19.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f4a2239b233f0768ab829fde460ea475b2e2aedd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-11 01:56:49.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"1cadcd392c89c1ed7e69ac8e21ff8f5c5ba31a1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-11 02:18:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3ba084e6933912e2483810f3548356a5eb063109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 03:07:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/bd479a9c03f54eee92c981b65c88f425 : SUCCESS in 39m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/67e319cf8cf548289a493d55f3ed577f : SUCCESS in 41m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fe824bdc8dbd4648bcf6c056a80e335c : SUCCESS in 42m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e78b08ec3ef44f13b9c459c5818bed68 : SUCCESS in 42m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ce7bbb0e6a2a445a845823f8063fe3e9 : SUCCESS in 41m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4b983bd27b354537b4476b53a0cfc176 : SUCCESS in 38m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6b225e3bdf3d44eca794689aa9e41d85 : SUCCESS in 39m 23s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6a7438a8bc9f4144a61b62c5996142f8 : SUCCESS in 42m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f03490a1659a41668e2489a99b84240d : SUCCESS in 42m 46s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/61ec47e87b19487ab3fb13e849522675 : SUCCESS in 37m 31s\n- loci-placement https://zuul.opendev.org/t/openstack/build/52925ee2a7a34e4399ec2d76f7de55b5 : SUCCESS in 32m 03s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/709697891a3e4a21b2c66139a090a760 : SUCCESS in 25m 45s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f43881f5d935423cbc39cb4ff25e008d : SUCCESS in 39m 23s","accounts_in_message":[],"_revision_number":8},{"id":"4b5e7212fa6da33b74e13b46cc748c81adb7f5a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-15 21:09:33.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6d13bbd27c955fa1e1596e22267fbfdebd8c098a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 22:11:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e09f05ca0ac8476b80ce3bbd105e539b : SUCCESS in 36m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9dc8728bc7834751a9e21195f77e10ea : SUCCESS in 40m 55s\n- loci-heat https://zuul.opendev.org/t/openstack/build/02e497a8062a45be8d33c2d93d890d23 : SUCCESS in 41m 34s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0adf48a3714e4e508489efd79184aee3 : SUCCESS in 52m 51s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1fb9e284002145d3a7dba289d2b5a532 : SUCCESS in 49m 56s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/456849c0f4e44e0a8e1b6855e80a1c2b : SUCCESS in 43m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/059eca5d409546d28e0783acffc2cfc7 : SUCCESS in 40m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/817e0effdcdd41039f937c3081107728 : SUCCESS in 42m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/64ba26be72474cdc87817b71637cc70e : SUCCESS in 44m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c6edd63727b8440395899ee11489e574 : SUCCESS in 43m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f34597335a904b3ca4f85bfa2ff19368 : SUCCESS in 41m 41s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b5b1ed322dd547d0899764d1906d9513 : SUCCESS in 26m 54s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/e92690cf85ea472d920959ef51cb984e : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":9},{"id":"5c55c9f28945893d81c5b3b0fbb56c1405d01b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-16 03:32:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7c205ed74a4f5f5dbec4a0bd1c64d93d8770c553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 03:51:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/541d45d16289487f96f105b0b1c49c21 : FAILURE in 5m 38s\n- loci-glance https://zuul.opendev.org/t/openstack/build/57a08b86654a4e89baddfc776661822d : FAILURE in 8m 39s\n- loci-heat https://zuul.opendev.org/t/openstack/build/eca722397fb94a7399195372a4204e5f : FAILURE in 8m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/555297681b1b4275895a505c9bb01e58 : FAILURE in 7m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/57de67b7e66a4154827037a17f768103 : FAILURE in 7m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0dd5b6a39d9c4dcba573b7e9725b4944 : FAILURE in 4m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/59bbd529f6ef47128a38d4b080376f52 : FAILURE in 5m 07s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b22bb152d2dc4b3aad93c6cec9ed83d6 : FAILURE in 6m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/24608960067f4af3b047cad66ed6b8f2 : FAILURE in 6m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/66028c494b0a4445a39ed86ddcae138a : FAILURE in 8m 21s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4f2b27b203eb4a0db70679e1f9b61d1a : FAILURE in 4m 41s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f484043ef1f246c39c136f42a93065b4 : FAILURE in 6m 16s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/ccb9576951fc404e90775ee7ee6bd60f : FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":10},{"id":"16f2213dad6d94963f884c632d8c9be9f42be1a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-16 03:59:49.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"ce7827425975c23c8f8eb47b263b144c00684ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 04:50:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5ee61a44b3f2408e83b99e15d7fbd4da : SUCCESS in 44m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d1b2ced19dd14e989dffcc01f1d6ddec : SUCCESS in 37m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c134332167d24c09952d167e62e5eaea : SUCCESS in 27m 45s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/732ac89607b04b188df2524303bf65d6 : SUCCESS in 28m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/12c0f3f2611b4ab188f65e6f59d243b0 : SUCCESS in 27m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1af7d1945a8a4e0f962b755505edc313 : SUCCESS in 26m 54s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9c57594e2e174fee9aeb819e545de35c : SUCCESS in 27m 35s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/80842f3a09c64bf194b95abd1bfb3519 : SUCCESS in 27m 50s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ab365e62de72434795abc9c3bec56591 : SUCCESS in 24m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1c93e44c0af149dabf2e95d29c46a6c2 : SUCCESS in 33m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/01adc974c170495eb8588c56a8696771 : SUCCESS in 29m 23s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d40174111d124d489777ca81385472cd : SUCCESS in 27m 08s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2c2f7292c1d44a61aa833674938f1d9a : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":11},{"id":"03affe7cb5274fd21c8a6753ee52fda92af9a067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-16 16:15:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"20e47c29804c74566785bbac779a5bee3c69901c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 17:06:15.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3ca0cba8bc3649e39f9d90d00b929f4d : SUCCESS in 40m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c40ee5297d8b4d83a8f407831291004a : SUCCESS in 36m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fcb42acb769141a2b205a33362c40758 : SUCCESS in 37m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9de00d7d9e5e4cbf88a23c1ba4462047 : SUCCESS in 39m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b26dd2e7b7cd4048bb3e24511ccfbcca : SUCCESS in 37m 24s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2f63b93d4fce419bb742c329bd512efe : SUCCESS in 35m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2998c73ddda44c9eac724698bf7b76d0 : SUCCESS in 37m 16s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/454aa3794c994470817c41e9ccaece55 : SUCCESS in 38m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6e0ff80770284653a3a43b5d92b2b351 : SUCCESS in 42m 25s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e2f7ba03ed6b40a1b47fb497a7e690b9 : SUCCESS in 21m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5d2eb7545aed414ab0c5024d944b84bb : FAILURE in 18m 52s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1222e005f699470ba098ee6b32e2c69d : SUCCESS in 31m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3361646ad3bd4f6cbbfb8537daaf6900 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":12},{"id":"08491b911a8bfa78cf3425aec306c22e8eb606cf","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-16 17:06:58.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1b4cbc108d40e1f1ed6bf16012e0e82b860fd58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 17:55:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/da148197e31b42a489e5abd05533a141 : SUCCESS in 20m 37s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c9b821ce087f48bfa2a32b0c360db6da : SUCCESS in 20m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fb82b5e75ffb4c63baa86cd890d9f59e : SUCCESS in 20m 03s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/79e0e249d8bd40f6a9393b38689ab3db : SUCCESS in 24m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/debf43b957034dc8a27770c832207a70 : SUCCESS in 19m 51s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9765a6356f414a2eaf7472ca5d32e4b5 : SUCCESS in 20m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/07ba28752d9745a2ba0d6ceed98206b0 : SUCCESS in 20m 52s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/286de0a335354b518bc87c33c4ac1a1b : SUCCESS in 37m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/459f143d8974480f8d707eddc1292f54 : SUCCESS in 37m 17s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c1e8736772404e7894f10fd4b4cb43ae : SUCCESS in 38m 05s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7b23db3804dc482b8a8111e1f11a059c : SUCCESS in 31m 23s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8f71c30fe9c74f929eec4b6b5773345e : SUCCESS in 26m 29s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c9316a447dc941d2a8902e7c76a46309 : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":12},{"id":"610f927b590dd0c7a65897bf6532306985cb8b4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-16 21:44:00.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"1fd42bda0aadbcae20db2aa3d5daaa4e36f8a32b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-16 21:45:00.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"f8cee088f2d0a9305a2f6d9378482eab32fd7ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:45:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8b88403fd38245d9aba287e7d6f8f63d : SUCCESS in 40m 48s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3bb30c008c1d4a45bdaceb327b452c42 : SUCCESS in 30m 55s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2ec4cf3345db429ab85e596826410cc4 : SUCCESS in 54m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7cb7a391b3e84a699c3c8d915adc6a09 : SUCCESS in 41m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/59608c6ab901430e89c83ab477edc64a : SUCCESS in 51m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9fd6b22b18674bd2872e239c2cc39071 : SUCCESS in 50m 38s\n- loci-manila https://zuul.opendev.org/t/openstack/build/13b34175413548229ef91f4d92027bd9 : SUCCESS in 39m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/60a8a0196c5d44d7a5106547bae4d6d5 : SUCCESS in 38m 02s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ce7a35616ba941e4bf6da703b5a805cd : SUCCESS in 41m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9f66aa3aedfe42b0a959136cb9ddd47e : SUCCESS in 39m 05s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bf2c3fb2f3684e4aaa2d6a689febf918 : SUCCESS in 16m 58s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/32321a06015e4a138010f989780fbd7c : SUCCESS in 27m 34s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5436b85db5624946b63bf726f8622ce6 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":14},{"id":"6035306b6fa5347658238c16bbc45d8da1022309","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-17 18:19:00.000000000","message":"Patch Set 14:\n\nreview please. Thank you!","accounts_in_message":[],"_revision_number":14},{"id":"0ad48f43bde705a65c9af3acee88f57de5032079","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-12-17 18:54:56.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"5a39837ab782291f4d57c5142264e2315b7dc124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-18 01:53:45.000000000","message":"Patch Set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"b9c8a50b2a75b71e44e10882df38922ba4973a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:52:08.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/578f65d377c7432995d22e7f6479f229 : SUCCESS in 41m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2b8c05b98f384985981e6302c625b959 : SUCCESS in 40m 30s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8ce0fd610c7648b5a29b0e4ee0fe91c1 : SUCCESS in 45m 35s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f09de77d24784f9bb61af5b40f08fb38 : SUCCESS in 41m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2eb375e1eed048cea13751cb0859308e : SUCCESS in 39m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6bcbcbbc0e8c457ca86158ac1b843255 : SUCCESS in 39m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3d30cd0ece5d4b12b3508050f36bb55f : SUCCESS in 43m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/18b6ba7fc18a4d16a8da58dd15384f9c : SUCCESS in 43m 43s\n- loci-nova https://zuul.opendev.org/t/openstack/build/4a75328affe64dc18e733237be0eb86b : SUCCESS in 53m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/57c27786e9e741c7b5109adfe854f06d : SUCCESS in 43m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/faca7cbe0cb841ffa0fabb6047cb8726 : SUCCESS in 30m 04s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f1f8c642e7664c248a5f218164add4ac : SUCCESS in 25m 18s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/940c975f2c494044a6f8e86e96f8c982 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":15},{"id":"3765732d6273511c964d0a45d369ea9d7e6e5453","author":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"date":"2020-12-18 20:54:15.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"777fa7ea788929449dd45d13c87fc75dacafcfb8","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-12-23 23:36:46.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"87a06b4522498e08c48e1bcdd452f73b5b426af2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 23:36:58.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"99d39cd2c933402f44311d49e410bdf57d74b090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 00:40:39.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d2b070e390824a00958d8dd1923615a6 : SUCCESS in 49m 35s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cbeccb8ee89f48cc8d617d7ddf7678f5 : SUCCESS in 43m 13s\n- loci-heat https://zuul.opendev.org/t/openstack/build/07031bc1386e4c28b38ad3bd6bb39742 : SUCCESS in 50m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d1502f863b4344378b03436544dae5c3 : SUCCESS in 55m 51s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5b569b03dcec490489086f6a94d9873a : SUCCESS in 43m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a4c73c6d524942848cef6dae542a695b : SUCCESS in 53m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7e8c11aa6a9843f29989823543dc7f17 : SUCCESS in 41m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8308f69c5e274c7399becc3ab80ad9d1 : SUCCESS in 49m 03s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0e0e969828744124a08ccd5146161ff8 : SUCCESS in 46m 17s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3abee5b63a81496ba42f39b886bf1faa : SUCCESS in 47m 56s\n- loci-placement https://zuul.opendev.org/t/openstack/build/911bd9a1930c4cefb80ed81fd945e5e7 : SUCCESS in 40m 07s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6b2d4799a91a45b7b449046fa7a5290d : TIMED_OUT in 40m 51s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6b7aeb93fec84995996dd86d39d498c0 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":15},{"id":"da3519c1c4ab36472b60c8a02e00046d63d27f39","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-24 02:15:00.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"f0b42056c3dc69a0dcd79bea6048f7d3611a3668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 02:52:58.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0ad8efc4022045c2a185f4f29ca609a4 : SUCCESS in 35m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/75649e9b9a9c413c8974866c3c66976e : FAILURE in 28m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/451a0e4b8979472db71112dbbef76d1b : SUCCESS in 33m 35s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2ea98ebd318c42e1b997e568989e2c05 : SUCCESS in 35m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/02bfa920e0c24671a4aa52dd21b145ff : SUCCESS in 32m 43s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d9abad5cbb8a4719a25ff2235b255982 : SUCCESS in 32m 07s\n- loci-manila https://zuul.opendev.org/t/openstack/build/49cee13723024a958e00ee3fba76b631 : SUCCESS in 32m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b41635b02775457bb4ea81d3f6373353 : SUCCESS in 36m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/994a8a169c1a40b6896fd72bfbfb73c3 : SUCCESS in 36m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/693d8657e80e4d13b6fa9ed4907dc4cc : SUCCESS in 35m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4076a96c00c44eae9a45980e530c1554 : FAILURE in 24m 38s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/55a821425ef74182be3b8e324013cc23 : SUCCESS in 29m 10s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6fcdae6b85e347b3ae44f47e50fecd68 : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":15},{"id":"53c6ada598c43a74339147fe17774db6579f6a30","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-24 03:04:34.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"a10e01c82ad342aebdb0cad148c81b678241eb8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 03:57:57.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3d527eabadc24d3daeaafa5868139aa3 : SUCCESS in 47m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c09371f540d14d29819c618a87529ad1 : SUCCESS in 40m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3fbf67c9ca4640859270dacb5d09f507 : SUCCESS in 45m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3e4ab21a63b94b88a04d85b5c0aea9dc : SUCCESS in 42m 41s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d99c4886580f4c58b76809152ec66106 : SUCCESS in 40m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7138bfe647fc445faa39b4578352a2ef : SUCCESS in 40m 27s\n- loci-manila https://zuul.opendev.org/t/openstack/build/130ac907ae5b4fad900d0c070f851d23 : SUCCESS in 40m 20s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b65c82f1026649ce80399c548775c8b4 : SUCCESS in 43m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9dca009efd454bd6b8ec94d953f867ef : SUCCESS in 48m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8a3e4dd3b3814b6a9613764ee6c68d31 : SUCCESS in 47m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1ad70075cbd54db6b0e22dd5abfd38a7 : SUCCESS in 35m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d54cc6522992449c85b6502c822bceb1 : SUCCESS in 28m 33s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/75f6b236a3694dabb13aff11dbb67d98 : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":15},{"id":"04b43886223f6f0cd2ab398622ea74d06ba45d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 03:58:09.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"5bfdfdfb0d07dac1128c00753d7b232abec7292b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 04:49:42.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/40ebe4dc78ec4a749a34c70604255e5c : SUCCESS in 43m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e1ec052b207346a3844b5510bc9eecf9 : SUCCESS in 41m 49s\n- loci-heat https://zuul.opendev.org/t/openstack/build/954d963ba3c3409ab5f0881803fc2bdb : SUCCESS in 40m 41s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5dd3b70eff0749a0bcf6b322ffe4810f : SUCCESS in 43m 41s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/db4eda53c52640458328f6b81c779bf6 : SUCCESS in 41m 23s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/34895534428a4a69b8bbc122260d880a : SUCCESS in 39m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/17faece653174b66b0d31aa3d9b1c78f : SUCCESS in 39m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/59ff7840581e4956a2d2610da385a5c7 : SUCCESS in 45m 33s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0e2269d52cb84df2b2d823c08b2578c9 : SUCCESS in 46m 26s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/cd497b5d902844ed961985dfdb856016 : SUCCESS in 42m 45s\n- loci-placement https://zuul.opendev.org/t/openstack/build/96d87e42080040459cb9857a4d0bd967 : SUCCESS in 34m 49s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c1490f5b864843c9a26b992250dd82ba : SUCCESS in 28m 24s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2d7923ec52cf4b43899d53e6ea4d0207 : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":15},{"id":"77f96ba675e077442f8bceebf1822b393237f083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 04:49:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","revisions":{"6cb759d510e65a07c466fdd58805cc3f02522132":{"kind":"REWORK","_number":1,"created":"2020-12-10 22:48:59.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/1"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit package will first be tried to be installed from local\nwheels only and, if failed, index is used.\n\nThis adds a minor overhead to no KEEP_ALL_WHEELS builds.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6cb759d510e65a07c466fdd58805cc3f02522132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6cb759d510e65a07c466fdd58805cc3f02522132"}]},"branch":"refs/heads/master"},"714c697c0f2e6500641a5f0710ccd621f5c648c5":{"kind":"REWORK","_number":2,"created":"2020-12-10 22:51:58.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/2"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:51:56.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit package will first be tried to be installed from local\nwheels only and, if failed, index is used.\n\nThis adds a minor overhead to no KEEP_ALL_WHEELS builds.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/714c697c0f2e6500641a5f0710ccd621f5c648c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/714c697c0f2e6500641a5f0710ccd621f5c648c5"}]},"branch":"refs/heads/master"},"b734ad248d39b23807e7bffc420a3a13ee7e374a":{"kind":"REWORK","_number":3,"created":"2020-12-10 23:54:38.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/3"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 23:53:40.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b734ad248d39b23807e7bffc420a3a13ee7e374a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b734ad248d39b23807e7bffc420a3a13ee7e374a"}]},"branch":"refs/heads/master"},"618b4935a2ddb8af83298c4f7b8d18faea0419fe":{"kind":"REWORK","_number":4,"created":"2020-12-11 00:24:44.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/4"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-11 00:24:32.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/618b4935a2ddb8af83298c4f7b8d18faea0419fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/618b4935a2ddb8af83298c4f7b8d18faea0419fe"}]},"branch":"refs/heads/master"},"b87a17fe22238925afc89261c14e114ab565ba73":{"kind":"REWORK","_number":5,"created":"2020-12-11 00:36:32.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/5"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-11 00:36:27.000000000","tz":0},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b87a17fe22238925afc89261c14e114ab565ba73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b87a17fe22238925afc89261c14e114ab565ba73"}]},"branch":"refs/heads/master"},"19c1db54a1fde099ce250bcc0d0c82bc97738268":{"kind":"REWORK","_number":6,"created":"2020-12-11 01:14:19.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/6"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-11 00:37:04.000000000","tz":0},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/19c1db54a1fde099ce250bcc0d0c82bc97738268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/19c1db54a1fde099ce250bcc0d0c82bc97738268"}]},"branch":"refs/heads/master"},"54e7dae514e64174258b97eec2e9f1dfd5d4ba85":{"kind":"REWORK","_number":7,"created":"2020-12-11 01:56:49.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/7"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-11 01:56:44.000000000","tz":0},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/54e7dae514e64174258b97eec2e9f1dfd5d4ba85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/54e7dae514e64174258b97eec2e9f1dfd5d4ba85"}]},"branch":"refs/heads/master"},"bf37f34a3fb0c5a81542635ed8c77db501930404":{"kind":"REWORK","_number":8,"created":"2020-12-11 02:18:34.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/8"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-11 02:18:28.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf37f34a3fb0c5a81542635ed8c77db501930404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf37f34a3fb0c5a81542635ed8c77db501930404"}]},"branch":"refs/heads/master"},"c9542b4209e05f0eb39e8a8293b7f1960c7a8cf5":{"kind":"REWORK","_number":9,"created":"2020-12-15 21:09:33.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/9"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-15 20:37:40.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9542b4209e05f0eb39e8a8293b7f1960c7a8cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9542b4209e05f0eb39e8a8293b7f1960c7a8cf5"}]},"branch":"refs/heads/master"},"3367526b1422db79f54deeeb3b9f42618511e9e0":{"kind":"REWORK","_number":10,"created":"2020-12-16 03:32:37.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/10"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-16 03:30:27.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nThis also adds PIP_UC_IGNORE dockerfile argument, allowing to provide a\nlist of components to exclude from upper-constraints.txt. This enables\nan ability to pass another version of the same component to PIP_PACKAGES.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3367526b1422db79f54deeeb3b9f42618511e9e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3367526b1422db79f54deeeb3b9f42618511e9e0"}]},"branch":"refs/heads/master"},"53c429c4d92e6195f92457360f1d43648981aa0b":{"kind":"REWORK","_number":11,"created":"2020-12-16 03:59:49.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/11"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-16 03:59:45.000000000","tz":0},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nThis also adds PIP_UC_IGNORE dockerfile argument, allowing to provide a\nlist of components to exclude from upper-constraints.txt. This enables\nan ability to pass another version of the same component to PIP_PACKAGES.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/53c429c4d92e6195f92457360f1d43648981aa0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/53c429c4d92e6195f92457360f1d43648981aa0b"}]},"branch":"refs/heads/master"},"8815529e1b1f15331715835769b7583cba9a04f9":{"kind":"REWORK","_number":12,"created":"2020-12-16 16:15:24.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/12"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-16 16:15:19.000000000","tz":-360},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel of the same version from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build\npackage will first be tried to be installed from local wheels only and,\nif failed, index is used.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8815529e1b1f15331715835769b7583cba9a04f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8815529e1b1f15331715835769b7583cba9a04f9"}]},"branch":"refs/heads/master"},"6b17f0c5593a3dccb11551879d5435e13a4178bd":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2020-12-16 21:44:00.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/13"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-16 21:43:46.000000000","tz":0},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers todownload wheel of the same version\nfrom index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image build packages will be installed\nfrom local wheels only. This brings a requirement to ensure all wheels are present - in the case\nwhen PIP_PACKAGES is provided to component image build, it should be also passed to requirements\nimage build.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b17f0c5593a3dccb11551879d5435e13a4178bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b17f0c5593a3dccb11551879d5435e13a4178bd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"addcaea0a96c2a7e6d20b73c5fa0378b50cb5593":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2020-12-16 21:45:00.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/14"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-16 21:44:56.000000000","tz":0},"subject":"Try to install packages from local wheels first","message":"Try to install packages from local wheels first\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image\nbuild packages will be installed from local wheels only. This brings\na requirement to ensure all wheels are present - in the case when\nPIP_PACKAGES is provided to component image build, it should be also\npassed to requirements image build.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/addcaea0a96c2a7e6d20b73c5fa0378b50cb5593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/addcaea0a96c2a7e6d20b73c5fa0378b50cb5593"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c1a589698e5675db0c6c586d9a3a00d8cfb435b9":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2020-12-18 01:53:45.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/35/766635/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/35/766635/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/35/766635/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/35/766635/15"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-12-10 22:41:45.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-18 01:53:38.000000000","tz":0},"subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","message":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno\n\nWith requirements image built with KEEP_ALL_WHEELS pip prefers to\ndownload wheel from index.\n\nWith this commit and KEEP_ALL_WHEELS\u003dyes during a component image\nbuild packages will be installed from local wheels only. This brings\na requirement to ensure all wheels are present - in the case when\nPIP_PACKAGES is provided to component image build, it should be also\npassed to requirements image build.\n\nChange-Id: Iea72ee68aa59f2fea2c82a98f7f637c4476f7d2a\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
