)]}'
{"id":"openstack%2Floci~771265","triplet_id":"openstack%2Floci~master~I30760e6f7378c35d4afea837b238967509017da5","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I30760e6f7378c35d4afea837b238967509017da5","subject":"Add monasca-api profile","status":"MERGED","created":"2021-01-18 18:55:56.000000000","updated":"2021-02-16 15:13:21.000000000","submitted":"2021-02-16 15:13:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"771265","meta_rev_id":"a76e0b367986b4957f5950781403ad598a28ed44","_number":771265,"virtual_id_number":771265,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-16 15:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":2,"date":"2021-02-16 14:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-05 17:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":1,"date":"2021-02-16 14:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 19:25:22.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2021-01-18 20:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 16:56:42.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2021-02-16 14:16:44.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"}],"messages":[{"id":"f877ec94d78fd38a31e8f1cc646ca015d2fedab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-18 18:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a3f8b1c719e95c351fca33a26914adc213b8400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-18 19:17:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1e8ddc35b9a6de0610c5537da6e1bcf614c645fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 20:15:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9c264006b31d4754bb4f358a16a9f377 : SUCCESS in 45m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a68b30fb9984496e869a58a7fa0c8ea8 : SUCCESS in 22m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5000d9cfc1f64d258b6d10be1072136f : SUCCESS in 44m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4287daa32c4d487a911832f43656d762 : SUCCESS in 45m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/aac3889c976149dfa3969f97a249b75e : SUCCESS in 43m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d3592f3f240443aca026973a4e30e0c0 : SUCCESS in 40m 21s\n- loci-manila https://zuul.opendev.org/t/openstack/build/974b91f6e2db4ae3bd08a86e284aa649 : SUCCESS in 38m 29s\n- loci-monasca https://zuul.opendev.org/t/openstack/build/579be7c8196a4cc99a8e8f2c3928c88a : FAILURE in 34m 52s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/88591f08098b4f439bdf49e99c5865b8 : SUCCESS in 22m 51s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9d54b4bbfe1f43ca8032810e080a283c : SUCCESS in 51m 51s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0935f0fdd233481a904acc35d6fa5848 : SUCCESS in 46m 51s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7de292086883452281cfe6d78115bb5a : SUCCESS in 41m 38s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/fd60ce263a62410db40f051315127b5a : SUCCESS in 31m 10s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/21cc8d197a624bfab50dcfbb89b872d4 : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e60e0a5c9742efc0d7e9b771240c7504f5a3b93c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-18 20:39:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"339573264e2f1b65fe6d0a8506c14b3acc95042c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-18 21:00:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"76d3903b116e5d56b111a568c8fb0fe51479dcba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 21:12:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0c7feab4538142948b8c20cd66018961 : FAILURE in 3m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b8bda086a38e44f48c9b8f21f16f3663 : FAILURE in 6m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3d3c024b108e4672bc6a4a62526d2d03 : FAILURE in 6m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/310ba7665ae44999904f8919360a17d9 : FAILURE in 2m 40s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/692dfff780cd4fd4a9c9f4e69e9c16d4 : FAILURE in 3m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8dc6043cc5494f9e8e72dd179de84ac5 : FAILURE in 3m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0cde914d0ad14512b7d88b2c60227b5a : FAILURE in 3m 08s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/9cb069ca9e4f420386e9a5ba8b801108 : FAILURE in 3m 14s\n- loci-monasca-events-api https://zuul.opendev.org/t/openstack/build/dd577e3cf6d44ada8b2198d0ee83ade9 : FAILURE in 2m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b93a8651db1b468484c9b0723a2a9899 : FAILURE in 3m 12s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dbb3b0fa689a40409a19253f13385abb : FAILURE in 4m 04s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/bd74154ccad94e5e9750bb19de71aba6 : FAILURE in 3m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a07be744492f4a2abc7161f98ee645f6 : FAILURE in 2m 56s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/545649c537ff4abf963b8d54699dac2b : FAILURE in 2m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f6e01b9509364e658e516fd8bc7f358c : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":4},{"id":"76837a6fe88b78922c600cc8d1cb0fdbf53ec494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-18 21:31:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"834ef1e0759dd7e638cba7b184e64bd660685e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 21:51:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/07fee53824f243289ebc04e58979e6cb : FAILURE in 3m 19s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c2781ae46e3f41d3a78d567275b9c77a : FAILURE in 4m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/bfbebc09349a4056af8ec9b035685a84 : FAILURE in 2m 47s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/858aca01fdb64daebab2321d291e871d : FAILURE in 3m 55s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2c8f89029b1a4294b61cec92c5be54f8 : FAILURE in 3m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9b4d6ecea2f6429fbfc246b9a2e08b39 : FAILURE in 2m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/39e2d2f8022343a0a6956988c613b40a : FAILURE in 4m 16s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d342f40eb35347eea25b304bbdbe02aa : FAILURE in 4m 00s\n- loci-monasca-events-api https://zuul.opendev.org/t/openstack/build/8f0552e97d7f46e0a133907ae5302adc : FAILURE in 3m 02s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cb0fb968329743d29ea6f43e6caa15b5 : FAILURE in 2m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3e6c1b6f6e0b4494953848856692328c : FAILURE in 2m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c74ae11754334461b645ac18bcb349c7 : FAILURE in 2m 56s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5b88427d78814e139ec369fe9077ed42 : FAILURE in 2m 47s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/66cfd5e317444eab8b132fe2df15b618 : FAILURE in 3m 49s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a218071e5b0b44efb5c39056d80a83aa : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"fe0b99858907cbb61b07502c0230a690a7990769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-19 08:43:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b92efaf931d06db39612a0c680190a74176171a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 08:52:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/86256c6a84644401b40b9e08bfda2d8a : FAILURE in 4m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4c4d1bf4ee6b4674af26a55f0a056ec4 : FAILURE in 2m 34s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7a728015562645029f9ba6133e0251d0 : FAILURE in 2m 52s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4d358afb01524e23920ebc0a245a2714 : FAILURE in 2m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/bca11cbc4bdb4109a2de765cfe4dd309 : FAILURE in 3m 16s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9ed1cca10d3a4eb4938098e7215c1146 : FAILURE in 2m 58s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c504c5a7d61a4f289724a9e2668c678f : FAILURE in 3m 46s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b209aaef0c2741b796580a2ab8b31b7c : FAILURE in 3m 41s\n- loci-monasca-events-api https://zuul.opendev.org/t/openstack/build/95991fcc112f4c2ab483e2acc50d8c58 : FAILURE in 3m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/682263ad2a9b4c288f18982d145f7d7b : FAILURE in 3m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b26416c81699412d85d2d4c2679936d9 : FAILURE in 3m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ea0f62a1149d40b1861d3d7bd85d236c : FAILURE in 3m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8d8592b6c7af440fb24e321708ee3954 : FAILURE in 3m 13s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8e6caf4a541a4954add19f0f8b0b29dd : FAILURE in 3m 11s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5ad04d9a1fe34ea28d4e9335d2acef28 : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":6},{"id":"0f3b7fd8cc6eee754c636dfefb38adaed3220deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-19 08:56:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1eddeafd5c282074a2aad1553f7bbd92b6ccba38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 09:50:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4f1621187f324688a54e2b6bf7653aaa : SUCCESS in 44m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/dfda0263d7b746c48fdb39caa85f63d6 : SUCCESS in 31m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/eff37271993843c7a5867f7c719012fd : SUCCESS in 38m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3f309993883f4f6ebfe1e13c74bf61ea : SUCCESS in 42m 27s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d1f3a565785843bfb87bcb948ff9ac95 : SUCCESS in 46m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e0a0ce0f39604a07a95a076450552467 : SUCCESS in 44m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c706374f7c5a47f4a4cd35c25b2f3ae8 : SUCCESS in 35m 40s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/61d8768eb1394786bfb9aa11109c6cf1 : SUCCESS in 42m 22s\n- loci-monasca-events-api https://zuul.opendev.org/t/openstack/build/f18b9db17e2047458972289992b72fee : FAILURE in 40m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a06c2ce48a2444f08a0f273320604f0b : SUCCESS in 38m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/05683356e3c04d5a934178598ff3b9a5 : SUCCESS in 44m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b50c91d09fd84a20b868a4601ed34e62 : SUCCESS in 47m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6645e4ed20ad42e690eff370ebbd2eba : SUCCESS in 31m 17s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/fb6d9d9ec3124edc9685f22a5e9fc1d8 : SUCCESS in 30m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fd90de01eb534a4dabd5f52434305dc2 : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":7},{"id":"f4ee4e58757bec69c9f1dc42c7459d8ffbdb7fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-19 10:02:53.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4a828a1edba23d310e6c33edb6e7f13534a8fe55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 10:59:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b6c8c5521751491b81ee9f0b0361a4da : SUCCESS in 32m 15s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a00d69e34493475d94ffb1f73a1a2346 : SUCCESS in 36m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5e51713a09c7443bbcb0df1767c400ff : SUCCESS in 37m 02s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/859864a18ad8467a8ff847cd529832b2 : SUCCESS in 41m 39s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d394b57dbe9d424eb7bbe218c36e3c6f : SUCCESS in 48m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6432bd8fd5bb4c4b903f3bf1deea365f : SUCCESS in 48m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e156a43fa52441e1ac7593a7d5c0f51b : SUCCESS in 47m 12s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6e4cdb631f2d4a97851fb8bc22f8928e : SUCCESS in 20m 40s\n- loci-monasca-events-api https://zuul.opendev.org/t/openstack/build/173a206251a944e9b766650b26280efc : FAILURE in 36m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/668e6e14db95469fab562259469baa7a : SUCCESS in 49m 45s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7c1208e5a6ed4324844b486eea16312c : SUCCESS in 48m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8413ae56d45e4c25a29e32465e15e0c3 : SUCCESS in 42m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/72fff9f3658b4e25a86b75a843f20afc : SUCCESS in 34m 06s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b44b58cd50c54be7b3d3e9547d33fce5 : SUCCESS in 29m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7ba01f57e89345a98ef6289b9c5f9297 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":8},{"id":"f4b957a903b79e00d432573347017e8e48904950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-19 15:04:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e7e87a5fbe6b7c46f8c8df7b16a7bc56724b3c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 16:30:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/bb9e316126574d439d86cae469d2441a : TIMED_OUT in 1h 00m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7684688051704317b05b16691cfd09bd : TIMED_OUT in 1h 00m 46s\n- loci-heat https://zuul.opendev.org/t/openstack/build/607d9b29b5dc482daa2d68b173b9e44a : TIMED_OUT in 1h 00m 51s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3d5a5f89f06b47c4a2d7447e6fb7b0bc : TIMED_OUT in 1h 00m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b01f68fe236247a6bcdc92c8e1317c1b : TIMED_OUT in 1h 00m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d7c432b7b3c5406ea22713ac2d72abdd : TIMED_OUT in 1h 00m 46s\n- loci-manila https://zuul.opendev.org/t/openstack/build/354600fe37cb457ea92bbe15004d7503 : TIMED_OUT in 1h 00m 49s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/0f91d12178304a6e8a5f1cb619af99da : TIMED_OUT in 1h 00m 51s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9eab23da71034f17ac3f5de8061bf86e : SUCCESS in 41m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cf4389a54e5e45d9ae5e64a2cf1aa44e : TIMED_OUT in 1h 00m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ee64f41181b14513a00768a861d43ce6 : TIMED_OUT in 1h 01m 00s\n- loci-placement https://zuul.opendev.org/t/openstack/build/20cc7fc4f4b14d6b98c8a9555ba43aec : SUCCESS in 18m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e43797138fb045a58abd630432d322cf : SUCCESS in 14m 49s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1e19582201244e7a8d5052e24c1a56a4 : SUCCESS in 21m 36s","accounts_in_message":[],"_revision_number":9},{"id":"a4c7dd7e3b86d1ea4324ddbf227c241635bb84bb","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-20 12:44:33.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0e1039cdb90fdb1839d987e390a7037fa11d871f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 14:01:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a24fab4f84e348c7af1b2145d3537300 : SUCCESS in 36m 42s\n- loci-glance https://zuul.opendev.org/t/openstack/build/eb4e6b98422f40b79ac921be56ca2bea : SUCCESS in 23m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/58b8b8ef4ad348b0a48bf343dd15f01d : SUCCESS in 57m 04s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/77a2fcb04d5c45a8b27f614b6c5817e1 : SUCCESS in 1h 00m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/86fbd3f4fa49436890841f2c7ac12da0 : SUCCESS in 36m 22s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6819084f97cb4e74b2f4ad1d347f7da0 : SUCCESS in 34m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d2567543769149dbba4afac3531adaef : SUCCESS in 44m 09s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/36c59df67eb64428b32d894f892e21b3 : SUCCESS in 37m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/327619e4af854e189913a61140934877 : SUCCESS in 41m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/aac607e9a9584d7fb380f4e2e521ee93 : SUCCESS in 45m 13s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e582061948e64b1da64873f4cdbc5960 : SUCCESS in 57m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d7513d385a7249f1928cb94412aa811b : SUCCESS in 50m 14s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d43a67d9e1aa47a7b94fbbc23581d776 : SUCCESS in 32m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bccf794dae974551809ab36acd5a8b1f : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":9},{"id":"4c45d464f3f0fb6534a737804b0bbf57f42ee1b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-01-25 12:11:34.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"1f07f2329322095d1eb0b0bc0ad6fc12ef3de478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 13:07:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5f6399c363eb43da8864904c340b1b98 : SUCCESS in 49m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d891af0ad345485a90d01573cbdabc16 : SUCCESS in 47m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5e8fb8699d8947d6995fc3ba2028f359 : SUCCESS in 48m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/905a4d20aa6d4d268389e03f5a5c7639 : SUCCESS in 39m 33s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/99073d7d19d9461bb70a22732970c1d4 : SUCCESS in 20m 03s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1120a19b80784ec6a8c58d35c2028ea6 : SUCCESS in 42m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6b1744259b484fffa9064b76efedc738 : SUCCESS in 44m 48s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2feb4cfb579248c193db199b73d5995c : SUCCESS in 24m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1cfc32006c414845a24c1d4c85fc0954 : SUCCESS in 26m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d5f62db2451d4a06aa50f96af4c8cafb : SUCCESS in 24m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0ac8748b398e438b80333f449536f26a : SUCCESS in 40m 36s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1fe8b43ccd1a4ede86cdddc56ea0d403 : SUCCESS in 30m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b940cdba4fcb407f8589fdddf9dc5730 : SUCCESS in 25m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8bc36399d57c4a5e9210777a2a0094df : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":10},{"id":"5719c9d2d34d3c559d9d459afcfe4f9972e3b95d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-04 16:56:42.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6334ab39f47c5dfdd295ecd7896f9ab148bc5c81","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-04 17:06:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ea0b6cedabb509f68196a2741d81535a7704a532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-05 16:16:27.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent, and commit message as Patch Set 10.","accounts_in_message":[],"_revision_number":11},{"id":"2d8dcce44e04d032b5816e1e799cca1f0d633326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-05 16:19:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"50fdf9394051f6fe34735fcdb387854e1e71509c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-05 16:24:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"335c1bf0116e1e3fadd966a3463dde307985141a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-02-05 16:26:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0a5e2ef10def175cbf8ef7b04908009fc9a50145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:24:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b8bdc737a30f43308268c3bb57093a8b : SUCCESS in 39m 35s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0cce1c4bebfc4cfba2253e7308263359 : SUCCESS in 50m 55s\n- loci-heat https://zuul.opendev.org/t/openstack/build/df6c019528a84049a846acbbca80f21d : SUCCESS in 39m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5f9d72ad792c4208beed47ea061cb7dc : SUCCESS in 38m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5d1fca8df63a44949f0edf480317e617 : SUCCESS in 37m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/40f1d582f341485eaa3fefcfb458ff4a : SUCCESS in 37m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/86680ff815af491eb511ded9b031ed9e : SUCCESS in 33m 59s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/37e23fc8fc4c44678bf76d9cfa8b6f82 : SUCCESS in 41m 52s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/40cf3dd7a1b042c897ff2eeb84122e60 : SUCCESS in 50m 27s\n- loci-nova https://zuul.opendev.org/t/openstack/build/99ee785045164d17ad994ed1520b8674 : SUCCESS in 51m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a0aa85f85bbe4944bea4c8a012340639 : SUCCESS in 48m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b23271d5614d4bd1b8c5b0bbb96fe082 : SUCCESS in 32m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/994f815ffdfa4358b3134557389dc906 : SUCCESS in 27m 44s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3289dfbb919f4c658c32c180ff33fb7d : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":13},{"id":"8a1d84af983825c713762e9873c18e605148a8be","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-05 17:30:39.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"2f852a57a6cc49c739c1c85b284d8a01520ed2de","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2021-02-16 14:16:44.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"ce6607a33a98bf0725b86da42e3112673022d0db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 14:17:06.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"175be0b6fafd49f1ca198e29e6e44aa6be680ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 15:13:19.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3932af25826b40c5ac5f58e063012993 : SUCCESS in 37m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3c43f618f2094392867f7df469b95160 : SUCCESS in 39m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e21ead81e8ce49fb8d4b6ac8333ab1fe : SUCCESS in 38m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/06bdac4d9ef041a1a00932da64733535 : SUCCESS in 40m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c2d8084edc204eddb5c849b159a6bc92 : SUCCESS in 37m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6b35f33390c1474693c7cfbad054053d : SUCCESS in 36m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/271998d2402244e69e66031488cf6ffa : SUCCESS in 36m 56s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/34efd03c10274e68afc50f78cb6a1048 : SUCCESS in 35m 17s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/82171a49c7854eb3a93b903ea8f9c859 : SUCCESS in 40m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/aac1a7f6873f4c5e9ab855f51d13c83f : SUCCESS in 49m 45s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/66a5feb924ca41718ef505fc0670b814 : SUCCESS in 43m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/af9c76303fef4ceeb65bef5dae82e7da : SUCCESS in 34m 03s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/34a45e55113141e9afce1471a82a2074 : SUCCESS in 28m 01s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a7c3af798d49428f82c397d93e7533d9 : SUCCESS in 38m 21s","accounts_in_message":[],"_revision_number":13},{"id":"a76e0b367986b4957f5950781403ad598a28ed44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 15:13:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"181ef533696ad1b167d6609f199773924b310486","revisions":{"d762e6393f124520020ad716ed9ed143660c0e6d":{"kind":"REWORK","_number":1,"created":"2021-01-18 18:55:56.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/1"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d762e6393f124520020ad716ed9ed143660c0e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d762e6393f124520020ad716ed9ed143660c0e6d"}]},"branch":"refs/heads/master"},"853e2ca7c7c2df6d0d5f93d5c1148fc73dbecfc6":{"kind":"REWORK","_number":2,"created":"2021-01-18 19:17:33.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/2"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 19:17:09.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/853e2ca7c7c2df6d0d5f93d5c1148fc73dbecfc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/853e2ca7c7c2df6d0d5f93d5c1148fc73dbecfc6"}]},"branch":"refs/heads/master"},"9b0af59ea84798a34e9de6f1e24aff11972ac414":{"kind":"REWORK","_number":3,"created":"2021-01-18 20:39:28.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/3"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 20:39:04.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9b0af59ea84798a34e9de6f1e24aff11972ac414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9b0af59ea84798a34e9de6f1e24aff11972ac414"}]},"branch":"refs/heads/master"},"af7870803c36d79205847b99b36d06fd53c1f711":{"kind":"REWORK","_number":4,"created":"2021-01-18 21:00:36.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/4"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 21:00:13.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/af7870803c36d79205847b99b36d06fd53c1f711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/af7870803c36d79205847b99b36d06fd53c1f711"}]},"branch":"refs/heads/master"},"c9412afc0fa5348fd52e3097aa29782c0644077d":{"kind":"REWORK","_number":5,"created":"2021-01-18 21:31:07.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/5"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 21:30:46.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9412afc0fa5348fd52e3097aa29782c0644077d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9412afc0fa5348fd52e3097aa29782c0644077d"}]},"branch":"refs/heads/master"},"2ac81fed9960e33a400283b389742b91f412514e":{"kind":"REWORK","_number":6,"created":"2021-01-19 08:43:16.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/6"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-19 08:42:30.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2ac81fed9960e33a400283b389742b91f412514e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2ac81fed9960e33a400283b389742b91f412514e"}]},"branch":"refs/heads/master"},"98a52c105c852591962000c94161b4305bdca1f2":{"kind":"REWORK","_number":7,"created":"2021-01-19 08:56:02.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/7"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-19 08:55:26.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/98a52c105c852591962000c94161b4305bdca1f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/98a52c105c852591962000c94161b4305bdca1f2"}]},"branch":"refs/heads/master"},"ca36ddfed5d5514c0c28e0fe7eb1a0862f85aa0c":{"kind":"REWORK","_number":8,"created":"2021-01-19 10:02:53.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/8"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-19 10:02:20.000000000","tz":120},"subject":"Add monasca profile","message":"Add monasca profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ca36ddfed5d5514c0c28e0fe7eb1a0862f85aa0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ca36ddfed5d5514c0c28e0fe7eb1a0862f85aa0c"}]},"branch":"refs/heads/master"},"ba5bd6aadb072920bffde57e806a550dc9fb4dc6":{"kind":"REWORK","_number":9,"created":"2021-01-19 15:04:50.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/9"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-19 15:04:04.000000000","tz":120},"subject":"Add monasca-api profile","message":"Add monasca-api profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ba5bd6aadb072920bffde57e806a550dc9fb4dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ba5bd6aadb072920bffde57e806a550dc9fb4dc6"}]},"branch":"refs/heads/master"},"5d055714db72a184319e2d2ccd22a1198322685a":{"kind":"NO_CHANGE","_number":10,"created":"2021-01-25 12:11:34.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/10"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-25 12:11:05.000000000","tz":120},"subject":"Add monasca-api profile","message":"Add monasca-api profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d055714db72a184319e2d2ccd22a1198322685a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d055714db72a184319e2d2ccd22a1198322685a"}]},"branch":"refs/heads/master"},"21e6777e554faa1f33457376d17a497bd4b8ff59":{"kind":"NO_CHANGE","_number":11,"created":"2021-02-05 16:16:27.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/11"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-02-05 16:16:23.000000000","tz":120},"subject":"Add monasca-api profile","message":"Add monasca-api profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/21e6777e554faa1f33457376d17a497bd4b8ff59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/21e6777e554faa1f33457376d17a497bd4b8ff59"}]},"branch":"refs/heads/master"},"e2f28362c0861cd2bd78f335ff149074ba0a3f64":{"kind":"REWORK","_number":12,"created":"2021-02-05 16:19:11.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/12"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-02-05 16:18:47.000000000","tz":120},"subject":"Add monasca-api profile","message":"Add monasca-api profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e2f28362c0861cd2bd78f335ff149074ba0a3f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e2f28362c0861cd2bd78f335ff149074ba0a3f64"}]},"branch":"refs/heads/master"},"181ef533696ad1b167d6609f199773924b310486":{"kind":"REWORK","_number":13,"created":"2021-02-05 16:26:27.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/65/771265/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/65/771265/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/65/771265/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/65/771265/13"}}},"commit":{"parents":[{"commit":"c1a589698e5675db0c6c586d9a3a00d8cfb435b9","subject":"Don\u0027t use pip index if KEEP_ALL_WHEELS!\u003dno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1a589698e5675db0c6c586d9a3a00d8cfb435b9"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-01-18 18:55:15.000000000","tz":120},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2021-02-05 16:26:14.000000000","tz":120},"subject":"Add monasca-api profile","message":"Add monasca-api profile\n\nAdd zuul ci\n\nChange-Id: I30760e6f7378c35d4afea837b238967509017da5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/181ef533696ad1b167d6609f199773924b310486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/181ef533696ad1b167d6609f199773924b310486"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
