)]}'
{"id":"openstack%2Floci~775714","triplet_id":"openstack%2Floci~master~Id99bab9f136978db40a1d26053afa86dce711e22","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"23928":{"account":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"last_update":"2022-01-21 21:17:31.000000000","reason":"Change was abandoned"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-01-21 21:17:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id99bab9f136978db40a1d26053afa86dce711e22","subject":"Default to python3","status":"ABANDONED","created":"2021-02-15 22:52:39.000000000","updated":"2022-01-21 21:17:31.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b87b6bc02e659c42c6664e1e046c14274fd80516","_number":775714,"virtual_id_number":775714,"owner":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-16 15:41:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 00:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-16 15:19:20.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"CC"},{"updated":"2022-01-21 21:06:24.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"CC"}],"messages":[{"id":"42e9a853b783605d55ff8d192ad09dc5918af961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2021-02-15 22:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"415389623ff89cc73a3c27453d785109ac18162b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 00:47:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5775bac7799541bca4016b77b644ec00 : FAILURE in 28m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/fb7f7a6033eb474c902c8de76acb4041 : FAILURE in 27m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/86f3a7311cbc43fca118e6ca72851be7 : FAILURE in 24m 03s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/85910a3bf02d4efd99801fb90667e463 : FAILURE in 24m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b311bc43ee5543ce95f0900c58a1d2b1 : FAILURE in 24m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e1816c2fdf024e4daa2af8028eabde8f : FAILURE in 27m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/adc69b19171c4483af9dbe436781e736 : FAILURE in 24m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ccad8c9184374ee0ba35e632d9faa7ee : FAILURE in 18m 20s\n- loci-nova https://zuul.opendev.org/t/openstack/build/029000c72b904f12bb9ade26d8de9ef1 : FAILURE in 24m 35s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2030bffc81cf4507be44bfc3b267a7ca : FAILURE in 23m 11s\n- loci-placement https://zuul.opendev.org/t/openstack/build/356a1bc5abfe40278e67c449736a6540 : SUCCESS in 36m 11s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c39ae3244652478c940947f841de4f03 : FAILURE in 23m 04s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c47a56ae97b44b8e9a6a2cae840c4991 : FAILURE in 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"34e4fba6e2d114a26882dee41689b7b57cea9d74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2021-02-16 14:17:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"783f1e17868ac49158b14ab1b286aaf0288b4f13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 14:57:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4bc7910d95d24d7781a3d67e637c66bb : FAILURE in 15m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c0adde469e2e4bb998119e9576927ccc : FAILURE in 26m 11s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c41ef549e866461a839467cba26dfe91 : FAILURE in 25m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/dd8c86a18b21446d8edacec2e303fb25 : FAILURE in 29m 39s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8a2194fcf343415586afd76bc168aa0d : FAILURE in 26m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5b4e47ed7d2f42469ca03dd5339d04df : FAILURE in 25m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/beee1811eac74f1b9321870d6f971661 : FAILURE in 23m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2ccdc3f484c646498a73a7ac7a396084 : FAILURE in 22m 05s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c0f9bd93790b4c61bb7303a8d88c9f4b : FAILURE in 26m 13s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e9c95d15226c468b82a4f0f2e2dc8b19 : FAILURE in 25m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b2bcd127cd1c44d3bf5fc8882c77e813 : SUCCESS in 17m 23s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/dc61af62e7754244bba3ad64b6c3befa : FAILURE in 25m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1d619a0098f24bf3a988d0a112aa0686 : FAILURE in 24m 59s","accounts_in_message":[],"_revision_number":2},{"id":"177fcde467c19e74f28e82831a5b027b576a7317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-04-16 14:29:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9b1eedc489d5d0f05f7d09886dcec9b0140ce291","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-04-16 15:19:20.000000000","message":"Patch Set 3:\n\nThat will take much more, Gage. Let me do it in a bit.","accounts_in_message":[],"_revision_number":3},{"id":"f9598a8545d8d431b0981ce6181325a254e598f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 15:41:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3a31db6df2744662b0451c78157a0b8d : FAILURE in 38m 48s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4e404bb7f2b44742845cc969cb1b7f89 : FAILURE in 38m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6f0fc701b0e04e5fa05540832e169cb8 : FAILURE in 42m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/012512dea8fd4560a1e9ad1c8d13689a : FAILURE in 42m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9f6601020750443d908a5178d689cc23 : FAILURE in 37m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/372266afe24e4c80b2b748a60b17dd8f : FAILURE in 36m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4ef42f7df0a84f81af3eedd076612f74 : FAILURE in 37m 13s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2fc613fb9bde4c0087122eaebe98c34e : FAILURE in 38m 03s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2981bd4c82c94a9eb379fb6ba89b0cce : FAILURE in 38m 32s\n- loci-nova https://zuul.opendev.org/t/openstack/build/28811e87c2764ac3b8a610f9455c6c3c : FAILURE in 50m 50s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ece8ae0a6b4c441d92d64413d036a9f1 : FAILURE in 46m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6e94d47f770d484bae9acd5bcea49e22 : SUCCESS in 1h 00m 21s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b4cb6aca5e6b4b5497dd249e2526c3ff : FAILURE in 46m 41s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1319496e5e0445aca47d2bd04b1d60ca : FAILURE in 45m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b15bc3ec33143c83c9463001db8426d3b2d0c840","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:06:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3126e91b8b612414e142a32bf0aeb82b044f9cce","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-01-21 21:15:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b87b6bc02e659c42c6664e1e046c14274fd80516","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:17:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aedbe508af5741bfe72de3df3053acd3682b57f3","revisions":{"ce5b83f40d08229523da87a881d05421189be7c9":{"kind":"REWORK","_number":1,"created":"2021-02-15 22:52:39.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/14/775714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/775714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/775714/1"}}},"commit":{"parents":[{"commit":"e66ffb928e9d0a8daff4859ced3c79f1994f71f3","subject":"Update debian distro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e66ffb928e9d0a8daff4859ced3c79f1994f71f3"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2021-02-15 22:52:03.000000000","tz":-360},"committer":{"name":"Pete Birley","email":"pete@port.direct","date":"2021-02-15 22:52:03.000000000","tz":-360},"subject":"Default to python3","message":"Default to python3\n\nIts the year 2021, LOCI should default to python3\n\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\nChange-Id: Id99bab9f136978db40a1d26053afa86dce711e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce5b83f40d08229523da87a881d05421189be7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce5b83f40d08229523da87a881d05421189be7c9"}]},"branch":"refs/heads/master"},"dbd98eda5c611f3216839e983ef2b5b4d16fd642":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-16 14:17:17.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/14/775714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/775714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/775714/2"}}},"commit":{"parents":[{"commit":"0157128488b9245157d68af69a05e3ff4c76bc9d","subject":"Merge \"Set DEBIAN_FRONTEND\u003dnoninteractive when installing from apt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0157128488b9245157d68af69a05e3ff4c76bc9d"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2021-02-15 22:52:03.000000000","tz":-360},"committer":{"name":"Pete Birley","email":"petebirley@gmail.com","date":"2021-02-16 14:17:17.000000000","tz":0},"subject":"Default to python3","message":"Default to python3\n\nIts the year 2021, LOCI should default to python3\n\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\nChange-Id: Id99bab9f136978db40a1d26053afa86dce711e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dbd98eda5c611f3216839e983ef2b5b4d16fd642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dbd98eda5c611f3216839e983ef2b5b4d16fd642"}]},"branch":"refs/heads/master","description":"Rebase"},"aedbe508af5741bfe72de3df3053acd3682b57f3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-16 14:29:19.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/775714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/775714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/775714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/775714/3"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Pete Birley","email":"pete@port.direct","date":"2021-02-15 22:52:03.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2021-04-16 14:29:19.000000000","tz":0},"subject":"Default to python3","message":"Default to python3\n\nIts the year 2021, LOCI should default to python3\n\nSigned-off-by: Pete Birley \u003cpete@port.direct\u003e\nChange-Id: Id99bab9f136978db40a1d26053afa86dce711e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/aedbe508af5741bfe72de3df3053acd3682b57f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/aedbe508af5741bfe72de3df3053acd3682b57f3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
