)]}'
{"id":"openstack%2Floci~777276","triplet_id":"openstack%2Floci~master~Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f","subject":"Fix virtualenv version comparison in setup_pip.sh","status":"ABANDONED","created":"2021-02-24 03:03:43.000000000","updated":"2022-05-16 20:42:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1a82f06ac0f266fbf26ec18056e64648b72ed0f","_number":777276,"virtual_id_number":777276,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-20 20:25:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 04:29:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 00:00:19.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2021-05-07 00:02:32.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"ac3fa6a4c3fdabe096c27a7e3df55bcc93a60b88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-24 03:03:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6f806e70a49900a9b967cbf6658a0e75e552c44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-24 03:05:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d7e495b89c5a90fb2922b611b46e176f3e768d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-02-24 03:31:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"500b02f8b1f2ed756eed53464f6ff84bb5d5ae09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:29:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c57a138b55964f738605e69b00909d93 : SUCCESS in 54m 42s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cef216497bdc41f598a3f6d6c9cb5096 : SUCCESS in 43m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b5127b22f6394a79b7e60f2fffcbe97e : SUCCESS in 48m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0f4a30491c7e46b4b6c4699da8ff5d5e : SUCCESS in 45m 27s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c1397e139b544c83bd5c33906648d6c2 : SUCCESS in 47m 22s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d899752aaa2c4429ac2e72959206f53d : SUCCESS in 46m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/39b402023ddc43dca442743a59c12a3b : SUCCESS in 50m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/91c665b2aeb548b58722cd4b4b02c690 : SUCCESS in 46m 19s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7fca11bda20842e7bdc31bc568760bfa : SUCCESS in 55m 15s\n- loci-nova https://zuul.opendev.org/t/openstack/build/627a423a1e5846a382f4fe69ceadfa9b : SUCCESS in 42m 08s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c088d069ac454590b09d5c0207b4a0b0 : SUCCESS in 53m 25s\n- loci-placement https://zuul.opendev.org/t/openstack/build/873456f2efdc459e8911d3f61b7b6820 : SUCCESS in 37m 07s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/788867c6123340c9824ae155d4865170 : SUCCESS in 39m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/844dad64d25141718f239b9db1e924c9 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f0f6df9ea7305006e68d22b51863d158e575d6b5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-05-07 00:00:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9c973435e59501dda3c00c63cc6aa390d69023c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-05-07 00:02:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"45fec264b5ef6aa53de0feed9682bfebe78a28a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 00:41:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f127b1d69a1744a5b6ad9d7e25bb96ab : FAILURE in 26m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c3289dee7f9d466997634976d728ecc6 : SUCCESS in 33m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3675d59aae904824ab6def094ce24df3 : FAILURE in 26m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cd989fa2de2e4a4087d48e023aa2f6bf : SUCCESS in 34m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/da441adb4ffd46678a6e35e8ca39df72 : SUCCESS in 33m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/897c21452e2d40c2a53967406c1fcc04 : SUCCESS in 33m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/21c6c70df6fc4751bac170f01a92f148 : SUCCESS in 33m 26s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/a5c955f19cc5437891a87dac826ec832 : SUCCESS in 33m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e3aed3d223b148329ebedbff5efffa8f : SUCCESS in 35m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8310024d86994567878b4617969da6cd : SUCCESS in 37m 01s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/52609b49d0264ea984df530e034b7d23 : SUCCESS in 33m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/65ae8b66c2f64534847795955529bd41 : FAILURE in 26m 38s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0612ebd54f98431ab6913cdd937bbb08 : SUCCESS in 26m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/40509c3a6c6a499c8130cb9fc7b15f1e : SUCCESS in 33m 32s","accounts_in_message":[],"_revision_number":4},{"id":"c10742821dc789a3d0e93728ccefda4b27d167b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-20 19:51:50.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b2ad3bb6bc11382420bcda1491b2d7fd8780ab55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 20:25:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8df4ecdb72794efd95c3886fe8f5d9e7 : SUCCESS in 23m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e6411ec511e545f28d79c007e7f11d31 : SUCCESS in 22m 33s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6bb408f79d8a431f97c132ac62b91fb7 : FAILURE in 17m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cd15ceec95af491cad6e40f9227e3f63 : FAILURE in 18m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/21c44c64d5a94217980b9c3d1311ddfc : SUCCESS in 22m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/515a1269763646eeb710a43e899e75d3 : SUCCESS in 23m 17s\n- loci-manila https://zuul.opendev.org/t/openstack/build/78ab9154499c4a60b0c03835d9d88e27 : SUCCESS in 22m 35s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d6f977385df14e49bf4bcd43b4774ba6 : SUCCESS in 20m 57s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c109d0c911454ea78babec78f9d74b64 : SUCCESS in 24m 44s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c0590557e4c74330a4453a259a4f5813 : SUCCESS in 25m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/cb41dc9cb3c545c0808877305e666401 : SUCCESS in 23m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0f751025b29f43639074740ce40609d2 : SUCCESS in 31m 27s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5304ec538aa141bda5e8abe68c7ecb3b : SUCCESS in 16m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/94a5b6f691a344a6828d29391410261b : SUCCESS in 22m 22s","accounts_in_message":[],"_revision_number":5},{"id":"f1a82f06ac0f266fbf26ec18056e64648b72ed0f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-16 20:42:37.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5d18d5e75738e7fb25a54851d535ceed11fc3cbb","revisions":{"44a3658968ca0a52d8d148252273732245fd30f4":{"kind":"REWORK","_number":1,"created":"2021-02-24 03:03:43.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/76/777276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/76/777276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/76/777276/1"}}},"commit":{"parents":[{"commit":"57a75eba725f2aeea97a993f01f830ac4938bfa3","subject":"Merge \"Enable Debian gate build and fix mirrors related issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/57a75eba725f2aeea97a993f01f830ac4938bfa3"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:03:11.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:03:11.000000000","tz":-360},"subject":"Fix virtualenv version comparison in setup_pip.sh","message":"Fix virtualenv version comparison in setup_pip.sh\n\nChange-Id: Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/44a3658968ca0a52d8d148252273732245fd30f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/44a3658968ca0a52d8d148252273732245fd30f4"}]},"branch":"refs/heads/master"},"37dff7ce093a53e3ae3511d4b2c3df1d09ec098c":{"kind":"REWORK","_number":2,"created":"2021-02-24 03:05:36.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/76/777276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/76/777276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/76/777276/2"}}},"commit":{"parents":[{"commit":"57a75eba725f2aeea97a993f01f830ac4938bfa3","subject":"Merge \"Enable Debian gate build and fix mirrors related issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/57a75eba725f2aeea97a993f01f830ac4938bfa3"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:03:11.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:05:31.000000000","tz":-360},"subject":"Fix virtualenv version comparison in setup_pip.sh","message":"Fix virtualenv version comparison in setup_pip.sh\n\nChange-Id: Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/37dff7ce093a53e3ae3511d4b2c3df1d09ec098c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/37dff7ce093a53e3ae3511d4b2c3df1d09ec098c"}]},"branch":"refs/heads/master"},"c49e5e30e69f7b854989493ff92eb8ea4a2e005f":{"kind":"REWORK","_number":3,"created":"2021-02-24 03:31:07.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/76/777276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/76/777276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/76/777276/3"}}},"commit":{"parents":[{"commit":"57a75eba725f2aeea97a993f01f830ac4938bfa3","subject":"Merge \"Enable Debian gate build and fix mirrors related issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/57a75eba725f2aeea97a993f01f830ac4938bfa3"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:03:11.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:30:02.000000000","tz":-360},"subject":"Fix virtualenv version comparison in setup_pip.sh","message":"Fix virtualenv version comparison in setup_pip.sh\n\nAlso remove special logic for pip\u003e\u003d20 since it\u0027s not compatible with pip\u003e20.3\n\nChange-Id: Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c49e5e30e69f7b854989493ff92eb8ea4a2e005f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c49e5e30e69f7b854989493ff92eb8ea4a2e005f"}]},"branch":"refs/heads/master"},"9699aa88a0c40a2e9102a9255a23aa21a10c2d2a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-05-07 00:02:32.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/76/777276/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/76/777276/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/76/777276/4"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:03:11.000000000","tz":-360},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-05-07 00:01:50.000000000","tz":-240},"subject":"Fix virtualenv version comparison in setup_pip.sh","message":"Fix virtualenv version comparison in setup_pip.sh\n\nAlso remove special logic for pip\u003e\u003d20 since it\u0027s not compatible with pip\u003e20.3\n\nChange-Id: Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9699aa88a0c40a2e9102a9255a23aa21a10c2d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9699aa88a0c40a2e9102a9255a23aa21a10c2d2a"}]},"branch":"refs/heads/master"},"5d18d5e75738e7fb25a54851d535ceed11fc3cbb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-20 19:51:50.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/76/777276/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/76/777276/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/76/777276/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/76/777276/5"}}},"commit":{"parents":[{"commit":"766179e0ea379313843bc2af25c9aa8f0add64e9","subject":"Reduce build targets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/766179e0ea379313843bc2af25c9aa8f0add64e9"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-02-24 03:03:11.000000000","tz":-360},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:51:50.000000000","tz":0},"subject":"Fix virtualenv version comparison in setup_pip.sh","message":"Fix virtualenv version comparison in setup_pip.sh\n\nAlso remove special logic for pip\u003e\u003d20 since it\u0027s not compatible with pip\u003e20.3\n\nChange-Id: Ieacd8922207a44ae5f3596c462cd74e1f9b7f14f\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d18d5e75738e7fb25a54851d535ceed11fc3cbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d18d5e75738e7fb25a54851d535ceed11fc3cbb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
