)]}'
{"id":"openstack%2Floci~801526","triplet_id":"openstack%2Floci~master~Ideb497c169828184d301b6be4359a7c2228aa444","project":"openstack/loci","branch":"master","topic":"build-targets","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-06-15 16:55:38.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-06-15 16:55:38.000000000","reason":"Change was submitted"},"33519":{"account":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"last_update":"2022-06-15 16:25:55.000000000","reason":"Tin Lam replied on the change"}},"hashtags":[],"change_id":"Ideb497c169828184d301b6be4359a7c2228aa444","subject":"Reduce build targets","status":"MERGED","created":"2021-07-20 19:49:31.000000000","updated":"2022-06-15 16:55:38.000000000","submitted":"2022-06-15 16:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801526-build-targets","meta_rev_id":"5f37cd2120120c30cb3d858dcaf23df7b9265f96","_number":801526,"virtual_id_number":801526,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-15 16:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-05-23 16:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-23 16:14:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2022-06-15 16:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2022-06-15 16:25:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-20 20:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-16 20:43:34.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"CC"},{"updated":"2022-05-23 16:30:19.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2022-06-03 14:44:58.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"state":"REVIEWER"}],"messages":[{"id":"c93393b462150afc0a54d0962f8b693ce7105f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-20 19:49:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f3b16f82f57f3a5fb6d525ce9df1fbe6478429a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 20:34:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8f8167c1ea34419ea9bd61503217b95f : SUCCESS in 24m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d64ff05e034045b68ca3c3d2f17dbcdd : SUCCESS in 40m 20s\n- loci-heat https://zuul.opendev.org/t/openstack/build/29db0ac4af584c9894710b3736d3f9d2 : SUCCESS in 29m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5f5ebe751fef487c92b95638e97b9b36 : SUCCESS in 20m 51s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/83224c6a4d1745d3955ba4d32ab27c61 : SUCCESS in 27m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/aec8ae7f105044f3bd9df3bc52368235 : FAILURE in 17m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f4c5954ac0844d66aa124e0d967e88b0 : FAILURE in 17m 39s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d22a7a3eaab84ea6b7fee5c1d0a05816 : SUCCESS in 28m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0ac4ddb3da754846a17ed4ec87a5ca9a : SUCCESS in 29m 52s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8c4c654782d24d4094ca041882b73c35 : FAILURE in 15m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b13c94a790b24eef83908de3d596e394 : SUCCESS in 41m 46s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f3de494e6a504e9a9a21f1925291a551 : SUCCESS in 27m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4d37e731fad946e5b591e9de503eb666 : SUCCESS in 17m 19s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/789d356497e145928df6e680e6f3b671 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e380d84d89fd77ca76ec7e92a1c495f32fb9503f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-28 16:34:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"728584f48a2e0ecff7a28d78bbecea174c8393e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-28 17:11:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03c3603ee932b9425a3dba1f9bb39a460550702f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:41:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-keystone https://zuul.opendev.org/t/openstack/build/311fe968291c465981f8ef52f7523ab7 : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":2},{"id":"34bd21afb792e12cebb4ab4ff1fae4deeb23d9d5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-28 19:55:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b8566a80cbf2df6ad7e0918b53b44e1cf944627a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 20:20:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- loci-keystone https://zuul.opendev.org/t/openstack/build/32f541ad633c4e5b93188448ed291398 : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":2},{"id":"91c6b0898d200233ecb15560ffe9d6319c98a6b2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-07-29 20:03:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"339c893f5c1a7d6923545313fd7d39697f32681d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 20:33:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- loci-keystone https://zuul.opendev.org/t/openstack/build/43b0ec8e0bdc4574842192b2e4800edf : SUCCESS in 26m 09s","accounts_in_message":[],"_revision_number":2},{"id":"9288b82f53f29465d208df7450781b4246d5714b","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-16 20:43:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee6fe604c55a91d8ca48ef5d24e8b86042e551ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-17 19:44:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d7897850b8e070ca0c0f9eeddd9cfbdc9ff6ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-17 19:45:50.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"adbea3ea078bf8ba16ebfebb6fe32020d585d2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 20:05:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ccb76a24b12c44dab41d01ca423fe985 : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":4},{"id":"44532347b343360df4c9ab7c9ec5fd75a33529da","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-17 20:36:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c23189db9ffd3b3f78f787c325b81a9b9d3d660","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-19 17:43:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ac8f92885a5100e99301fcb560b50e04be78c904","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-19 17:45:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fa670e6be260209c69d00a7d59edd5e5f8d94834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:25:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4f8a030625254817a7bae538511be435 : SUCCESS in 35m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cc1dc3e872bf4e77ac2138e1f38d61bc : SUCCESS in 23m 34s\n- loci-heat https://zuul.opendev.org/t/openstack/build/563cb6674dd54dc78b715b4e57bc410b : SUCCESS in 34m 47s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fec15d03ac3d4b7ea4b0a63237db97cd : SUCCESS in 34m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b23a7de6f2a44fdb97e6a7273f5c0575 : SUCCESS in 34m 24s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b29845731be942988664e6a7c509ca9c : SUCCESS in 20m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/29016ba3d4614d0aa19047f4ae2f2d3c : SUCCESS in 20m 01s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/08900c348d47422bb4c37e85d2763150 : SUCCESS in 19m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6d454fe6d2ac4de88db3f9fd90f33a90 : SUCCESS in 18m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2ea9fc1f5cd74fd99641a8c37c51b6da : SUCCESS in 22m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/262468c3c911491b8e2ae14bc4f4c95f : SUCCESS in 19m 29s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3f7484d10ad2447da3f61e3756838b5e : SUCCESS in 19m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/16c48844d7e4409fa1b0a7d049222130 : SUCCESS in 16m 40s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/e6a8240be86e4fd8af9c14e5569e9beb : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":5},{"id":"c9127e6f83e8450e9e50060b84e319176749b0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-23 15:59:49.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"674238eb67c0199f9bce3b9adfab2e4a034b40a1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2022-05-23 16:14:03.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"25154c0184128e7f00ffbf76e70f729f760b5b37","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-05-23 16:30:19.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ea34b66d67fd599672481b19f5ca9d2143bd33f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 16:30:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1df67385cc4a44d4b14ffdb79165777c : SUCCESS in 21m 21s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3a7278e8d7d24a6f85797c41b019f7a8 : SUCCESS in 19m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0db64ac91cd64a23be2e895cebb88302 : SUCCESS in 22m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6596cca4057c44a5be5e64b206b080a1 : SUCCESS in 21m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e741997a008945f0a1c0a393c5b70c1f : SUCCESS in 22m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/714dd740db454bcda84efb20b892e359 : SUCCESS in 19m 38s\n- loci-manila https://zuul.opendev.org/t/openstack/build/34236cb54b4942e3b79da2c447336f1c : SUCCESS in 20m 08s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4b06859de4e544e3a8d8c025355d3c14 : SUCCESS in 19m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ea29545b9f07442e8efd5a7c86dd9462 : SUCCESS in 21m 06s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6319f491daee4625b969807baca6258c : SUCCESS in 25m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/39f6908b7a8442488a2c04e4028a347d : SUCCESS in 21m 13s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2f7e4e7b1abf4edea63f67d91b763e00 : SUCCESS in 20m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/972a4c387d784a06964ae69ac524ff27 : SUCCESS in 15m 46s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/01d5b940256843889c00fc06e11ccab5 : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":6},{"id":"7eb0235d3392ce57da6b619532c18380a445040a","author":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"date":"2022-06-15 16:25:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"aacfe4d989fda021c696ef348df166eea4b54387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 16:26:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f3227aa561c6e4cc084f036d77d4cff1ead8433d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 16:55:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0a6e8cf9312546d3a9cf62225b173833 : SUCCESS in 18m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6f2945ac6c07425a85caa7437855414c : SUCCESS in 22m 08s\n- loci-heat https://zuul.opendev.org/t/openstack/build/044101c8ba2646ae9a54e7120cb0d74b : SUCCESS in 18m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/38506bd2e1e240f5939d2c3eb5f468ce : SUCCESS in 23m 51s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/714add01b3644e4c93bce2c3cd40af57 : SUCCESS in 19m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1d0edc0d0d8b449d984a8e3837a67ed5 : SUCCESS in 17m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2c93247233ed4c858fd0d75786d0dfa8 : SUCCESS in 19m 42s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2b2142d233bb40fb9d5697b28757ff3b : SUCCESS in 20m 30s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4a43a0fa6fdd4a43a1dff87f6e3df04c : SUCCESS in 19m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8cfa83f638944bcd8045952bbe8800ab : SUCCESS in 22m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d05cc950c6354a5bb080bff28d2b6e11 : SUCCESS in 17m 12s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c66e0518107c482987e9801400a6ca34 : SUCCESS in 16m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5a9b198596cd483a9bdb20535e3af9fd : SUCCESS in 13m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d04430f2e4864f4187b70271c2eb153d : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":6},{"id":"5f37cd2120120c30cb3d858dcaf23df7b9265f96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 16:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9e7b71c3043245baeae16ecc0c8f3563e2c25c98","revisions":{"766179e0ea379313843bc2af25c9aa8f0add64e9":{"kind":"REWORK","_number":1,"created":"2021-07-20 19:49:31.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/26/801526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/801526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/801526/1"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:25.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:31.000000000","tz":-240},"subject":"Reduce build targets","message":"Reduce build targets\n\nUbuntu Xenial is EOL since April 2021, which also runs builds\nfor Ocata only that also has been in EM for a long time and\nnot really maintained.\n\nOpenSUSE LEAP is currently not in use by any known downstream\nusers and it has been broken for quite sometime, leaving the gate\nin a broken state and unable to land any code.  There\u0027s no active\nmaintainer for it as well.\n\nDebian support was added by us (VEXXHOST) but we decided that\nwe\u0027re not going to continue using it and move towards Ubuntu, so\nwe do not want to maintain it, as well as that it is building images\nfor Train only.\n\nIt also updates the README file accordingly as well as includes\nfocal in there which has been recently added.\n\nChange-Id: Ideb497c169828184d301b6be4359a7c2228aa444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/766179e0ea379313843bc2af25c9aa8f0add64e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/766179e0ea379313843bc2af25c9aa8f0add64e9"}]},"branch":"refs/heads/master"},"e929d6861c8fd99fe6e3d5c83ecd6c80a117ad94":{"kind":"REWORK","_number":2,"created":"2021-07-28 17:11:37.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/26/801526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/801526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/801526/2"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:25.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-28 17:11:33.000000000","tz":-240},"subject":"Reduce build targets","message":"Reduce build targets\n\nUbuntu Xenial is EOL since April 2021, which also runs builds\nfor Ocata only that also has been in EM for a long time and\nnot really maintained.\n\nOpenSUSE LEAP is currently not in use by any known downstream\nusers and it has been broken for quite sometime, leaving the gate\nin a broken state and unable to land any code.  There\u0027s no active\nmaintainer for it as well.\n\nDebian support was added by us (VEXXHOST) but we decided that\nwe\u0027re not going to continue using it and move towards Ubuntu, so\nwe do not want to maintain it, as well as that it is building images\nfor Train only.\n\nIt also updates the README file accordingly as well as includes\nfocal in there which has been recently added.\n\nChange-Id: Ideb497c169828184d301b6be4359a7c2228aa444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e929d6861c8fd99fe6e3d5c83ecd6c80a117ad94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e929d6861c8fd99fe6e3d5c83ecd6c80a117ad94"}]},"branch":"refs/heads/master"},"529a0ec746baabcbdc17a6e3c9d9368602dd59d1":{"kind":"REWORK","_number":3,"created":"2022-05-17 19:44:26.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/801526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/801526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/801526/3"}}},"commit":{"parents":[{"commit":"eb95b4adab5b8659ba95131c4bc234776b21cda5","subject":"Merge \"Remove supported releases for placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eb95b4adab5b8659ba95131c4bc234776b21cda5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:25.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-05-17 19:44:03.000000000","tz":-300},"subject":"Reduce build targets","message":"Reduce build targets\n\nUbuntu Xenial is EOL since April 2021, which also runs builds\nfor Ocata only that also has been in EM for a long time and\nnot really maintained.\n\nOpenSUSE LEAP is currently not in use by any known downstream\nusers and it has been broken for quite sometime, leaving the gate\nin a broken state and unable to land any code.  There\u0027s no active\nmaintainer for it as well.\n\nDebian support was added by us (VEXXHOST) but we decided that\nwe\u0027re not going to continue using it and move towards Ubuntu, so\nwe do not want to maintain it, as well as that it is building images\nfor Train only.\n\nIt also updates the README file accordingly as well as includes\nfocal in there which has been recently added.\n\nChange-Id: Ideb497c169828184d301b6be4359a7c2228aa444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/529a0ec746baabcbdc17a6e3c9d9368602dd59d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/529a0ec746baabcbdc17a6e3c9d9368602dd59d1"}]},"branch":"refs/heads/master"},"ac7291257ae5c1053ad3df1bb8c935d9501d8aff":{"kind":"REWORK","_number":4,"created":"2022-05-17 19:45:50.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/801526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/801526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/801526/4"}}},"commit":{"parents":[{"commit":"eb95b4adab5b8659ba95131c4bc234776b21cda5","subject":"Merge \"Remove supported releases for placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eb95b4adab5b8659ba95131c4bc234776b21cda5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:25.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-05-17 19:45:50.000000000","tz":0},"subject":"Reduce build targets","message":"Reduce build targets\n\nUbuntu Xenial is EOL since April 2021, which also runs builds\nfor Ocata only that also has been in EM for a long time and\nnot really maintained.\n\nOpenSUSE LEAP is currently not in use by any known downstream\nusers and it has been broken for quite sometime, leaving the gate\nin a broken state and unable to land any code.  There\u0027s no active\nmaintainer for it as well.\n\nDebian support was added by us (VEXXHOST) but we decided that\nwe\u0027re not going to continue using it and move towards Ubuntu, so\nwe do not want to maintain it, as well as that it is building images\nfor Train only.\n\nIt also updates the README file accordingly as well as includes\nfocal in there which has been recently added.\n\nChange-Id: Ideb497c169828184d301b6be4359a7c2228aa444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac7291257ae5c1053ad3df1bb8c935d9501d8aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac7291257ae5c1053ad3df1bb8c935d9501d8aff"}]},"branch":"refs/heads/master"},"86a51eeac795119c004f2db9385ab432e877de5d":{"kind":"REWORK","_number":5,"created":"2022-05-19 17:43:50.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/801526/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/801526/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/801526/5"}}},"commit":{"parents":[{"commit":"eb95b4adab5b8659ba95131c4bc234776b21cda5","subject":"Merge \"Remove supported releases for placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eb95b4adab5b8659ba95131c4bc234776b21cda5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:25.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-05-19 17:43:45.000000000","tz":0},"subject":"Reduce build targets","message":"Reduce build targets\n\nUbuntu Xenial is EOL since April 2021, which also runs builds\nfor Ocata only that also has been in EM for a long time and\nnot really maintained.\n\nOpenSUSE LEAP is currently not in use by any known downstream\nusers and it has been broken for quite sometime, leaving the gate\nin a broken state and unable to land any code.  There\u0027s no active\nmaintainer for it as well.\n\nDebian support was added by us (VEXXHOST) but we decided that\nwe\u0027re not going to continue using it and move towards Ubuntu, so\nwe do not want to maintain it, as well as that it is building images\nfor Train only.\n\nIt also updates the README file accordingly as well as includes\nfocal in there which has been recently added.\n\nChange-Id: Ideb497c169828184d301b6be4359a7c2228aa444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/86a51eeac795119c004f2db9385ab432e877de5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/86a51eeac795119c004f2db9385ab432e877de5d"}]},"branch":"refs/heads/master"},"9e7b71c3043245baeae16ecc0c8f3563e2c25c98":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-05-23 15:59:49.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/801526/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/801526/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/801526/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/801526/6"}}},"commit":{"parents":[{"commit":"eb95b4adab5b8659ba95131c4bc234776b21cda5","subject":"Merge \"Remove supported releases for placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eb95b4adab5b8659ba95131c4bc234776b21cda5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-07-20 19:49:25.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-05-23 15:59:49.000000000","tz":0},"subject":"Reduce build targets","message":"Reduce build targets\n\nOpenSUSE LEAP is currently not in use by any known downstream\nusers and it has been broken for quite sometime, leaving the gate\nin a broken state and unable to land any code.  There\u0027s no active\nmaintainer for it as well.\n\nDebian support was added by us (VEXXHOST) but we decided that\nwe\u0027re not going to continue using it and move towards Ubuntu, so\nwe do not want to maintain it, as well as that it is building images\nfor Train only.\n\nIt also updates the README file accordingly as well as includes\nfocal in there which has been recently added.\n\nChange-Id: Ideb497c169828184d301b6be4359a7c2228aa444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9e7b71c3043245baeae16ecc0c8f3563e2c25c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9e7b71c3043245baeae16ecc0c8f3563e2c25c98"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
