)]}'
{"id":"openstack%2Floci~802920","triplet_id":"openstack%2Floci~master~I2b5b1a31165e98ce71380824e8f57af8906701c6","project":"openstack/loci","branch":"master","topic":"victoria-wallaby-support","hashtags":[],"change_id":"I2b5b1a31165e98ce71380824e8f57af8906701c6","subject":"Fix setup_pip.sh and remove xenial based builds","status":"MERGED","created":"2021-07-29 17:03:49.000000000","updated":"2021-08-02 18:50:04.000000000","submitted":"2021-08-02 18:50:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802920-victoria-wallaby-support","meta_rev_id":"617b36c0db83b88a025c5215c6eb0ac4c1edcee1","_number":802920,"virtual_id_number":802920,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 18:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2021-08-02 17:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":2,"date":"2021-08-02 16:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2021-08-02 17:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"}],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 19:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-02 15:49:51.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2021-08-02 15:49:51.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"state":"REVIEWER"},{"updated":"2021-08-02 15:49:51.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":24780,"name":"Sangeet Gupta","email":"sg774j@att.com","username":"sgupta"},"state":"REVIEWER"},{"updated":"2021-08-02 15:49:51.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":30777,"name":"Steven Fitzpatrick","email":"steven@fitzpatrick.wtf","username":"sf280x"},"state":"REVIEWER"},{"updated":"2021-08-02 15:49:51.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"421f381ac1651e99869c54ba10f22278d294aabd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 17:03:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f61ae5c8156843684d45882badc64d544a8e2e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 18:10:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7f13296bf4ffb85c73a855650b61f2d4f387920f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 19:38:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b70a005d0a464d01ac6a8b4ccf0505ec : FAILURE in 58m 28s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8a0b997d0bbd41d9be67aec5aaa1563e : FAILURE in 56m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/501966a40ff742778b17ce14a7963799 : FAILURE in 1h 12m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8c41da3d046c440db291ff15d4db4975 : FAILURE in 58m 01s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8ca514ba00ea4df2879bd134d36425e1 : FAILURE in 1h 09m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/460fcfa44b2a4955ab08dcc0ca1eeaa8 : FAILURE in 1h 24m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/64dc3744579242f486d2c3a68486c264 : FAILURE in 1h 09m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d3e767931ce54fd3a3d04108dcd0d4e5 : FAILURE in 55m 10s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b355c65ad60b4b739ece77f343f7a788 : FAILURE in 1h 02m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cef48b0055b04865a22950eb24b0b05c : FAILURE in 58m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5b24e8e7ba77462788eee3ab3cfffff2 : FAILURE in 57m 12s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a63c6628ec0247ac9a0b591e61cd7a64 : FAILURE in 51m 34s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b6f60e70fd0a4b2885971103132a7d6c : SUCCESS in 43m 32s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/22d4cb418af34521a67bd8a989b53045 : FAILURE in 57m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f999a657d41cbc57ca6db49c15f411458b3a59f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 20:27:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"85ebef356d6a5f5a1922805f857127627c9a45f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 21:33:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7e404a194d07487d8c4bf53477036bcd : FAILURE in 40m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b947548e31944353a0cb82382f1a8222 : FAILURE in 58m 33s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1326ca758afd498d8b7804db363fe754 : FAILURE in 54m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a55ae4b0ef184606817cf7ade02b66f0 : FAILURE in 51m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/47bd1fd861fd4cb586bd611856b3fb0c : FAILURE in 48m 53s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b3b0d85889314fc58d4b367f7d036c81 : FAILURE in 57m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9d53ad899aff4c53bed76ce7cc01bf8e : FAILURE in 55m 12s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2ea4903d2f704583b237395406869455 : FAILURE in 52m 51s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/88e1d32c941845f4a1becf8b47b52712 : FAILURE in 44m 44s\n- loci-nova https://zuul.opendev.org/t/openstack/build/73f953d4cae24d05ab9ae2b2aee62d94 : FAILURE in 59m 23s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9c85417ad1f44404b6196a68d84b86ed : FAILURE in 56m 25s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5dc779019a3f467bbbb01af3b031cdcd : FAILURE in 53m 37s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b17287bf51a94d359d2da8f1c862ddf3 : FAILURE in 40m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8070c1b3eecd4822a1bcc467505152a4 : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":3},{"id":"4473bcab9454b9a94aa1a8245ba814013abd596e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 21:33:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"00f5d7514f9acc0d1e497154f8931a642ef2599f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 22:22:41.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"edb14cf9e692d304e1cf93793cb5aab4ac99d616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 23:26:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"69a2cc3d44c6fce783a83420ea5d2bef5efe1da0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-07-29 23:28:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eb03ce86a4a3c967d77f22abf28c0148e1177157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 00:36:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7a49fc0d8dff4aa5882366cc92edfacd : SUCCESS in 40m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f5e55885ce434f989a88a04ed6253eb5 : SUCCESS in 53m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/791de2238dfd478985f36e55461b272f : SUCCESS in 55m 16s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b2ab23e3a2ca45a08398e26ddad0802e : SUCCESS in 53m 17s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e3e58c39a3cc4cbe9add30eb3411cddd : SUCCESS in 1h 04m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5af6034a50364ca89c71bc663090b904 : SUCCESS in 1h 03m 31s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3ab092da74b044a3abc05979b001bf40 : SUCCESS in 49m 34s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/decc695837b244d98068278b28ef8b1e : SUCCESS in 47m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0d6efd7238f94db8983292034f23f789 : SUCCESS in 57m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/53ee616afab44c4e90dc505ea925524b : SUCCESS in 48m 43s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7fd45ef9f07043ff955a689bd0749e39 : SUCCESS in 55m 11s\n- loci-placement https://zuul.opendev.org/t/openstack/build/07dad2dee0884bccb3375fbea889f20e : SUCCESS in 1h 03m 11s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4511a5572d2845fda6903b93292738f9 : SUCCESS in 30m 07s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c66863facb374353803f935b5fa3a78f : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":7},{"id":"884c2dd8c65ddde146626ea7a68fdb8dc911317b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-08-02 16:15:27.000000000","message":"Topic set to victoria-wallaby-support","accounts_in_message":[],"_revision_number":7},{"id":"0c3c748f53058098020f74b85bddfa6a00a0dac8","author":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"date":"2021-08-02 16:35:33.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b8213a283f2245b333550a2f803940856fc1a96a","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2021-08-02 17:49:28.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"070d5bc24d0144388824f6b4496e53690b6bf452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 17:49:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e8218b4730d6c8dddc22f4b94fe37b87e7510df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 18:50:04.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/37669a0023f54b178fcc10ab3ee6c6aa : SUCCESS in 54m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/5cecc2cd13b6432ab4505869ce25b8b5 : SUCCESS in 53m 20s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1fcfcadcbdf3414ead0bc589f3c7efc1 : SUCCESS in 54m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/bab811905aa2448ca30ecf5a2e3b2653 : SUCCESS in 55m 32s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b9668235378c480d8f3ac5422909bc0b : SUCCESS in 53m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/56e1cb9a5c174f708872028a6071d6fe : SUCCESS in 53m 17s\n- loci-manila https://zuul.opendev.org/t/openstack/build/fc94df60affa4217812674b0e9b177de : SUCCESS in 52m 59s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4dd0b08bbbf8439887038a150c1bb528 : SUCCESS in 52m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7d8b015f45d8428f9bdfb2b802c988ee : SUCCESS in 56m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/23fa4791b16640d8a5565d9f5cb5ad51 : SUCCESS in 59m 07s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e17caf9409824908a43e825e7ab0cf9e : SUCCESS in 54m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bbf8f6366aeb49fd9c7ce6929f8bcc6e : SUCCESS in 53m 19s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/de5c1a1d444546048372d5a80a374b4c : SUCCESS in 32m 46s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bc4f1b50646e4814935ee4552291c50c : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":7},{"id":"617b36c0db83b88a025c5215c6eb0ac4c1edcee1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 18:50:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c19db02c6dec9fba5b450bc36f12b9d5623f699c","revisions":{"59ea83b481df349b7cd94d6b8effdd01343c300f":{"kind":"REWORK","_number":1,"created":"2021-07-29 17:03:49.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/1"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"subject":"Fix setup_pip.sh","message":"Fix setup_pip.sh\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59ea83b481df349b7cd94d6b8effdd01343c300f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59ea83b481df349b7cd94d6b8effdd01343c300f"}]},"branch":"refs/heads/master"},"e16e6af589163c15ec8bd8123e5c9c3c30e92df7":{"kind":"REWORK","_number":2,"created":"2021-07-29 18:10:35.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/2"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-07-29 18:10:29.000000000","tz":0},"subject":"Fix setup_pip.sh","message":"Fix setup_pip.sh\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e16e6af589163c15ec8bd8123e5c9c3c30e92df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e16e6af589163c15ec8bd8123e5c9c3c30e92df7"}]},"branch":"refs/heads/master"},"22650222ffbed8b72848f3cd63be293d7273d537":{"kind":"REWORK","_number":3,"created":"2021-07-29 20:27:59.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/3"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 20:27:40.000000000","tz":-360},"subject":"Fix setup_pip.sh and remove xenial based builds","message":"Fix setup_pip.sh and remove xenial based builds\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/22650222ffbed8b72848f3cd63be293d7273d537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/22650222ffbed8b72848f3cd63be293d7273d537"}]},"branch":"refs/heads/master"},"4f35c49b9599d6b4d1bf0bd7ac08998227e3f640":{"kind":"REWORK","_number":4,"created":"2021-07-29 21:33:38.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/4"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 21:33:32.000000000","tz":-360},"subject":"Fix setup_pip.sh and remove xenial based builds","message":"Fix setup_pip.sh and remove xenial based builds\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4f35c49b9599d6b4d1bf0bd7ac08998227e3f640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4f35c49b9599d6b4d1bf0bd7ac08998227e3f640"}]},"branch":"refs/heads/master"},"d3545a7a8b01de067642e572cd56ffee238a9df4":{"kind":"REWORK","_number":5,"created":"2021-07-29 22:22:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/5"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-07-29 22:22:31.000000000","tz":0},"subject":"Fix setup_pip.sh and remove xenial based builds","message":"Fix setup_pip.sh and remove xenial based builds\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d3545a7a8b01de067642e572cd56ffee238a9df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d3545a7a8b01de067642e572cd56ffee238a9df4"}]},"branch":"refs/heads/master"},"0ad067dcb6c92d30300255d447e76d6dda8f93ba":{"kind":"REWORK","_number":6,"created":"2021-07-29 23:26:42.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/6"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 23:26:19.000000000","tz":-360},"subject":"Fix setup_pip.sh and remove xenial based builds","message":"Fix setup_pip.sh and remove xenial based builds\n\nAlso fix monasca bindep.\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0ad067dcb6c92d30300255d447e76d6dda8f93ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0ad067dcb6c92d30300255d447e76d6dda8f93ba"}]},"branch":"refs/heads/master"},"c19db02c6dec9fba5b450bc36f12b9d5623f699c":{"kind":"REWORK","_number":7,"created":"2021-07-29 23:28:19.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/20/802920/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/802920/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/802920/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/802920/7"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 17:03:37.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-07-29 23:27:51.000000000","tz":-360},"subject":"Fix setup_pip.sh and remove xenial based builds","message":"Fix setup_pip.sh and remove xenial based builds\n\nAlso fix monasca bindep.\n\nChange-Id: I2b5b1a31165e98ce71380824e8f57af8906701c6\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c19db02c6dec9fba5b450bc36f12b9d5623f699c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c19db02c6dec9fba5b450bc36f12b9d5623f699c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
