)]}'
{"id":"openstack%2Floci~810070","triplet_id":"openstack%2Floci~master~I47008d2947aad4440563c20810c64a97d6c5ce69","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I47008d2947aad4440563c20810c64a97d6c5ce69","subject":"[test]","status":"ABANDONED","created":"2021-09-20 15:58:40.000000000","updated":"2022-05-16 20:44:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"817c1be4ab5989dfdc8463ffa4c63889e3710ba5","_number":810070,"virtual_id_number":810070,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-20 17:13:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 17:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5ff31b159fa78489e92148cbc0a85f2fc43c2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-09-20 15:58:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a98b07926324f0a8de76695925207645d1c4dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 17:13:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/229e45b3abff4283978c62fa8196f293 : FAILURE in 40m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/527d8fa85d23455cb5438a6a6fafe7b1 : FAILURE in 42m 10s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d4f7ff2587d14be1b2b2df8710c4388f : FAILURE in 37m 22s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4a2138f1efa54e3ea35a8f04adda4fdf : FAILURE in 35m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/158db31a1e10475896536f9601b6cd93 : FAILURE in 41m 43s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9e75035d0add408c8452952c56b22eca : FAILURE in 34m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/65c1e06077cc429b8debcd3978f3f361 : FAILURE in 36m 47s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/695b55e1e0aa4e36b928e6d255af1461 : FAILURE in 35m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/165f0016d28c46ba8d5bc5ac01cc45e6 : FAILURE in 34m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/db522d05478440919b6dabde917819ab : FAILURE in 34m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/65087cc06f2348e59ae9976ddaf4f7e6 : FAILURE in 37m 36s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cc257fa0304d4f88b4b18c1a47f486c7 : FAILURE in 34m 59s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/df9e77bcf58d490ebb44a2a13e2398b4 : TIMED_OUT in 1h 01m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/81c191b090d54d75bed4472adeecba8e : FAILURE in 38m 40s","accounts_in_message":[],"_revision_number":1},{"id":"817c1be4ab5989dfdc8463ffa4c63889e3710ba5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-16 20:44:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbc7beb5891a92cd5c7edd4915a43f53c45a9a64","revisions":{"dbc7beb5891a92cd5c7edd4915a43f53c45a9a64":{"kind":"REWORK","_number":1,"created":"2021-09-20 15:58:40.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/70/810070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/70/810070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/810070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/810070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/810070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/70/810070/1"}}},"commit":{"parents":[{"commit":"c19db02c6dec9fba5b450bc36f12b9d5623f699c","subject":"Fix setup_pip.sh and remove xenial based builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c19db02c6dec9fba5b450bc36f12b9d5623f699c"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-09-20 15:58:29.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-09-20 15:58:29.000000000","tz":-360},"subject":"[test]","message":"[test]\n\nChange-Id: I47008d2947aad4440563c20810c64a97d6c5ce69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dbc7beb5891a92cd5c7edd4915a43f53c45a9a64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dbc7beb5891a92cd5c7edd4915a43f53c45a9a64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
