)]}'
{"id":"openstack%2Floci~810110","triplet_id":"openstack%2Floci~master~I81ec48dbd123e6f24ff05d64906d47c31ae693f7","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I81ec48dbd123e6f24ff05d64906d47c31ae693f7","subject":"Pin virtualenv to 20.7.2","status":"MERGED","created":"2021-09-20 18:21:30.000000000","updated":"2021-09-20 23:02:49.000000000","submitted":"2021-09-20 21:44:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810110","meta_rev_id":"44618ff83fb51fe9db617cd9bd934156edcfe8a1","_number":810110,"virtual_id_number":810110,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 21:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2021-09-20 20:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":2,"date":"2021-09-20 23:02:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":2,"date":"2021-09-20 19:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-20 23:02:49.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2021-09-20 20:13:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":1,"date":"2021-09-20 19:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 18:32:52.000000000","updated_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"reviewer":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"state":"REVIEWER"},{"updated":"2021-09-20 18:36:10.000000000","updated_by":{"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"},"reviewer":{"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"},"state":"REVIEWER"},{"updated":"2021-09-20 19:51:54.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2021-09-20 20:13:34.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2021-09-20 20:39:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"117efc66e6edf456db214dfb55c170e4ac0d4201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-09-20 18:21:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6254262acc78380e1658ad73b57ad205ff903d30","author":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"date":"2021-09-20 18:32:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"011b83d0d627b5e5a04e49783eda5e1ecda88cf6","author":{"_account_id":33789,"name":"Josh Herrera","email":"v-josherrera@microsoft.com","username":"jh813b"},"date":"2021-09-20 18:36:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8354a70a76bb88af7240473910276653f7866979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-09-20 19:40:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0a425a039071e6ceeaac8a88c07ae47b11bcccbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-09-20 19:40:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1f15015f6a489f7453875fb127171bc9234bbafe","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-20 19:51:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a792205f1898839ea4a8aac58d6b5f02f9dad1da","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2021-09-20 20:13:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"93dee40e39ab9f0c59f25023a6bef96c3d5110cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 20:39:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/932c4c6390b24f7daea4446039e5e5b3 : SUCCESS in 55m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0d761c75c16a4dce93390a11e0c0a82a : SUCCESS in 44m 10s\n- loci-heat https://zuul.opendev.org/t/openstack/build/accce2363a1843ef92131727488a393d : SUCCESS in 56m 14s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/351c5a82be8e486d851bcbc32ebf453f : SUCCESS in 49m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e5b074eb554e41c1ab6ed6248e5f8504 : SUCCESS in 46m 15s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6a632c7d57854c9ca81fa947fa64513c : SUCCESS in 54m 40s\n- loci-manila https://zuul.opendev.org/t/openstack/build/583dfa6fe5c2457daa219b0e2a396b99 : SUCCESS in 43m 10s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/7c53d2daa24f4b72b25de1ca83beeeff : SUCCESS in 44m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2233dbced726422bb87426819ec2ae0b : SUCCESS in 50m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b8a93ba4754c45c7bed988f725f5429b : SUCCESS in 46m 49s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/44e324586bb14fa5b51760c426ecb446 : SUCCESS in 44m 07s\n- loci-placement https://zuul.opendev.org/t/openstack/build/94f0a477b97b473bb3aa9e49615af63d : SUCCESS in 43m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6b99f42f67d44d90afad09a96b06e42d : SUCCESS in 43m 08s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/beefd78386544664849acb74733678c2 : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":3},{"id":"0645d7fc1aed1b30597b32fd565273d53abb94a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 20:39:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4437fdd77a4b1098e1ddd6576323e986d828789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 21:44:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ef6c7c2af2da4197a700ec87be0a2a59 : SUCCESS in 50m 15s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ff9eeaa02d354d1b9288c9fbf42e29f7 : SUCCESS in 47m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7ec84b3121884392be3023aa1e2fad3c : SUCCESS in 58m 15s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b8a3eb1ce4ca47b7a80d7da76c7cfc88 : SUCCESS in 50m 13s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/afbe83bb7a56455ab96d51f5b4ce6558 : SUCCESS in 48m 30s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3c3f976c908c44c7ac018a708085c2b9 : SUCCESS in 47m 31s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f722e33953fb494999cce8e6c1c6dfd5 : SUCCESS in 47m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b964b7f5064d4b5494feb57adf5e7ae2 : SUCCESS in 46m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1da421c6645749c9a7d7220f9994b06c : SUCCESS in 51m 57s\n- loci-nova https://zuul.opendev.org/t/openstack/build/394de5bb4799477bb4ff154644f22200 : SUCCESS in 52m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/613e63b91a624fa1b3afd81ceb3ab99f : SUCCESS in 48m 49s\n- loci-placement https://zuul.opendev.org/t/openstack/build/99831680a8084ce3becac5035a9c2375 : SUCCESS in 47m 46s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/bb99eca8f8fc47dabf3183f7cc62b814 : SUCCESS in 36m 03s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/04546b005a3746bbbe0de650afe7d6a9 : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5a062b051c539eb1c61b360615ca4243fc5e99a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 21:44:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5ef275660065c76252bbbbe80cbd1f1d317b3b6e","tag":"autogenerated:gerrit:revert","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-09-20 23:02:49.000000000","message":"Created a revert of this change as I52d3e7b98ba425da55efb44aa80deb02b911e22a","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","revisions":{"2891f7525e0c180426a1dfbb906431bcd81cd41c":{"kind":"REWORK","_number":1,"created":"2021-09-20 18:21:30.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/10/810110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/810110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/810110/1"}}},"commit":{"parents":[{"commit":"c19db02c6dec9fba5b450bc36f12b9d5623f699c","subject":"Fix setup_pip.sh and remove xenial based builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c19db02c6dec9fba5b450bc36f12b9d5623f699c"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-09-20 18:19:04.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-09-20 18:19:04.000000000","tz":-360},"subject":"Pin virtualenv to 20.7.2","message":"Pin virtualenv to 20.7.2\n\nCurrent latest 20.8.0 have updated embedded pip 21.2.4 and setuptools\n58.4.8 incompatible with some of constrainted packages.\n\nChange-Id: I81ec48dbd123e6f24ff05d64906d47c31ae693f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2891f7525e0c180426a1dfbb906431bcd81cd41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2891f7525e0c180426a1dfbb906431bcd81cd41c"}]},"branch":"refs/heads/master"},"67e6bb1b34bdd95670eecaba20788aab03b25591":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-20 19:40:22.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/10/810110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/810110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/810110/2"}}},"commit":{"parents":[{"commit":"8baeb8ecea689642d56d82a2e6c4e35a2fefaa07","subject":"Add focal base to dockerfiles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8baeb8ecea689642d56d82a2e6c4e35a2fefaa07"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-09-20 18:19:04.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-09-20 19:40:22.000000000","tz":0},"subject":"Pin virtualenv to 20.7.2","message":"Pin virtualenv to 20.7.2\n\nCurrent latest 20.8.0 have updated embedded pip 21.2.4 and setuptools\n58.4.8 incompatible with some of constrainted packages.\n\nChange-Id: I81ec48dbd123e6f24ff05d64906d47c31ae693f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67e6bb1b34bdd95670eecaba20788aab03b25591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67e6bb1b34bdd95670eecaba20788aab03b25591"}]},"branch":"refs/heads/master","description":"Rebase"},"68c3d9177f83501c0d44dcb36bae0c286fa66d43":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-20 19:40:29.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/10/810110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/810110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/810110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/810110/3"}}},"commit":{"parents":[{"commit":"c19db02c6dec9fba5b450bc36f12b9d5623f699c","subject":"Fix setup_pip.sh and remove xenial based builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c19db02c6dec9fba5b450bc36f12b9d5623f699c"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-09-20 18:19:04.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-09-20 19:40:29.000000000","tz":0},"subject":"Pin virtualenv to 20.7.2","message":"Pin virtualenv to 20.7.2\n\nCurrent latest 20.8.0 have updated embedded pip 21.2.4 and setuptools\n58.4.8 incompatible with some of constrainted packages.\n\nChange-Id: I81ec48dbd123e6f24ff05d64906d47c31ae693f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
