)]}'
{"id":"openstack%2Floci~811757","triplet_id":"openstack%2Floci~master~I3f4f60756887ad07229cf522803b30d4e16463cd","project":"openstack/loci","branch":"master","topic":"remove-py3","attention_set":{},"removed_from_attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-11-10 17:51:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3f4f60756887ad07229cf522803b30d4e16463cd","subject":"Remove PYTHON3 hard-coded values","status":"ABANDONED","created":"2021-09-29 17:36:40.000000000","updated":"2022-11-10 17:51:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7141d37defb82558bd875c224feb6648b9274392","_number":811757,"virtual_id_number":811757,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-11 21:04:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-29 18:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7616dd0295d9f7b65d5953777f79983f9e024b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-09-29 17:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cde6a28e3af21bbda65923b553c980ae6c2a596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 18:13:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7744890172a6479eb28115aaecf9c60a : FAILURE in 33m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c092c833e30b4130ba2503c6140a9e3f : FAILURE in 35m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/616f212bf8934a46ac68008bf383c692 : FAILURE in 34m 42s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/dc719a7d5dc7461d94d9168f44ccfb07 : FAILURE in 29m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/67a3d85c5a4c4775bf7932a116178602 : FAILURE in 31m 04s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9bc98c86320a449e8edafd99a4e7360b : FAILURE in 35m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a33466e04b8b4f97b3e51081f801e911 : FAILURE in 28m 55s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f6931422e4fa4245a95095caf979157f : FAILURE in 30m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/25540be232ab4d51b61c166b682326fc : FAILURE in 34m 44s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6ece6a50c2a143268da8f900df4751be : FAILURE in 31m 31s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ecbf9c3b3f29474aae88822bbb4bd96c : FAILURE in 31m 14s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7b07b5775bda487eb72aeae199eb1397 : FAILURE in 30m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4336ed0e1b634fe9a4372fdc942367da : FAILURE in 29m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2f8d6b8ae5a64aaeb90ca4980a19426c : FAILURE in 30m 21s","accounts_in_message":[],"_revision_number":1},{"id":"742d545ef9986939f01b2277da5d18788ae53fec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2022-04-11 20:33:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77551fa6b4a260b623ed9b4d517e2c06ac6d58d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 21:04:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6a205384d2cd4f4cae42d648a3a65cc5 : FAILURE in 11m 42s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e33d865001804be7832385f435201b4c : FAILURE in 15m 15s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2cf0618f19fe4c55b5acbf22b6c28507 : FAILURE in 15m 28s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fc206a5fe73d4539b1b6b40343a64bc8 : FAILURE in 17m 48s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d2cd2442fc8e4b35b5541087188cc6c9 : FAILURE in 15m 24s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ab3d69803b664586bc44d3628f7e6843 : FAILURE in 15m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/25c7795ebc934df1bc5165e971092099 : FAILURE in 17m 02s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/9f8f67a25ed5400b971e0b7b93017658 : FAILURE in 17m 54s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b7cb6ee15e3e45db8dac736d9d3f1ab4 : FAILURE in 13m 08s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5c3b076fcc854cc39ad456e427ad24f1 : FAILURE in 17m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/886d231b50c645ada2a5c970006a466f : FAILURE in 14m 29s\n- loci-placement https://zuul.opendev.org/t/openstack/build/16938d126317444b95dcd5467f04e175 : SUCCESS in 11m 14s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4bd266e71a2d42a59ae4a508f6aefd83 : FAILURE in 18m 01s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/23dc662bbe8841d29f2bc4f0c129b1aa : FAILURE in 14m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7141d37defb82558bd875c224feb6648b9274392","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-10 17:51:54.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 212 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86e4fb05302462309102e940ed6b7cf3bc8efb80","revisions":{"972712dbb67deecdbe6ab3dfceb58ab890ff0802":{"kind":"REWORK","_number":1,"created":"2021-09-29 17:36:40.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/57/811757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/57/811757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/57/811757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/57/811757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/57/811757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/57/811757/1"}}},"commit":{"parents":[{"commit":"058ea993c6ecf5404ac7d631f686142f182ea4aa","subject":"Switch PYTHON3 to YES by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/058ea993c6ecf5404ac7d631f686142f182ea4aa"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-09-29 17:36:34.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-09-29 17:36:34.000000000","tz":-240},"subject":"Remove PYTHON3 hard-coded values","message":"Remove PYTHON3 hard-coded values\n\nChange-Id: I3f4f60756887ad07229cf522803b30d4e16463cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/972712dbb67deecdbe6ab3dfceb58ab890ff0802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/972712dbb67deecdbe6ab3dfceb58ab890ff0802"}]},"branch":"refs/heads/master"},"86e4fb05302462309102e940ed6b7cf3bc8efb80":{"kind":"REWORK","_number":2,"created":"2022-04-11 20:33:56.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/57/811757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/57/811757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/57/811757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/57/811757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/57/811757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/57/811757/2"}}},"commit":{"parents":[{"commit":"08b559993ba776a5cfc673168e77400e32ad36a5","subject":"debian: stop building images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/08b559993ba776a5cfc673168e77400e32ad36a5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-04-11 20:33:51.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-04-11 20:33:51.000000000","tz":-240},"subject":"Remove PYTHON3 hard-coded values","message":"Remove PYTHON3 hard-coded values\n\nChange-Id: I3f4f60756887ad07229cf522803b30d4e16463cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/86e4fb05302462309102e940ed6b7cf3bc8efb80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/86e4fb05302462309102e940ed6b7cf3bc8efb80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
