)]}'
{"id":"openstack%2Floci~811889","triplet_id":"openstack%2Floci~master~I1e293f345d01ee81fb7b685c6e8eba9e6fbe0c8f","project":"openstack/loci","branch":"master","topic":"masakari-profile","hashtags":[],"change_id":"I1e293f345d01ee81fb7b685c6e8eba9e6fbe0c8f","subject":"Added Masakari profile","status":"ABANDONED","created":"2021-09-30 08:20:18.000000000","updated":"2022-05-16 20:45:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92e0f8687e9a1c42d0aaf033127f688ab882a6d4","_number":811889,"virtual_id_number":811889,"owner":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-30 09:19:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-30 09:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42c9e41523f093947f24d0321c42302a252c2988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-09-30 08:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6d11609edc7b76495073d32235bf5d250dd8c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 09:19:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a0128863211d451cbf80c317eab20281 : FAILURE in 43m 54s\n- loci-glance https://zuul.opendev.org/t/openstack/build/95987913d0b046989c8bdc0049454739 : FAILURE in 43m 46s\n- loci-heat https://zuul.opendev.org/t/openstack/build/748536bace8a4d37a571fb9cd82b49e7 : FAILURE in 37m 34s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0bb1490e7cb94313990c604714f30152 : FAILURE in 38m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8365cce9b22d4ca6bccddce6b968c9ff : FAILURE in 38m 22s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7ccba285d8a24e1c9e226fec7ce36c43 : FAILURE in 37m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d7dc3929aa1346b6b53dc820ee819dc8 : FAILURE in 36m 14s\n- loci-masakari https://zuul.opendev.org/t/openstack/build/4e53dc9873f34780a05239c4150c71d6 : FAILURE in 34m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/75fdef11fd134c24985a028ffbe32d51 : FAILURE in 36m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/83dd89713bcb439ea5c0613078af174e : FAILURE in 32m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bfeb4c31e2a040138f4ab2168b748232 : FAILURE in 38m 11s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f505abb92e04433480f84c3a38ad002e : FAILURE in 53m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5170aca310e74093bfd3352e3fcbd211 : FAILURE in 37m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/29e6b3d77442456db4eda30069b15f6a : FAILURE in 32m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/13ef6d1c0670483fb42d654a7c84c3db : FAILURE in 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e5ee6736628301cf3eba752864b41f326a363a47","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-09-30 09:33:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a3ef402a4862f307fc42fc8f445cbac6ce9b35df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 10:26:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/44139698f0e14b009d7761ee0207915d : FAILURE in 41m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a180b7a5f1bf458598f4b1784423c18e : FAILURE in 37m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d761460aa34b441990a6b04b0fa6d90b : FAILURE in 36m 58s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/68da53fc29584515b3b7728e8639ad87 : FAILURE in 46m 12s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a988e42ce55d44199aa0b6131f3817d5 : FAILURE in 34m 45s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/928b0a8eae2c4746b62232bc7dc1ab7c : FAILURE in 47m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3d4db0fa5c314c0184ac1cdbea393076 : FAILURE in 34m 59s\n- loci-masakari https://zuul.opendev.org/t/openstack/build/110e0ca9d7574642aaa6770943b78569 : FAILURE in 48m 58s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/3262534e62ca41faa82ac84bc0b316e9 : FAILURE in 35m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d33333c5ce7d46d688380758e2027546 : FAILURE in 49m 18s\n- loci-nova https://zuul.opendev.org/t/openstack/build/435b9b05fed4464faee796236a47b37c : FAILURE in 49m 06s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/da899458a60540c797025acfabdcf898 : FAILURE in 45m 43s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cbe7b96681b54a6a8fcdb8c9b330816d : FAILURE in 47m 08s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d5a65481a8dc4cb4b35bbc02ff56a2b4 : FAILURE in 44m 19s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d3e830b5811c4ee7bf9057c661b63b45 : FAILURE in 37m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e8352e32719dd0823b954f633f4e636dc6005eef","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-09-30 13:52:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3758d03833c2c83c1a084e6b7febcf0afd4e2fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 14:35:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a530680c87a34ef084dfe521336102f2 : FAILURE in 40m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ce98506e33c64196a6b871d462eec072 : FAILURE in 39m 37s\n- loci-heat https://zuul.opendev.org/t/openstack/build/552a5bef4f0c4cfc9d8c98f31695b478 : FAILURE in 37m 15s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a7aeb132c82c4d1695e463a38f720452 : FAILURE in 39m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/cf00835a77224d098e4f07e56b0fcbe5 : FAILURE in 35m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/768fd8f930524a31a1c5307186582f1b : FAILURE in 35m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/388890ec8f0b4c43a008e5dac5e901e4 : FAILURE in 37m 49s\n- loci-masakari https://zuul.opendev.org/t/openstack/build/c6aff07f08614cdf84d4e2300e3a7522 : FAILURE in 36m 49s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/1fa65e2969f5404d9d588c8c9fa54a65 : FAILURE in 37m 30s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/240bbb4bd2f948bda12ba9a9b05934c3 : FAILURE in 37m 09s\n- loci-nova https://zuul.opendev.org/t/openstack/build/021f4c1fb49241d79ff91d709b6f6f54 : FAILURE in 37m 01s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e29f2a6a4d7d4b099d0d1907deb4f04e : FAILURE in 34m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5563d0ad8fc64c4196f4b880af6d55a3 : FAILURE in 38m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0a4bd33a775e41fca1f670250ccdb052 : FAILURE in 35m 00s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/88609d7241664bd2ac228360c5c05584 : FAILURE in 40m 01s","accounts_in_message":[],"_revision_number":1},{"id":"746b8a3e66eb0c4626fd5c8fb3897f845f740ffc","author":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"date":"2021-10-03 16:24:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb6ddc6e1e12243a250fb5b9eae6439371d83db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 17:12:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d33133b7b5164dfab123baacc1e448e5 : FAILURE in 37m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/61b024cb2a6743739ccbad6812eb869a : FAILURE in 40m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/09723f7109ec40acbc5ff8692b1c977d : FAILURE in 38m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ec38b58dda4644bc8de423ee503181e0 : FAILURE in 35m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7d571bf6568b41f19f70485a97dc7c89 : FAILURE in 44m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/63640436d0cd451d85583762367e38f2 : FAILURE in 34m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/05e7dc507dd24e239e52ab6c3eb969b5 : FAILURE in 34m 53s\n- loci-masakari https://zuul.opendev.org/t/openstack/build/138b0c97c7b6445e8e27a9bccb3d3cb2 : FAILURE in 35m 14s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f39b512a6e994bc4a9ed3900592a869a : FAILURE in 32m 21s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/592dfd40bf3f4294baeeee97db131053 : FAILURE in 33m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cbb0531230ae44af92b5214ac1de2dfa : FAILURE in 35m 06s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/defd09e4695f4ad7bc395066fa381ca7 : FAILURE in 44m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/54988be8eac2451bb904404c97b04e4e : FAILURE in 32m 03s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a3ab40cb90ae41379514dc378d0019db : FAILURE in 28m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5762476108fc40389fec482a49705687 : FAILURE in 34m 51s","accounts_in_message":[],"_revision_number":1},{"id":"92e0f8687e9a1c42d0aaf033127f688ab882a6d4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-16 20:45:07.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"209ba7ff93799c39d26c254fe105413c089b9fb1","revisions":{"209ba7ff93799c39d26c254fe105413c089b9fb1":{"kind":"REWORK","_number":1,"created":"2021-09-30 08:20:18.000000000","uploader":{"_account_id":33130,"name":"Susanta gautam","email":"xuxant02@gmail.com","username":"sgautam"},"ref":"refs/changes/89/811889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/89/811889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/89/811889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/89/811889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/89/811889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/89/811889/1"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-09-30 08:19:53.000000000","tz":345},"committer":{"name":"xuxant02@gmail.com","email":"xuxant02@gmail.com","date":"2021-09-30 08:20:14.000000000","tz":345},"subject":"Added Masakari profile","message":"Added Masakari profile\n\nChange-Id: I1e293f345d01ee81fb7b685c6e8eba9e6fbe0c8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/209ba7ff93799c39d26c254fe105413c089b9fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/209ba7ff93799c39d26c254fe105413c089b9fb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
