)]}'
{"id":"openstack%2Floci~813574","triplet_id":"openstack%2Floci~master~I9a70aebd9636e170c6e04398356933668a2bfbba","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"8863":{"account":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"last_update":"2021-10-13 16:30:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a70aebd9636e170c6e04398356933668a2bfbba","subject":"Update gates","status":"MERGED","created":"2021-10-12 05:21:05.000000000","updated":"2021-10-13 16:30:25.000000000","submitted":"2021-10-13 16:30:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813574","meta_rev_id":"6f6d41ccac0f3eecba4ffd6deabc568fb2617ddb","_number":813574,"virtual_id_number":813574,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 16:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":2,"date":"2021-10-13 14:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-13 15:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"all":[{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-13 15:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 06:23:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 14:36:02.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2021-10-13 15:26:11.000000000","updated_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"reviewer":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"state":"REVIEWER"}],"messages":[{"id":"403ecd731424220d36532dc204168802c014818b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 05:21:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcc084d476f03d260132477440d5c798ed64bb1d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 05:24:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"53a11443afc32116125a0a6455b0f281f8b92076","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 05:24:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a8ee5c053c9425ad699e2403f3d877f00336c63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 06:23:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/fc1c6a4057c44723bea07dead8dba4c0 : FAILURE in 40m 15s\n- loci-glance https://zuul.opendev.org/t/openstack/build/960eb8d1f7444c1886b8e126fd470f86 : FAILURE in 51m 48s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ea9856718901473aa7586103fbb60c0d : FAILURE in 36m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ddff6fd47b0b49a6b569debebd834f82 : FAILURE in 41m 14s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5b8b0ad5033b46ae91b168ae742e249b : FAILURE in 31m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fa1d1fe0c1ed4d9c82f3e6956820862f : FAILURE in 37m 43s\n- loci-manila https://zuul.opendev.org/t/openstack/build/89d5622892ce4cc2bdd2bb3cb31e3970 : FAILURE in 30m 34s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ef928ea9315f4fb8b94526728496cb51 : FAILURE in 38m 21s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/788bd71c0cdc4cc89ce1b2b69a8432ca : FAILURE in 39m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f933f7fcd85a4ad09020d03381c0e6df : FAILURE in 38m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/077614cd3c454c9c957abc5d8cb8b8e1 : FAILURE in 36m 31s\n- loci-placement https://zuul.opendev.org/t/openstack/build/75800c61ace24d8b94e2ff47a8c08dd0 : FAILURE in 40m 32s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4f19029780b04360a8eaa562ae161221 : FAILURE in 39m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3e8888cb65fe4ef88d2e02ca9a56f34a : FAILURE in 40m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f75d293127ad55f0e2ebb966a5962ff622cc9526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 18:46:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e01f7904c788551d7c7df8d3ff4465fa2dc60ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 19:45:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/402cce86767840f4b600828b0e2f89cd : FAILURE in 49m 23s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3657d2aca4584285ad972dd50cedac28 : FAILURE in 33m 46s\n- loci-heat https://zuul.opendev.org/t/openstack/build/cf88ccb73373468ca7a13ae43b0b388a : FAILURE in 45m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f15677ff6b414cbda77bb4b2a533796c : FAILURE in 33m 59s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/283604fcbc614d278cf1802efdb9863a : FAILURE in 37m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a55ab83c253e4d24b7a4171149781861 : FAILURE in 33m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/12056a00cd234fc49e66f82d6d091a86 : FAILURE in 34m 36s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/8fc7266568f045c0a363d15f3ac9f65e : FAILURE in 41m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c2fc183ab479449891397f2fbed33822 : FAILURE in 33m 32s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fee17159ba324fba9b7ddd155af0c724 : FAILURE in 33m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/04e29cb139394d96939708a151b91ec2 : FAILURE in 35m 51s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0f58f85a823e46719735f64b07c59162 : FAILURE in 30m 45s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1e43bb884ab741569fd11d46618d9483 : FAILURE in 34m 08s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/82c3061814de454e9331dd7c0b7faceb : FAILURE in 33m 23s","accounts_in_message":[],"_revision_number":3},{"id":"49d3f243a43cc56cf57b18430daf58dba06a770e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 21:08:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"33a4f2d00dac11ad192b642067a73e33c94becd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 21:14:10.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f2ec2d346616e855150d82677a8f33828c91f6c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 21:49:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a6924c9b989016e86dc94db886a9540cbfa6cf4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 22:21:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1ef5e940a5218d723f924fc0433088421bd4b0b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 23:01:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/38c398751ad945f4884df0abbf71346e : SUCCESS in 25m 19s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7b520110659f4af293126116002d3ef0 : SUCCESS in 22m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8bef329249f44999b98295158b09763b : SUCCESS in 23m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8c0b5cef637d443b88d021e6398f5ac6 : SUCCESS in 32m 11s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d3a48a9334394e60b2e879c4a58a87cd : SUCCESS in 24m 07s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ced876e7aadc4d3e973c2dbaa6ab42d6 : SUCCESS in 22m 17s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e33e1fb6404d47e39d5118f8aaf3b9f0 : SUCCESS in 19m 10s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/a1de8475f1f042ab82faaafc36c89ecb : SUCCESS in 23m 05s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ed57734bd532444f894b207fb79cff28 : SUCCESS in 31m 45s\n- loci-nova https://zuul.opendev.org/t/openstack/build/42248ef943f34893a38e1470a3fc8508 : SUCCESS in 31m 10s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/17d9cf87d09545418321091fbd3d65ff : SUCCESS in 24m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/72d48879066448339dd7f9482ac2f1d3 : SUCCESS in 17m 32s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/709ce71a52e948108307041e6e0a31d3 : SUCCESS in 21m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a97dbca3370944f681558b9dc23074cf : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":7},{"id":"de8c3fc799a1ee6f3cf9581a8eddc0cbb670c2a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 23:29:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"da9c947cb63c7405dba2152a7e67198974bb3c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 23:55:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/11ca04d98ae24c778b0ab9c930ea5df5 : FAILURE in 16m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bc4a3ddd855b432683c90a84b2cd4b9b : FAILURE in 16m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/68010715f1a64bdab88cbb420cc61f3e : FAILURE in 24m 16s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/837cf2b441f0447f85f218ff2eb363da : FAILURE in 16m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/87eafc0e438d4ede83e5597e52056fae : FAILURE in 16m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7b1624d7991949fcaf5360a60b8ad411 : FAILURE in 16m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8a67b34dde2f40339566e4fd1b24ddb0 : FAILURE in 16m 37s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/da6aea6622fd4ce5a9fcce67aadee4e1 : FAILURE in 16m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3a9fc04f03ba4a12a0297de9f356ee90 : FAILURE in 15m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3dd9b740ff9c407d90d3f387bff73197 : FAILURE in 16m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8bdb0e147bcf4fcf93d2e7fa1e9ce522 : FAILURE in 17m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9e9a87e9b88848aabdd8490a4cc9cf46 : FAILURE in 15m 39s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9b4cc83f92ff499da1d6a4258f2be96f : FAILURE in 15m 05s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/02a3bd43f16e44be9cab25f7a58b9dd2 : FAILURE in 17m 15s","accounts_in_message":[],"_revision_number":8},{"id":"9ade068532d9c640448ba9b9a0b394561a7debbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-13 00:30:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8028164a9586311d785ca5f6bf5354abc8986018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 01:34:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4e49d486303546ecb78e53e60284060d : SUCCESS in 53m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f81761a6852f4ff7958724458bb23502 : SUCCESS in 48m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/684ef8fdf70749aca8e440327e31071a : SUCCESS in 56m 22s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/be7c2fcd5dd24301a7845a10fea55f61 : SUCCESS in 53m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5e77dfa5d56e499ea7a29e4f069be07a : SUCCESS in 50m 02s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3602738490304dabb20dab3d876588c2 : SUCCESS in 51m 51s\n- loci-manila https://zuul.opendev.org/t/openstack/build/01558b7fc9364077944eab4325c50b88 : SUCCESS in 50m 33s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/a69b8055593c4678aa5690293603ea58 : SUCCESS in 45m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/49b9128af4c746938383aa17c7039131 : SUCCESS in 58m 37s\n- loci-nova https://zuul.opendev.org/t/openstack/build/618c503543614ec4b866e18c8bca2b72 : SUCCESS in 55m 30s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/572a50e198c04381aed36f59dc9c7d7a : SUCCESS in 47m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c8d7988f09d84991a0b375ada48dd991 : SUCCESS in 47m 30s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e2af5a25a59347fbb911dd5ceb6e030d : SUCCESS in 33m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7cb1ad73e9344d5caae2954f1c77f8ee : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":9},{"id":"735e03e91ffd3708c7ebce76dfada2055de3f3e2","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-10-13 14:36:02.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"569902e9dc35fe574776bc932b7380b624bdce10","author":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"date":"2021-10-13 15:26:11.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0e5531330537a5d6ab959c51eaaa7f56cb2b80b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:28:22.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4d6b4fc41449591c020e36802df2069acdc5aa38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:30:25.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2f442e7597a1477b81c5346ab83f1208 : SUCCESS in 1h 00m 33s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f4e9b57c4cb64278a31439e7e35634fc : SUCCESS in 48m 13s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7989d0d01aff4a949c24bab8adc81dc6 : SUCCESS in 38m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/09bacc07f5fe4f90a2509027e0c4246b : SUCCESS in 43m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/aa81b7cec2864cb68adc67d04e32e04f : SUCCESS in 48m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d8508ffec2344917bf432d21a3e5d518 : SUCCESS in 45m 30s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b3e1306f889c42a4a87450538bef1438 : SUCCESS in 49m 41s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/e53ed028ee7d43b397f701bc48a7f0ce : SUCCESS in 39m 00s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8b1bb2c7defb461dbf7cc1afc6f171fe : SUCCESS in 52m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2e8ebdd984a042f88a521623c8bbb06d : SUCCESS in 49m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ba03fc2c234e4989b460a507d52243f9 : SUCCESS in 48m 29s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0eaa8ec4a5d54e7d91ecc95f99077228 : SUCCESS in 38m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/abc830bdfcc547f69702174a15977fa4 : SUCCESS in 36m 03s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/cc4b9a53bdf94d06a4e18497f598b2a0 : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":9},{"id":"6f6d41ccac0f3eecba4ffd6deabc568fb2617ddb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:30:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"a61c73e9b224286a10cf927de7c58f0ac65db69a","revisions":{"89ad89dcf014c62b2dcd721655174a9a15a3c923":{"kind":"REWORK","_number":1,"created":"2021-10-12 05:21:05.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/1"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"subject":"Remove newly introduced setuptools from constraints","message":"Remove newly introduced setuptools from constraints\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/89ad89dcf014c62b2dcd721655174a9a15a3c923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/89ad89dcf014c62b2dcd721655174a9a15a3c923"}]},"branch":"refs/heads/master"},"f9e8c3288807b576abfc86c77e0ee7cfd001fe50":{"kind":"REWORK","_number":2,"created":"2021-10-12 05:24:07.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/2"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:24:07.000000000","tz":0},"subject":"Remove newly introduced setuptools from constraints","message":"Remove newly introduced setuptools from constraints\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9e8c3288807b576abfc86c77e0ee7cfd001fe50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9e8c3288807b576abfc86c77e0ee7cfd001fe50"}]},"branch":"refs/heads/master"},"dd04e2f2ea8c320827a0e737271a2e9f7c4dd7e0":{"kind":"REWORK","_number":3,"created":"2021-10-12 18:46:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/3"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 18:44:42.000000000","tz":-360},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported release [0][1]\n* Switch to centos 8 to run master [1][2]\n* Switch default base image to ubuntu:focal\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dd04e2f2ea8c320827a0e737271a2e9f7c4dd7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dd04e2f2ea8c320827a0e737271a2e9f7c4dd7e0"}]},"branch":"refs/heads/master"},"5db61e97cf01cf327cec4c458961f4a507cf28dd":{"kind":"REWORK","_number":4,"created":"2021-10-12 21:08:20.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/4"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:08:05.000000000","tz":-360},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported release [0][1]\n* Switch to centos 8 to run master [1][2]\n* Switch default base image to ubuntu:focal\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5db61e97cf01cf327cec4c458961f4a507cf28dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5db61e97cf01cf327cec4c458961f4a507cf28dd"}]},"branch":"refs/heads/master"},"22d0c6eb8b70af73554468dff0b034a694ea72ad":{"kind":"REWORK","_number":5,"created":"2021-10-12 21:14:10.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/5"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 21:14:07.000000000","tz":0},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported\nrelease [0][1].\n* Switch to centos 8 to run master [1][2] and disables related\njobs until [3] is resolved.\n* Switch default base image to ubuntu:focal.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n[3] https://bugs.centos.org/view.php?id\u003d18329\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/22d0c6eb8b70af73554468dff0b034a694ea72ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/22d0c6eb8b70af73554468dff0b034a694ea72ad"}]},"branch":"refs/heads/master"},"b65741036ef34fe91085f92c12f4486edc81c423":{"kind":"REWORK","_number":6,"created":"2021-10-12 21:49:35.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/6"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:48:28.000000000","tz":-360},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported\nrelease [0][1].\n* Switch to centos 8 to run master [1][2] and disables CentOS\nand Leap jobs until [3] is resolved.\n* Switch default base image to ubuntu:focal.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n[3] https://bugs.centos.org/view.php?id\u003d18329\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b65741036ef34fe91085f92c12f4486edc81c423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b65741036ef34fe91085f92c12f4486edc81c423"}]},"branch":"refs/heads/master"},"d090ae9567b347c85afb3e46abd2d444ecda9b0e":{"kind":"REWORK","_number":7,"created":"2021-10-12 22:21:22.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/7"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 22:21:17.000000000","tz":-360},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported\nrelease [0][1].\n* Switch to centos 8 to run master [1][2] and disables CentOS\nand Leap jobs until [3] is resolved.\n* Switch default base image to ubuntu:focal.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n[3] https://bugs.centos.org/view.php?id\u003d18329\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d090ae9567b347c85afb3e46abd2d444ecda9b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d090ae9567b347c85afb3e46abd2d444ecda9b0e"}]},"branch":"refs/heads/master"},"44ad8952f2dbb29c30b1369ae9e3d93918138ea9":{"kind":"REWORK","_number":8,"created":"2021-10-12 23:29:13.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/8"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 23:28:53.000000000","tz":-360},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported\nrelease [0][1].\n* Switch to centos 8 to run master [1][2].\n* Switch default base image to ubuntu:focal.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/44ad8952f2dbb29c30b1369ae9e3d93918138ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/44ad8952f2dbb29c30b1369ae9e3d93918138ea9"}]},"branch":"refs/heads/master"},"a61c73e9b224286a10cf927de7c58f0ac65db69a":{"kind":"REWORK","_number":9,"created":"2021-10-13 00:30:20.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/74/813574/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/74/813574/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/74/813574/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/74/813574/9"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 05:21:05.000000000","tz":0},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-13 00:30:09.000000000","tz":-360},"subject":"Update gates","message":"Update gates\n\n* Use bionic for ussuri build as it\u0027s the latest supported\nrelease [0][1].\n* Switch to centos 8 to run master [1][2].\n* Switch default base image to ubuntu:focal.\n\n[0] https://governance.openstack.org/tc/reference/runtimes/ussuri.html\n[1] https://governance.openstack.org/tc/reference/runtimes/victoria.html\n[2] https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I9a70aebd9636e170c6e04398356933668a2bfbba\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a61c73e9b224286a10cf927de7c58f0ac65db69a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a61c73e9b224286a10cf927de7c58f0ac65db69a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
