)]}'
{"id":"openstack%2Floci~813709","triplet_id":"openstack%2Floci~master~Iaa3f707e48dec32aadc2f8849196990915e36451","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"8863":{"account":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"last_update":"2021-10-13 16:30:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa3f707e48dec32aadc2f8849196990915e36451","subject":"Make py3 default","status":"MERGED","created":"2021-10-12 21:52:12.000000000","updated":"2021-10-13 16:30:27.000000000","submitted":"2021-10-13 16:30:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813709","meta_rev_id":"36f5ddcf04afe7e4ef0b2b27d2344169fc4c9122","_number":813709,"virtual_id_number":813709,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 16:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"all":[{"value":2,"date":"2021-10-13 15:23:09.000000000","_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":2,"date":"2021-10-13 14:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"all":[{"value":1,"date":"2021-10-13 15:23:09.000000000","_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 22:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 14:34:57.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2021-10-13 15:23:09.000000000","updated_by":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"reviewer":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"state":"REVIEWER"}],"messages":[{"id":"5c1fcb1a28f54f990ea36f1d683418107a715278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 21:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1652829fb5c6dbd991355a8f5d381b04ac7592ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 22:22:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bc3be9a06368c9354471aa294a26c9cfef663768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 22:57:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ee9cd3738fcc4bfc89cb26aec865a61a : SUCCESS in 24m 49s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cf1ea3e6a62343d4ae04283222c075b9 : SUCCESS in 24m 47s\n- loci-heat https://zuul.opendev.org/t/openstack/build/66902c42024c4f09995932ab4c11c671 : SUCCESS in 20m 28s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/af3ca3cfa6734d74b56f61baf25d4c97 : SUCCESS in 28m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b7d7836d16834c7db630f68c0947d754 : SUCCESS in 27m 47s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a4b8b39f7caf405ea9a4363bc0007c3a : SUCCESS in 27m 30s\n- loci-manila https://zuul.opendev.org/t/openstack/build/cbb49eef0fcb42f9b2d881971d66db8f : SUCCESS in 20m 37s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5a83d7468ae5437ebc95f7ed2f2be2b9 : SUCCESS in 19m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6cc5c963f5d24226afdd9b0b2a7da97b : SUCCESS in 23m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fc6b798f82684bb2930f3f69e743bea4 : SUCCESS in 21m 55s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/023eafc6d0c049e6af96341758c6d8a8 : SUCCESS in 23m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f6c0b93b1de747f2a77475f5d2b5de8b : SUCCESS in 23m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/05f78c8c8a324b80955bb6d3769bb353 : SUCCESS in 16m 44s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/08a444a223104f169f8eb2e43a526085 : SUCCESS in 22m 19s","accounts_in_message":[],"_revision_number":2},{"id":"615443dca1477253811f679778bb03ba518d26e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 23:30:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aecfc6d14974824fe1faab7a7c8c579228d2acef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 23:54:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1ae8f0ca6af3406fba940080953dbdbf : FAILURE in 16m 05s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a22286da38ed48e8b827f81dbeac2b47 : FAILURE in 17m 10s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ca0cfa5bfb784ab48eec4fb9be4af8d5 : FAILURE in 16m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/de7bf61ebf004170a1a6e1010e889f26 : FAILURE in 16m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3621c05e5b064272a9c7cfc4b70b140b : FAILURE in 15m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b0d1c8f7bb9c4aca8c96532499ea4815 : FAILURE in 16m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/af557ea39b8d4bcb82bcc9226fa163c6 : FAILURE in 16m 34s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6c5ba5e6199d4e67a7cf458412f69c91 : FAILURE in 15m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/58fff3a942644652aa5df041361c08a4 : FAILURE in 15m 47s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d3f075cc8d224c96965aa568e0c8e843 : FAILURE in 16m 06s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3eec4d0b09f3455f994ee7f19fd7125a : FAILURE in 15m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4fc1bce4cc5d49908d3bb5c73e6dcfbf : FAILURE in 15m 05s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/499386b6580149268b20478e69baa918 : FAILURE in 17m 11s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/214eb9665f1b468d8efcdd1d08f4bbb6 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":3},{"id":"cb33312a4134dfdf03c6d814753068b0899d855d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-13 00:30:23.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2d6119cd24c9584927ad0488d55a000dae65ce2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 01:22:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0611e677fd6d425ba37e85c8cb86efd5 : SUCCESS in 49m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1ce13146896441b29d74e3b05ffd004f : SUCCESS in 40m 01s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0ea52a2e06ba49fcbd099eef4f5bef7c : SUCCESS in 41m 30s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fac500f687124a95b86a2f216a421bc9 : SUCCESS in 43m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b0fa5c1138564ca9a6b10f065a39f81a : SUCCESS in 41m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/78e832b992ab4877bab344de2ff15cc8 : SUCCESS in 45m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f2712a8d0caa4d11ac3f1ca26be5f9e7 : SUCCESS in 43m 35s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/84a0347b6f0f4d91bcec4ee0f261d986 : SUCCESS in 42m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b83c991275404df8a31acd845275a2d1 : SUCCESS in 46m 54s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b21fae1fcf564adaac570862f5e20362 : SUCCESS in 48m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/855f59b1cdf84ed1b62de2a20b07dfe2 : SUCCESS in 41m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6ed0b298470f439188f648d74f463aaa : SUCCESS in 40m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/078db26ea8b44fa5b452c675a93fc16d : SUCCESS in 34m 58s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f711921140954597b5675a1dd5b78e64 : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":4},{"id":"4adb7fbdf4653bf989f570fcbc829cf64884c1fd","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-10-13 14:34:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d15022ee3eeff0a5fea4c0895d3d3eba0d66430d","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-10-13 15:23:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5f6d66307e5f6042cc8e4e525befe5f984abd826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:27:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ab8fba4c53107f299d8d0e1c3d94a3dcf63872a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:30:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/becc183a91cc42ad9dd85eda9a7462ff : SUCCESS in 42m 56s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4f19932c1fd54d84997be7d4d121d53d : SUCCESS in 43m 27s\n- loci-heat https://zuul.opendev.org/t/openstack/build/699c390446b64ba289917f3ac7cc4276 : SUCCESS in 46m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f2a6921da44d4557a5f49bf248c7deb1 : SUCCESS in 45m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/69546fdec31d4854852d675919ae4719 : SUCCESS in 44m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ef441d77a37042bcb45a344c86476d48 : SUCCESS in 45m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/814248b2c46848e0b7adc77134e7449c : SUCCESS in 47m 11s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5543993ed6f74a2ca6df378631f0d42e : SUCCESS in 45m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2bc2bec038f64b11aeae24a075967353 : SUCCESS in 47m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2d16c1302abf4fafa344cc7f587f50ac : SUCCESS in 50m 15s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/49314fde239148c683e1f1434758e375 : SUCCESS in 44m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cd5e1badd00d4025b0e2f5cd46b296c6 : SUCCESS in 39m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5cf433e00d2940398ce4791bd5ed3e8e : SUCCESS in 32m 10s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f55f5775b2e249b490cfed77bc33e89c : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":4},{"id":"36f5ddcf04afe7e4ef0b2b27d2344169fc4c9122","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 16:30:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","revisions":{"9d0202a7a1c55d8926aa757bfa3ec3292f93725b":{"kind":"REWORK","_number":1,"created":"2021-10-12 21:52:12.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/813709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/813709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/813709/1"}}},"commit":{"parents":[{"commit":"b65741036ef34fe91085f92c12f4486edc81c423","subject":"Update gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b65741036ef34fe91085f92c12f4486edc81c423"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:51:03.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:51:03.000000000","tz":-360},"subject":"Make py3 default","message":"Make py3 default\n\nChange-Id: Iaa3f707e48dec32aadc2f8849196990915e36451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d0202a7a1c55d8926aa757bfa3ec3292f93725b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d0202a7a1c55d8926aa757bfa3ec3292f93725b"}]},"branch":"refs/heads/master"},"c5a947f101d3be00e4fdc9538f258fa70dbe3bf3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-12 22:22:39.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/813709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/813709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/813709/2"}}},"commit":{"parents":[{"commit":"d090ae9567b347c85afb3e46abd2d444ecda9b0e","subject":"Update gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d090ae9567b347c85afb3e46abd2d444ecda9b0e"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:51:03.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 22:22:39.000000000","tz":0},"subject":"Make py3 default","message":"Make py3 default\n\nChange-Id: Iaa3f707e48dec32aadc2f8849196990915e36451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c5a947f101d3be00e4fdc9538f258fa70dbe3bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c5a947f101d3be00e4fdc9538f258fa70dbe3bf3"}]},"branch":"refs/heads/master","description":"Rebase"},"c6ee0fcb7483488dc9e687ff5af503eb452a2428":{"kind":"REWORK","_number":3,"created":"2021-10-12 23:30:33.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/813709/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/813709/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/813709/3"}}},"commit":{"parents":[{"commit":"44ad8952f2dbb29c30b1369ae9e3d93918138ea9","subject":"Update gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/44ad8952f2dbb29c30b1369ae9e3d93918138ea9"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:51:03.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 23:30:28.000000000","tz":-360},"subject":"Make py3 default","message":"Make py3 default\n\nChange-Id: Iaa3f707e48dec32aadc2f8849196990915e36451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c6ee0fcb7483488dc9e687ff5af503eb452a2428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c6ee0fcb7483488dc9e687ff5af503eb452a2428"}]},"branch":"refs/heads/master"},"a2a88b8ffbc85683bd5be71ebb9de55289800d7a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-10-13 00:30:23.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/813709/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/813709/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/813709/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/813709/4"}}},"commit":{"parents":[{"commit":"a61c73e9b224286a10cf927de7c58f0ac65db69a","subject":"Update gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a61c73e9b224286a10cf927de7c58f0ac65db69a"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 21:51:03.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-13 00:30:23.000000000","tz":0},"subject":"Make py3 default","message":"Make py3 default\n\nChange-Id: Iaa3f707e48dec32aadc2f8849196990915e36451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
