)]}'
{"id":"openstack%2Floci~815013","triplet_id":"openstack%2Floci~master~I8c522975144881972ebf3d394e0fa93f77fea6cd","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"8863":{"account":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"last_update":"2021-10-23 01:05:42.000000000","reason":"Andrii Ostapenko replied on the change"}},"hashtags":[],"change_id":"I8c522975144881972ebf3d394e0fa93f77fea6cd","subject":"Remove obsolete code","status":"ABANDONED","created":"2021-10-21 14:30:54.000000000","updated":"2022-07-13 13:19:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ecdc0d9b1f3dd8fe84b307c6095082ec07873184","_number":815013,"virtual_id_number":815013,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-23 00:45:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 15:13:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5bcbce84e422958d05c62f2019a551cd2819c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-21 14:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"314db8375f170dba28e23e80bbba080d14296940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:13:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/308ff4749e0c499c9e181940a3183bf8 : FAILURE in 39m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7f5846f0cb9945f690e004a022dcc2d8 : FAILURE in 32m 08s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a626c6849409424887cdbe8f72c74207 : FAILURE in 38m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c763c5f2364447fa87bf101e7a6b3830 : FAILURE in 37m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/074c9a896e5a4ee6baf4eeb012328e76 : FAILURE in 32m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7a42c95319774972bf7338c53c71c931 : FAILURE in 39m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7ffa66da7f6046b88d0c608d45266dc4 : FAILURE in 31m 58s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/36d20c0f14d9492c87c8f2926e058a75 : FAILURE in 33m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5344e774155d46b7aef34a6ff46d3bc6 : FAILURE in 33m 35s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a3aa6e04dfa84413874d5a027becea35 : FAILURE in 38m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e5d21e0a3d1e47738430289fce957731 : FAILURE in 33m 56s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2f78496407d248c5b0e01201a9e15610 : FAILURE in 30m 08s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6459864d1ea9487585da8bb6210e62b7 : FAILURE in 37m 57s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/beb26783bec644c78cee59aa605cd1d4 : FAILURE in 31m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f80d8fc24fc7e029e927e0a46fac2a5a739398ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-22 22:42:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f68ad660feea80969fc885865b600bbecc9d8723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-23 00:45:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0bc4f30d1534403fac6bf5ee2af439aa : FAILURE in 29m 49s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8ad2b9a3b6f54bf788fc3a286340f8db : FAILURE in 35m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2861a97b28ec49b1ba6009b70f568f1e : FAILURE in 53m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/30f23ed6d3454ac7ba427c0ea128864e : FAILURE in 53m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/65501306af3b4dfe82d54cb82c977e58 : FAILURE in 31m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bbcc270d7dc2453281c6e0c6f74865d1 : FAILURE in 30m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5a585ed0fc6746628dcc1d1f577351d9 : FAILURE in 30m 39s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/504ba769f7d144b9b9b26f5551f26144 : FAILURE in 53m 56s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d5dc3b70b362475ea4581d9974ce5f72 : FAILURE in 35m 33s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fa00f18e8835491f98db19d076382565 : FAILURE in 51m 01s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/62dd9da21cea47eca52c170398600cf8 : FAILURE in 28m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/823dc39ce3b849f79fddc06d10e5fa29 : FAILURE in 31m 21s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/38bc79af8c4c438d887fb6e640a91791 : FAILURE in 43m 58s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/09f53f8ab7d646dba10ec9c3866ee581 : FAILURE in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1be100e320ff03c4214c2a909d4de0f9c9ff27bd","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-23 01:05:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ecdc0d9b1f3dd8fe84b307c6095082ec07873184","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-13 13:19:20.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08ce7e1a408994c46fc3eec0f5bbae0532a43875","revisions":{"97adc2af151fbe8b0833f48474e5351411023cc6":{"kind":"REWORK","_number":1,"created":"2021-10-21 14:30:54.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/13/815013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/13/815013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/13/815013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/13/815013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/13/815013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/13/815013/1"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-21 14:30:05.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-21 14:30:05.000000000","tz":-360},"subject":"Remove obsolete code","message":"Remove obsolete code\n\nWith [0] merged this code can be removed.\n\n[0] https://review.opendev.org/c/openstack/requirements/+/813693\n\nChange-Id: I8c522975144881972ebf3d394e0fa93f77fea6cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97adc2af151fbe8b0833f48474e5351411023cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97adc2af151fbe8b0833f48474e5351411023cc6"}]},"branch":"refs/heads/master"},"08ce7e1a408994c46fc3eec0f5bbae0532a43875":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-22 22:42:06.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/13/815013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/13/815013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/13/815013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/13/815013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/13/815013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/13/815013/2"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-21 14:30:05.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-22 22:42:06.000000000","tz":0},"subject":"Remove obsolete code","message":"Remove obsolete code\n\nWith [0][1] merged this code can be removed.\n\n[0] https://review.opendev.org/c/openstack/requirements/+/813693\n[1] https://review.opendev.org/c/openstack/requirements/+/814934\n\nChange-Id: I8c522975144881972ebf3d394e0fa93f77fea6cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/08ce7e1a408994c46fc3eec0f5bbae0532a43875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/08ce7e1a408994c46fc3eec0f5bbae0532a43875"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
