)]}'
{"id":"openstack%2Floci~817704","triplet_id":"openstack%2Floci~master~If85687046297c582b39ca591b21e93c76b7de876","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2022-02-21 08:01:05.000000000","reason":"Oleksandr Kozachenko replied on the change"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-03-01 23:01:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If85687046297c582b39ca591b21e93c76b7de876","subject":"uwsgi: fix occasional failures","status":"MERGED","created":"2021-11-12 06:50:39.000000000","updated":"2022-03-01 23:01:56.000000000","submitted":"2022-03-01 23:01:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817704","meta_rev_id":"34916cf4362c793f5900cb3a0c14e88e7776fc58","_number":817704,"virtual_id_number":817704,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 23:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"recommended":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"all":[{"value":1,"date":"2022-03-01 23:01:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":2,"date":"2022-03-01 21:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-01 23:01:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":2,"date":"2022-03-01 23:01:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":1,"date":"2022-03-01 21:41:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 07:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 14:37:31.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2022-01-21 21:35:26.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-01 15:57:40.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2022-02-17 15:11:35.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"CC"},{"updated":"2022-02-21 08:01:05.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"REVIEWER"}],"messages":[{"id":"dda5211e182db54fc0a6a5f612523f971d3e6f76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-11-12 06:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b3cbf0d444506d2757f03d21bfcc1274a5bb560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 07:36:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/47b11da8fd0542c6b2a04f9318b25a41 : FAILURE in 41m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/77f2ad090fde4c2eb823ea951907fd6b : FAILURE in 36m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/058186ca87bd406ca1266d624eb190b2 : FAILURE in 41m 37s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b709bb98fb2944aabe1319bea01a5d2d : FAILURE in 36m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d97009528f074d4a97ad2ac5dbde19a8 : FAILURE in 27m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6f4439119cab437fa9ab226b43108a0d : FAILURE in 42m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/63b201f3173648a9844dd6d5549d86ac : FAILURE in 43m 02s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ac023046605e44ef9c16e83dacb4dc89 : FAILURE in 27m 10s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/62c768346a8f40ecb074afbda51bdc0e : FAILURE in 27m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8bb51ec83e824fec8c8effb1494c2947 : FAILURE in 26m 21s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a13d838ecbfc443fb8000c22bd739628 : FAILURE in 36m 38s\n- loci-placement https://zuul.opendev.org/t/openstack/build/02b22ef0168e4b42a8856e5f54597317 : FAILURE in 23m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ca041512ac0d4cefb47d466f79d2e084 : FAILURE in 26m 55s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6e83445b91c14234a5a472678f9779ae : FAILURE in 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"40590da783836849b1a88faf49ed758bada6cd8b","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2022-01-12 14:37:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54ec15a9172ab08e6cab47a48fd6b2b2b7ba7b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-18 08:36:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"23715da0ccd1f2127c808662b5c4cdff624e13dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 09:38:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1347e74d8c04439a8c5e7df7ccaa3a82 : SUCCESS in 33m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/73aa1c070f6942648f591b490ffd1284 : SUCCESS in 56m 04s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fb4478a320454ee59d7a072360b85e8d : SUCCESS in 33m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c66512eee172431f8714d5b944f3b7dd : SUCCESS in 33m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/49bdd5da6ee744f6a1ac6d7d52207e5b : SUCCESS in 35m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0fd7c256aba8426291c62da532ff13d6 : SUCCESS in 39m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a59e0a8f67524116a500bfc040067f82 : SUCCESS in 32m 38s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/79dc5ab3cb50406b85d80d7791b46c19 : SUCCESS in 31m 01s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/15b332ed9a0540a9a1ebba1fd7ec87df : SUCCESS in 33m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/17a15f70f57646f68415acb11c2777d5 : SUCCESS in 36m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0005de614be44d79b2e217e18add4b3c : SUCCESS in 28m 09s\n- loci-placement https://zuul.opendev.org/t/openstack/build/53df22423e324415ab32486f6627b0eb : SUCCESS in 24m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/82c1df9f2b7a4054bc7d090fbae59b3e : SUCCESS in 23m 49s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/588c184d2afc45119a3fb13fcf3986db : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"698661a1b9cfd780be6ad9b9201ea72729ba5004","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:35:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"295d6f15362750779f004d1f193c459ed7056a7d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-02-01 15:57:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b3f586808f93a96617ffa9d6e373783ca4b5bbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 15:58:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"010a913b1ef2c392751bc6c103d82572f2e7f387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:41:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f38a5d4393184d65b5749a2372aa441e : FAILURE in 5m 47s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b1da7d5f9951454596ac1fac9f5f51f4 : FAILURE in 4m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/67da125397ed49ad9679700bc67eb00d : FAILURE in 3m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/67f4ee5a221d48ae8914f8730ada17a7 : FAILURE in 4m 07s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f432a9bf11104c878870e593412a41b5 : FAILURE in 5m 28s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8c996ba710c441a6908fda80fd90ff16 : FAILURE in 4m 01s\n- loci-manila https://zuul.opendev.org/t/openstack/build/dc028ff0da1a463ca4fe01263706e349 : FAILURE in 3m 54s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d15d9f1f789c4cf188cc119ec56c8f83 : FAILURE in 5m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b5e94a1269a04a57847fd1c97156e9ca : FAILURE in 5m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7ce9690bec174213a4d3c08e7a307aa4 : FAILURE in 5m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3aa00228fc8e4f9d85f0d979e2ea1080 : FAILURE in 6m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0bffb0d19598472997f9e600408dc877 : FAILURE in 5m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4de9248550334c7e9892f3784b453eea : FAILURE in 4m 32s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/803346d71f474671a836e914ad29c081 : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1892a03eb34f61c81f709e414114d070b6d599c2","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-01 16:45:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"862811c7ae79db01b4cc77de1b89bab239ceb9bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:54:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0b7a02a5cd6e442b974be5cd3f61e379 : FAILURE in 4m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8accca5c685c4e2d84785e6e80cca1db : FAILURE in 5m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9de6e4fb6d924172ba46b919787cc85e : FAILURE in 5m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2b7363a862c24a6894dac1d6a8f07a14 : FAILURE in 4m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2c75b155ff1e49898e83f1a62fae89c8 : FAILURE in 3m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f5001ef843d242d1bdbf62fe3cc26382 : FAILURE in 4m 30s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0c2a1628aecb4e90ae6c4e12b07e6031 : FAILURE in 5m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6bdcd49d15b84187a132ddb81197b5c7 : FAILURE in 3m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a126748672504d1ea1f83e5b5ffdfc51 : FAILURE in 3m 51s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b55f1bef41304146a34e181ca518ba1c : FAILURE in 4m 16s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/862c7be765ac4262a763bf35150bb6f8 : FAILURE in 4m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b03aa56fc7c841d89be2a19e77f0e55d : FAILURE in 3m 38s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a63ba5adec974ccf9d5d09ab9021d73e : FAILURE in 4m 49s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d9f0fa4bb1254b02918eb8cafc8fbb44 : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a27d73cd1c3ca7a20d3e8cbac2c1f4f974323d39","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-02-17 15:11:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fca914f73fd9ab1bda70206080b5bbfe69ce75f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 15:24:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e53814018cb44b689dee0d6c083c4d5e : FAILURE in 4m 03s\n- loci-glance https://zuul.opendev.org/t/openstack/build/793f40533d484383b2e455406aa029f5 : FAILURE in 3m 49s\n- loci-heat https://zuul.opendev.org/t/openstack/build/25259e4cadd3428b8dfc62d3748f3673 : FAILURE in 3m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6632500042c64fc0bc3b433a0681b6a0 : FAILURE in 3m 57s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6eec228d7b0743daa14dd2ab40e8f738 : FAILURE in 4m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/060301dfe1a940c69bf8ebe9bd58d102 : FAILURE in 3m 33s\n- loci-manila https://zuul.opendev.org/t/openstack/build/82665efd823f4d168428389de9377182 : FAILURE in 4m 00s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2481ebc98ca7422d8a7e7325451915c3 : FAILURE in 3m 23s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c27151f68abf4c90ad4e1b63a2729f3a : FAILURE in 4m 19s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2e843c1516e54ba88bdeec3eb189baca : FAILURE in 3m 40s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2cbc533fd2df43dd81c94de0b56253f8 : FAILURE in 4m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1654d0b290144689a274d552bd5b61fd : FAILURE in 3m 30s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/21b40191ffb94f91ae417b0268a6df5e : FAILURE in 3m 54s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/680ae3177d494cc1804600dd39306b70 : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4a67f18d7c005275cb5ecd1541039864ba6e7d9b","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-02-21 08:01:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d14fde44bf3feb56da84a3d3aa450ee2986acfc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-01 21:41:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9963b3975a17b78ab2ca8996772d8001c77d228a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-01 21:41:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"28964a00ac5444fe13a62d35b181a1758a7227fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:23:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3326b129ec954cebbb13dd7526696c67 : SUCCESS in 28m 28s\n- loci-glance https://zuul.opendev.org/t/openstack/build/315f99d70a8d46b18302c32f3b1c70d9 : SUCCESS in 32m 01s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5c0f2e2587054858a8ce78a78a808863 : SUCCESS in 32m 46s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d8c639abd32e4a84a5e98d326c9e88a8 : SUCCESS in 33m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e50047ce7593462689c1e8bac88e8730 : SUCCESS in 30m 03s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b67de0cfebf34796a086f4301ae300eb : SUCCESS in 26m 13s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3e2a32ba4ae846569881bf677942fe09 : SUCCESS in 25m 56s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/bcb4beafdb92484182be7488711aebd6 : SUCCESS in 28m 35s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/11c76bd136ea4fb69e98efaf66018ad6 : SUCCESS in 28m 18s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fdf9e4f76fb440a0a3a8491d1d4373a6 : SUCCESS in 34m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b527c546467d449495b4c33470b1a41b : SUCCESS in 27m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4278392778d14bdc90041e4a0e70ffa6 : SUCCESS in 19m 08s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/78e1d909e4154935823656614d1b08e8 : SUCCESS in 27m 07s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3f66fd27b74e451e9e15a8346687c02f : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":3},{"id":"536af0602c44b395654f5759b30863983c55b196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:24:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"38c11106f1d1e12c4f33f7f3b80c7d9052669672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 23:01:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d78326180c554f058ba7116a7fe098f9 : SUCCESS in 26m 16s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e3059db6cd11456bb4f6f91c7c4444c7 : SUCCESS in 27m 50s\n- loci-heat https://zuul.opendev.org/t/openstack/build/006b8b9bcbf74fcca39c5a0f0c9477c7 : SUCCESS in 25m 37s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d3ea996aa9d047039dfd3a721479ec6d : SUCCESS in 27m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8c26769bdb184776ad856bbf898283ca : SUCCESS in 29m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4c06d9d6648743a0b0e048c1018f73ee : SUCCESS in 29m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6fa3ac7700e04537a6e657392658e7cc : SUCCESS in 27m 59s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/19aa880025574417834dd6dcc7bb41c1 : SUCCESS in 28m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b79d0d06c8dc4733b3674e0488ab560a : SUCCESS in 27m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f8d203c6183348ad86e8b575eccf5168 : SUCCESS in 35m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c65e6b5f65784b0783a7d16e094ee5ce : SUCCESS in 28m 31s\n- loci-placement https://zuul.opendev.org/t/openstack/build/721f69e987bb4861b5ecb2a78c16f98d : SUCCESS in 25m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/16499223f5654ea4bacef1f51a32dff6 : SUCCESS in 17m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a6cb7e93672a4383be3cddaf05bb81c4 : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":3},{"id":"b0c92d2af36bffb0c0e7eccc465bdc6f30a0b27a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 23:01:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0dd8eb61b62424e9ded329bce33bdc61815851f9","revisions":{"b9679398b2183840cce70f73bc77758e4f3ced6d":{"kind":"REWORK","_number":1,"created":"2021-11-12 06:50:39.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/04/817704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/04/817704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/04/817704/1"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:49:24.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:50:39.000000000","tz":-300},"subject":"uwsgi: fix occasional failures","message":"uwsgi: fix occasional failures\n\nWithout CPUCOUNT\u003d1, uwsgi frequently fails to build in\ncertain environments.  Once this was implemented, the\nbuilds were stable again.\n\nChange-Id: If85687046297c582b39ca591b21e93c76b7de876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b9679398b2183840cce70f73bc77758e4f3ced6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b9679398b2183840cce70f73bc77758e4f3ced6d"}]},"branch":"refs/heads/master"},"a924cc4b4d96239990770f2ee99bde014ce336fa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-18 08:36:54.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/04/817704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/04/817704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/04/817704/2"}}},"commit":{"parents":[{"commit":"efccd0a853879ac6af6066eda09792d0d3afe9c0","subject":"Specify a different source for novnc/spice-html5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:49:24.000000000","tz":-300},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-01-18 08:36:40.000000000","tz":660},"subject":"uwsgi: fix occasional failures","message":"uwsgi: fix occasional failures\n\nWithout CPUCOUNT\u003d1, uwsgi frequently fails to build in\ncertain environments.  Once this was implemented, the\nbuilds were stable again.\n\nChange-Id: If85687046297c582b39ca591b21e93c76b7de876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a924cc4b4d96239990770f2ee99bde014ce336fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a924cc4b4d96239990770f2ee99bde014ce336fa"}]},"branch":"refs/heads/master"},"0dd8eb61b62424e9ded329bce33bdc61815851f9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-01 21:41:36.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/04/817704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/04/817704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/817704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/04/817704/3"}}},"commit":{"parents":[{"commit":"c488166421e868c3b348208ad8af8a4922942f66","subject":"Add neutron vpn packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c488166421e868c3b348208ad8af8a4922942f66"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:49:24.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-03-01 21:41:36.000000000","tz":0},"subject":"uwsgi: fix occasional failures","message":"uwsgi: fix occasional failures\n\nWithout CPUCOUNT\u003d1, uwsgi frequently fails to build in\ncertain environments.  Once this was implemented, the\nbuilds were stable again.\n\nChange-Id: If85687046297c582b39ca591b21e93c76b7de876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0dd8eb61b62424e9ded329bce33bdc61815851f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0dd8eb61b62424e9ded329bce33bdc61815851f9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
