)]}'
{"id":"openstack%2Floci~817705","triplet_id":"openstack%2Floci~master~Ib3eac4269f7ac8365623f95a81b07a6436464ed0","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2022-03-01 22:23:57.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-03-01 22:23:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3eac4269f7ac8365623f95a81b07a6436464ed0","subject":"keystone: add mod_oauth2 to image","status":"MERGED","created":"2021-11-12 06:55:41.000000000","updated":"2022-03-01 22:23:57.000000000","submitted":"2022-03-01 22:23:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817705","meta_rev_id":"d90dcbbbe7b0d11555dbe1bb9da524980d922d90","_number":817705,"virtual_id_number":817705,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 22:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-03-01 22:23:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-01 21:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-01 21:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 07:36:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 12:54:47.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"CC"},{"updated":"2022-01-21 21:35:53.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-01 16:01:10.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"CC"},{"updated":"2022-02-22 08:37:55.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"f4f7ef3ea37881a8f6904afa27730bae7fd5bd4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-11-12 06:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d40cc4bd62562e4b6e5a61aca7bccd9c86f3853d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 07:36:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0b7db575083f462e8a9b6b818372d2c7 : FAILURE in 38m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/68169dd526d147ddacde8414b00f5084 : FAILURE in 30m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/61e83ec4dee34f4d9cdeb73182dfa82e : FAILURE in 33m 58s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e0ac16adfd39433ba77138337f7f944b : FAILURE in 34m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4c6ba247c47946c7b0243adb1ada6e20 : FAILURE in 26m 19s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5ed461f0ac8f42a885073b74cfafbc0d : FAILURE in 37m 22s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8057868fc7384e4f9206b8b74f473b8c : FAILURE in 35m 16s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/fa4370c17c4849e6a3de7928caedde11 : FAILURE in 29m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/29ed8251caf14143803b89e0bfe06bd9 : FAILURE in 32m 17s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0f79256af3924074b75e3457789299fb : FAILURE in 35m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8aa356d448764274b16a7b849463de4d : FAILURE in 33m 34s\n- loci-placement https://zuul.opendev.org/t/openstack/build/03a7d2ddb0994d8abb3023c2a2118fa0 : FAILURE in 31m 11s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7ef6a617c327405e91e515de7acad7af : FAILURE in 32m 15s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d27f3f76d00b4b0db2282613b474c519 : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":1},{"id":"aa0bafc49840ebe3962fe8b7beb4b3b8b75fbee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-18 08:38:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0aea9b5ce89d549277dffaebc53bbdd660c72e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 09:33:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/193369b897bd44ff9be93bfb9312e66e : SUCCESS in 35m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d164d715f9644cc5b60ebfa19b3b51b8 : SUCCESS in 33m 48s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e0b555f0c2e147ce9981f7f4a092685c : SUCCESS in 35m 34s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c438a0a3e2954f09914cc10082f43909 : SUCCESS in 35m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0f0c1efa5cba4816a90af3d76bdef0e2 : SUCCESS in 33m 01s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8c76f67aa9504e8d84fd43f467e53860 : FAILURE in 30m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5621fb00afd440c295a5632a89123e6f : SUCCESS in 33m 44s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/076447a875244745b2c2443e843e70d5 : SUCCESS in 30m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/401d19b23e1c446e8ea3f3ef919d111f : SUCCESS in 38m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/204be3fef6004346bb1d66eb691b9ef9 : SUCCESS in 45m 16s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9e81e2c4442a4241808b93f7d57f65e8 : SUCCESS in 32m 36s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d3d535b2dccc4254988dea229dbc1e62 : SUCCESS in 23m 13s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7a379ffabee146d2b7ec6e91e3e4129f : SUCCESS in 23m 09s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/48b8977bdd4a4003ac8880ec68a404fb : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4a81ab0e20896ea67de36fa913ae24eff8c7aaf3","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-18 12:54:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4962835a108ff0f8de402a7aea242b31e6ed604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 13:50:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9589da6223de4385b9fe16b2adffb15b : SUCCESS in 32m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/abc177a7ca24434fae0fcdc5cc84db8f : SUCCESS in 35m 03s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f2fdf99a359d4532a55c14f6e3381229 : SUCCESS in 48m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a03824c6496647a0b00b27159f1fe397 : SUCCESS in 43m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/eb4f9aa03fa4451fa90c25de6b66c0a7 : SUCCESS in 46m 21s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ee05cdd7f02943eba2d7b8af662889b3 : FAILURE in 44m 32s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2ef44f18c5104f33bee95a57c5a2e335 : SUCCESS in 35m 48s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/30bef55fa0a04ab292490829481505e5 : SUCCESS in 30m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8c69f9a1c023419ba8de6a2c7ec6d65c : SUCCESS in 41m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d29b86df84e148699b78fffb9aa06c93 : SUCCESS in 47m 21s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/94a7926e24f747ce817e816535f2ae1f : SUCCESS in 26m 02s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a5ee7326aac34713953082515c706c8c : SUCCESS in 24m 15s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/318898f6188645299091b5947332dd66 : SUCCESS in 31m 46s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bc4e309b624b44c6b892d68db143cc24 : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8670a5698741b49cd46a5377477bf3316a48626c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-18 14:41:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"825b06c2750f9841be3fd5b5ef5a02370591bb5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 15:42:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9db63eaf8a424381b1abe82e3d37fd65 : SUCCESS in 35m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a2500e4fbd99447db315fce2a7a91e48 : SUCCESS in 31m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/40a3e9a3fdee4bb2a3766a57aabd40b7 : SUCCESS in 38m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6ddedc087a684f338ff5590be81dd07e : SUCCESS in 30m 27s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2f7dc6053ef944ccb1b895f119cab94b : SUCCESS in 37m 10s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ffd98aa89633465e84d147141ec5c9da : SUCCESS in 40m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5736f77ecfbf436a84adcd7e1ebc9d3e : SUCCESS in 31m 49s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/e7fdb369c95946aa87f3cfd876227ce0 : SUCCESS in 36m 18s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e7d633c11b5c4ed4b8d692b2d55b2a77 : SUCCESS in 35m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6f86d191c39f4deb9f8e73d569740bd3 : SUCCESS in 35m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/bac8d49e9cf94e2daa4822174524284a : SUCCESS in 37m 41s\n- loci-placement https://zuul.opendev.org/t/openstack/build/47fa5053b1dc411398c194d130d2d3b2 : SUCCESS in 28m 24s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/46a9b3cd2a21436eb1c40d74c57b4fe8 : SUCCESS in 27m 47s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/520b08181b61412caa6f79ebac0ea8c5 : SUCCESS in 30m 48s","accounts_in_message":[],"_revision_number":3},{"id":"b27dd111bb3ee8b1de5940ffa684a51c1285aed1","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:35:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c828185d035889990f411354d0f00c287a9089a2","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-02-01 16:01:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ca01c10caca3f1ada5fbbe97370cd1fdb89154f","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-02-22 08:37:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d29f53174733cf5aee6836706b108faea98a0c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 08:38:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e765db71a485faffc4374bfd6d4137e731550f28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 08:45:46.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b5bb5047e60546b9b66bd84a2ce9fa38 : FAILURE in 5m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/541b8baf4c044787a1b4998499552587 : FAILURE in 5m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c22f3709015146bc93b24d0c1af6fa59 : FAILURE in 3m 42s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/800f9271280a481cab8794be45977411 : FAILURE in 3m 48s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/40a93a58065e402fb7bdd238bb66e791 : FAILURE in 3m 59s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9208f8bf361e4ff0b5ebda94e8655fc0 : FAILURE in 4m 13s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4c95799cb1e64d2eb1a5b26452e5d7f1 : FAILURE in 4m 01s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/08dcb1e0e5c54f6c81b8d409a1c7b9a1 : FAILURE in 4m 35s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/929c1a9f129e46a3ac5e617905e224c6 : FAILURE in 3m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/47d1ffbe34e347b6badebfc23f42f5bf : FAILURE in 4m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9a99bbb5469a42a9b8428ee703e0ca5c : FAILURE in 3m 49s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b24fe6357c9c4adcb167de5d7b5aacd5 : FAILURE in 3m 41s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/498119b6272e4559adbe46c7cedbcdf5 : FAILURE in 4m 35s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2a1857127ddc49729abf0b382e893c86 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"c00a33db3db41dc119f291dc8c1109f6074877b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-03-01 14:24:58.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d9fcd58092385ae826fa9e0b286893e66ad8174a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 15:04:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/33253e24da68412c9d6498eef91e671c : SUCCESS in 25m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b9721c57250e43af93753fdf9e1d82b7 : SUCCESS in 24m 34s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8a3f10044b914fa29a0a0614ad0c2a87 : SUCCESS in 24m 28s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9cbb1e81a8aa4bd68a2c1daf87a21cfd : SUCCESS in 25m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/09e5fdcb65c84d0d8679316113fb387e : SUCCESS in 24m 41s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e64a3f017f214ecdb79b76c41bc1791e : SUCCESS in 24m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/248dbc721f114ee7a4f02a07be041260 : SUCCESS in 24m 32s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b5f2812a49f64f038462a65f295e3f2e : SUCCESS in 23m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/43a154515572443daec19d2335160631 : SUCCESS in 25m 17s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8354462696b642f1bf6229f273f5fcbe : SUCCESS in 31m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/520bc10595fd45aaaa83a284c95e5bd6 : SUCCESS in 28m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/35626feebbc24491b51c52299716b5e3 : SUCCESS in 19m 48s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1d95da854d0b45a086b9f0b60bbdf67d : SUCCESS in 27m 44s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/80ce231a20a14cd093337202f7f45b71 : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":4},{"id":"50e4d5213b7c331069962c62046084539f9532ad","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-01 21:41:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3be9131f981b4986b95e407fe5f8e47fe1e0c55a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 21:42:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c74736c1eac2c9981a4dc09363464dc169eafdfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:23:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8c1fa99090eb4687808c091cb46ddca4 : SUCCESS in 31m 28s\n- loci-glance https://zuul.opendev.org/t/openstack/build/55fa22fc8a024d6192bf84ffc6a9e9ba : SUCCESS in 31m 41s\n- loci-heat https://zuul.opendev.org/t/openstack/build/be6cc2f8850b4a3a9e44a72f57d7cba0 : SUCCESS in 35m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f9edd678fc164dd2be99d2606bc5a3be : SUCCESS in 32m 59s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e8abfc00354c407ca86fbc5c1ba5da87 : SUCCESS in 31m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/01c1d8be77dc4b3abdb6abafdc29c339 : SUCCESS in 30m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a4bab0e131e14ec6860559551ba876a7 : SUCCESS in 36m 38s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ac537f6c2e0b4aa297728b089fd4882b : SUCCESS in 33m 19s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6a53c84d1cc64d2dba7616819fd320d5 : SUCCESS in 36m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dfd08081e07344fba2085f875b91d8b6 : SUCCESS in 33m 32s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/94888d41d7c94dcf8f5044522c3c20db : SUCCESS in 29m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f7dfd9978f7c4130a95ce14b1d8c0b7c : SUCCESS in 21m 25s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/06b9d9ead53d4832ae398c225d3cadba : SUCCESS in 24m 53s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6774cc1a9ce6402f8cecc40b53ee07a1 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":4},{"id":"6d617c25eb688b2418d399aaeb6c41ed1f307573","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:23:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cb0650d09818883dc0da68774bfe4760a55a4586","revisions":{"d20e877209c53422e7ee4e8a27a222323918b050":{"kind":"REWORK","_number":1,"created":"2021-11-12 06:55:41.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/05/817705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/817705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/817705/1"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:54:43.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:54:43.000000000","tz":-300},"subject":"keystone: add mod_oauth2 to image","message":"keystone: add mod_oauth2 to image\n\nmod_oauth2 is needed to enable oauth2 as mod_auth_oidc is\ndeprecating that feature. It\u0027s also not included by default\nin the repos so we have to script installing it.\n\nChange-Id: Ib3eac4269f7ac8365623f95a81b07a6436464ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d20e877209c53422e7ee4e8a27a222323918b050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d20e877209c53422e7ee4e8a27a222323918b050"}]},"branch":"refs/heads/master"},"55a4f3ed21a6224cd27a3b6ec8c28d43f136cfcf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-18 08:38:33.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/817705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/817705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/817705/2"}}},"commit":{"parents":[{"commit":"efccd0a853879ac6af6066eda09792d0d3afe9c0","subject":"Specify a different source for novnc/spice-html5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:54:43.000000000","tz":-300},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-01-18 08:38:22.000000000","tz":660},"subject":"keystone: add mod_oauth2 to image","message":"keystone: add mod_oauth2 to image\n\nmod_oauth2 is needed to enable oauth2 as mod_auth_oidc is\ndeprecating that feature. It\u0027s also not included by default\nin the repos so we have to script installing it.\n\nChange-Id: Ib3eac4269f7ac8365623f95a81b07a6436464ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/55a4f3ed21a6224cd27a3b6ec8c28d43f136cfcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/55a4f3ed21a6224cd27a3b6ec8c28d43f136cfcf"}]},"branch":"refs/heads/master"},"5d8e4fe06d762dc4364c166059f4e0e4f7ffb3c2":{"kind":"REWORK","_number":3,"created":"2022-01-18 14:41:34.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/817705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/817705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/817705/3"}}},"commit":{"parents":[{"commit":"efccd0a853879ac6af6066eda09792d0d3afe9c0","subject":"Specify a different source for novnc/spice-html5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:54:43.000000000","tz":-300},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-01-18 14:41:13.000000000","tz":660},"subject":"keystone: add mod_oauth2 to image","message":"keystone: add mod_oauth2 to image\n\nmod_oauth2 is needed to enable oauth2 as mod_auth_oidc is\ndeprecating that feature. It\u0027s also not included by default\nin the repos so we have to script installing it.\n\nChange-Id: Ib3eac4269f7ac8365623f95a81b07a6436464ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d8e4fe06d762dc4364c166059f4e0e4f7ffb3c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d8e4fe06d762dc4364c166059f4e0e4f7ffb3c2"}]},"branch":"refs/heads/master"},"cb0650d09818883dc0da68774bfe4760a55a4586":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-01 14:24:58.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/817705/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/817705/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/817705/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/817705/4"}}},"commit":{"parents":[{"commit":"dee49b0ed09d906d5597e7359e9bf70aee494ec3","subject":"Disable centos-8 builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dee49b0ed09d906d5597e7359e9bf70aee494ec3"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2021-11-12 06:54:43.000000000","tz":-300},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-03-01 14:24:31.000000000","tz":660},"subject":"keystone: add mod_oauth2 to image","message":"keystone: add mod_oauth2 to image\n\nmod_oauth2 is needed to enable oauth2 as mod_auth_oidc is\ndeprecating that feature. It\u0027s also not included by default\nin the repos so we have to script installing it.\n\nChange-Id: Ib3eac4269f7ac8365623f95a81b07a6436464ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cb0650d09818883dc0da68774bfe4760a55a4586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cb0650d09818883dc0da68774bfe4760a55a4586"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
