)]}'
{"id":"openstack%2Floci~821612","triplet_id":"openstack%2Floci~master~I2e272ce1ed971509e79b3518f5493e68ec8d6e63","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2022-03-01 17:20:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e272ce1ed971509e79b3518f5493e68ec8d6e63","subject":"Add neutron vpn packages","status":"MERGED","created":"2021-12-13 18:40:11.000000000","updated":"2022-03-01 17:20:15.000000000","submitted":"2022-03-01 17:20:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821612","meta_rev_id":"717b6042e78b245b4c1bfe64dfbdfe794400d4c9","_number":821612,"virtual_id_number":821612,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 17:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-03-01 16:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2022-03-01 17:20:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2022-03-01 17:20:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-03-01 16:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 19:21:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-14 07:35:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2022-01-21 21:48:20.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-01 16:05:29.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2022-03-01 16:39:22.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"09fe0d2bf87e0fb96114756f23b211a2fd1ea02b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-12-13 18:40:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58db04e7c57024ce6323a0a581d1400712fef5df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 19:21:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8383af03aedb41aab7d1c627f879fb87 : FAILURE in 25m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/5a6c8c174c8541d4b286f8495e322b6f : FAILURE in 31m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a27bc7a5879b4162b5d2474ca9eb1f65 : FAILURE in 30m 04s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b9e184bc47f74e5584f42dc9443e6ba6 : FAILURE in 30m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e852e6adccb341adbf66870348373e5a : FAILURE in 31m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0b3504f6e7a84135b32d04462d320474 : FAILURE in 34m 05s\n- loci-manila https://zuul.opendev.org/t/openstack/build/00baa887c3974caab16e8b09c7829d81 : FAILURE in 31m 38s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2510979432e443c78869ff3ef414bdd4 : FAILURE in 38m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4d1b0783ea6b4e219a339f695cb3d491 : FAILURE in 29m 13s\n- loci-nova https://zuul.opendev.org/t/openstack/build/772c4231021f471ca29c352d6ec1a3c6 : FAILURE in 22m 44s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/453f18dc64224abbab5049aece740025 : FAILURE in 26m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/dd4a5fb66bb046d7a999fd479c3dd70e : FAILURE in 28m 35s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8edf5a553b2d4e509c9d75e0f7c359a1 : FAILURE in 32m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5d46752af0af4f86afafae429fb31dc9 : FAILURE in 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1e8e62a2c6fbe90ab4e40271fab0fcdd3bcb663c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-12-14 07:35:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d02d3bd37917ae7d9f912488803d753ab04be89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2021-12-14 09:37:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"409db103d54406419d76275556fec8c18cd1d7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 10:06:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ea5ae134d12d4edf9975f7fbcd8ce897 : FAILURE in 21m 01s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e30ef2a1bbaa4382a1d15d86b3274b3d : FAILURE in 25m 58s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8a584364233a4108b7bd45576c91a361 : FAILURE in 25m 13s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ca4ef40b9ed74d67bb09bb0999769d4e : FAILURE in 21m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/699b868b9d08493881351ab942cfa29d : FAILURE in 20m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/63ba54cd0ce14466a5ecdb09a31b81a4 : FAILURE in 25m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6bab91d661de4621948298b3f4efcdcf : FAILURE in 21m 24s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/dc4fce126e014cee93ecd76240694b7b : FAILURE in 25m 55s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4746b8cf08bc4210949d51db5039f133 : FAILURE in 24m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/39b76610645749919c404d5df64375b6 : FAILURE in 21m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5b02cab1b9f646a2a37c968ab144d6ba : FAILURE in 20m 57s\n- loci-placement https://zuul.opendev.org/t/openstack/build/32596454113b438a91a8925441dc22f7 : FAILURE in 19m 48s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4c0d8c1421b14aeaa6a6e2dd8ac576b0 : FAILURE in 21m 38s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/4ae2f5e308c143d29e2c479b714eba9e : FAILURE in 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f70957cca4ee1001500913d6e9d3c2f478ef6b72","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-10 08:45:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67bd3383cf153c7d7a592991dcba9b82d8749f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:14:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9cdfde4fdc5c4fe2aac3aaa69ff51a13 : FAILURE in 16m 23s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bd5752dd290c43f68cc6a952da1b1007 : FAILURE in 22m 11s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4a1fc198d4ce437ab03ddde473096fb0 : FAILURE in 20m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/18a5fcafe71a43b49fe4cd07a11b7a40 : FAILURE in 18m 39s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/10c8b8ff7bc34a528bc834e234f250a5 : FAILURE in 19m 45s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7f4b772641124b5eaf49f5a628c33820 : FAILURE in 16m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f10bd1b6621f44768a0722a4200f8d78 : FAILURE in 17m 52s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f223089e4fd04faea894f7ab08a1771d : FAILURE in 15m 42s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/be591aabdc1641f0b7fd1d72585193ac : FAILURE in 15m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d81477cd52f249fe95ad5d7b6ddb5541 : FAILURE in 20m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fdbeafe032ae4174b975773ddcd5075d : FAILURE in 15m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/025730f37219454ca1c6028639934023 : FAILURE in 13m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1cff5f095a7b4a3ca613a204ee1a1060 : FAILURE in 17m 25s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5e4018418c434718a55603faa711453d : FAILURE in 18m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4392c383000190e109e324c0a325c90bbba4bf71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-18 08:30:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1d8ddf8cf3f5a46ce5d83478569ef4d751a66237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 09:27:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/55d4de7555be4fdba7bc71fde436a5ac : SUCCESS in 37m 48s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ada09e1426af460cabb4b0bb37214a71 : SUCCESS in 37m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/01cffb10f46e4fada78d7e5c1d047a78 : SUCCESS in 48m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/078fd48a5945491f878102f8a0b9db4b : SUCCESS in 49m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e591caaee7124888806cc56cc218781f : SUCCESS in 49m 22s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/be0565f9ca884c1a8d552473b84a71a5 : SUCCESS in 48m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/45d0e1bd3ad74c6e8adeb7e0d8389538 : SUCCESS in 35m 01s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b80aa2b72523433aa7b28067387bd5f8 : SUCCESS in 43m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/72f526899bff4305a70c5a0b7dab2a99 : SUCCESS in 45m 18s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d10bf366cb724cc5b9a0357b52e9393d : SUCCESS in 36m 17s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7fc08a0630264519bdef4389e7cddee1 : SUCCESS in 47m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c1e1998660af42b1b663d5c473c05e62 : SUCCESS in 42m 05s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/64ac33eca4d94595a753a8fcad13766c : SUCCESS in 26m 26s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1965ce25b04f47dea0f49329aac0bce2 : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6f9dc7a3110ca86c5a0d5c5bda0054c3d172bc35","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:48:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"08037b6959309c113a2d6a31479c34a1042c083c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-02-01 16:05:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39d1f1af8da828b59cc58cc24c09a9f9f09f3045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:12:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4b2174783e3932cfdc36340378a18001b8f2a7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:30:16.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/eeda5fd66503487fac0623619b0e0208 : FAILURE in 3m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/56be8cd788f6407b9f7620840d57c80c : FAILURE in 4m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fc970a7be42d4b6b8393907263bda8d1 : FAILURE in 4m 34s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/54857af4c0b84748a29e95953da6fed0 : FAILURE in 4m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/55a34579be714fb99adae52e819fb3c3 : FAILURE in 4m 24s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c54977e5f0764f619fafbfdb0a50acb2 : FAILURE in 4m 23s\n- loci-manila https://zuul.opendev.org/t/openstack/build/31301dd2ef9c4cc8b0443a48acd356af : FAILURE in 3m 55s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/97355bc33788472a9925ef785b6dcf8e : FAILURE in 3m 47s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/13a4af62b2194bd495df8bbae06eec1e : FAILURE in 4m 07s\n- loci-nova https://zuul.opendev.org/t/openstack/build/75197222121b4e6ab4c6ebdd47cba2e1 : FAILURE in 4m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4110f9c311f64cad8c45e8fc2718d73e : FAILURE in 4m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ae060b4ddf2346458681cb49141ed00f : FAILURE in 4m 16s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/58fa4dcd8d2c486186f38ded9dda4548 : FAILURE in 4m 18s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d903806ce0534beb97d9d271261462d6 : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"c426d3a8aa2dd9688e1473c4120ecda293ccf6e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-03-01 14:27:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"849177176204886e5c91c000ffa75cb6b86ffd74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 15:07:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8f65599ad59b409e943f9817f2f671a2 : SUCCESS in 22m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/be52948a40d34624ad3207d0dad7422d : SUCCESS in 33m 58s\n- loci-heat https://zuul.opendev.org/t/openstack/build/88b373d32e5e489d8d2bbe5c168fab4a : SUCCESS in 36m 24s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/79271a8f40be4047bcf90480404a7eb7 : SUCCESS in 30m 41s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/76a0cbee4ee94cf79d37dbb786ce0e0e : SUCCESS in 31m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b1192b07df0440659a946ce8856d2997 : SUCCESS in 23m 50s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ed4003b0944c4b1dad4adde0b59fb1f7 : SUCCESS in 30m 58s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5831821ee62647f9a3e6d90ff3afbc1b : SUCCESS in 23m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f9c67a6d7bf7492e8e54bb3498fc8ef6 : SUCCESS in 25m 17s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3c517fab07694fe19cccdea8a376fcc4 : SUCCESS in 27m 50s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f6ba38fd2c914dfe899093f51e2b084f : SUCCESS in 25m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bfeb5a024c10407e97b855276539f102 : SUCCESS in 24m 46s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b4424a80280a411f8882a843691c75c2 : SUCCESS in 25m 22s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d671423408e643b48615cc5455381d3e : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":4},{"id":"395414913c7efcdb9424070ac716edeb9760bd95","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-03-01 16:39:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"56f143005b156cf90780e00070a61f3536cde5d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 16:40:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4f2ed8a80220acef7a57e4edddb64512037d1e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:20:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7deff4c4619f43e7bbdcbbf02ff8f28d : SUCCESS in 28m 07s\n- loci-glance https://zuul.opendev.org/t/openstack/build/26f185c6174040dba951de8278ecadf8 : SUCCESS in 24m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/52551a05742945209f1f7f864b4a12b7 : SUCCESS in 25m 19s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5f9b983e270b4fea8a33291cede90bfb : SUCCESS in 27m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f033e1ad405e470f9d791e168201f4de : SUCCESS in 27m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f9bb218b0a734fb7922f5af6324ed61b : SUCCESS in 29m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/105913924e5a42d5834d5a04d0e118d1 : SUCCESS in 25m 10s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6de3eaf54b2c43cb857405f47ef55166 : SUCCESS in 30m 47s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b84152e3aa9142869a27c218350ab04f : SUCCESS in 27m 54s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9059ba31c27846468ca377f46b7c9249 : SUCCESS in 35m 24s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a89417d16f7d4ca88144685980dfb8f5 : SUCCESS in 27m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e00263dff2b040edbc8a18424e7cd7ff : SUCCESS in 20m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7b894ce2e4ba4f7bbdffb4d69b7b6f43 : SUCCESS in 23m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/56686a8338c84e2392874f616954a03e : SUCCESS in 26m 41s","accounts_in_message":[],"_revision_number":4},{"id":"a0621842d1b0c1b85ef42f70e1e5fdb80fff05fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:20:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c488166421e868c3b348208ad8af8a4922942f66","revisions":{"927571406c786f77a0e35eeb9dd362530139bc39":{"kind":"REWORK","_number":1,"created":"2021-12-13 18:40:11.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/12/821612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/12/821612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/12/821612/1"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2021-12-13 18:39:44.000000000","tz":60},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2021-12-13 18:40:07.000000000","tz":60},"subject":"Add neutron vpn packages","message":"Add neutron vpn packages\n\nChange-Id: I2e272ce1ed971509e79b3518f5493e68ec8d6e63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/927571406c786f77a0e35eeb9dd362530139bc39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/927571406c786f77a0e35eeb9dd362530139bc39"}]},"branch":"refs/heads/master"},"118fc2e0efc25025aedfd7d7a7fe2d141a804194":{"kind":"REWORK","_number":2,"created":"2021-12-14 09:37:42.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/12/821612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/12/821612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/12/821612/2"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2021-12-13 18:39:44.000000000","tz":60},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2021-12-14 09:37:31.000000000","tz":60},"subject":"Add neutron vpn packages","message":"Add neutron vpn packages\n\nChange-Id: I2e272ce1ed971509e79b3518f5493e68ec8d6e63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/118fc2e0efc25025aedfd7d7a7fe2d141a804194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/118fc2e0efc25025aedfd7d7a7fe2d141a804194"}]},"branch":"refs/heads/master"},"9e7f105bd3e9682924656c2919ecf01ba912786c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-18 08:30:10.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/12/821612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/12/821612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/12/821612/3"}}},"commit":{"parents":[{"commit":"efccd0a853879ac6af6066eda09792d0d3afe9c0","subject":"Specify a different source for novnc/spice-html5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2021-12-13 18:39:44.000000000","tz":60},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-01-18 08:29:51.000000000","tz":660},"subject":"Add neutron vpn packages","message":"Add neutron vpn packages\n\nChange-Id: I2e272ce1ed971509e79b3518f5493e68ec8d6e63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9e7f105bd3e9682924656c2919ecf01ba912786c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9e7f105bd3e9682924656c2919ecf01ba912786c"}]},"branch":"refs/heads/master"},"c488166421e868c3b348208ad8af8a4922942f66":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-01 14:27:07.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/12/821612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/12/821612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/821612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/12/821612/4"}}},"commit":{"parents":[{"commit":"dee49b0ed09d906d5597e7359e9bf70aee494ec3","subject":"Disable centos-8 builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dee49b0ed09d906d5597e7359e9bf70aee494ec3"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2021-12-13 18:39:44.000000000","tz":60},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-03-01 14:26:46.000000000","tz":660},"subject":"Add neutron vpn packages","message":"Add neutron vpn packages\n\nChange-Id: I2e272ce1ed971509e79b3518f5493e68ec8d6e63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c488166421e868c3b348208ad8af8a4922942f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c488166421e868c3b348208ad8af8a4922942f66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
