)]}'
{"id":"openstack%2Floci~823577","triplet_id":"openstack%2Floci~master~I3001dad62a0caed26347dcb193d101c80901b5d8","project":"openstack/loci","branch":"master","topic":"bug/1956173","attention_set":{},"removed_from_attention_set":{"32349":{"account":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"last_update":"2022-01-17 14:05:05.000000000","reason":"Change was submitted"},"33519":{"account":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"last_update":"2022-01-17 13:18:41.000000000","reason":"Tin Lam replied on the change"}},"hashtags":[],"change_id":"I3001dad62a0caed26347dcb193d101c80901b5d8","subject":"Specify a different source for novnc/spice-html5","status":"MERGED","created":"2022-01-05 20:59:49.000000000","updated":"2022-01-17 14:05:05.000000000","submitted":"2022-01-17 14:05:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823577-bug/1956173","meta_rev_id":"0b86d75ab5b7397ebb6245ba535b7273d398aee8","_number":823577,"virtual_id_number":823577,"owner":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-17 14:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"recommended":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"all":[{"value":1,"date":"2022-01-17 14:05:05.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-14 22:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2022-01-17 13:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":1,"date":"2022-01-17 13:18:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 21:24:50.000000000","updated_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2022-01-05 21:28:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 22:06:10.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-01-17 13:16:51.000000000","updated_by":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"reviewer":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"state":"REVIEWER"}],"messages":[{"id":"553bb827a374b27bc0f3a1e1b1bef12b1e89684f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2022-01-05 20:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e509969d0adca483628b063cf60068125dae5247","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2022-01-05 21:24:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf195f8d7e0a0eb58d8ef96ef60669ff332f4b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 21:28:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/028e0a819abd4f928b2097fb12c541ce : FAILURE in 18m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9b4821e1a0214181871082ec4f991934 : FAILURE in 17m 08s\n- loci-heat https://zuul.opendev.org/t/openstack/build/cf75c6b76085439b8858eaca1de8327c : FAILURE in 19m 02s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c596ec69fb1c471e9a09f2745c783f4b : FAILURE in 17m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/25c660af46b74e478bdfb1d24d5dac2f : FAILURE in 16m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6e2e234fc6114522b4eac5d8fd85b58b : FAILURE in 18m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1d72955cdfea4c9f86ff6fc1d2c5d818 : FAILURE in 22m 22s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/38651072f1294355b79fbc0215d8afac : FAILURE in 18m 06s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9f26d9137d3f4a1ba40c0ca245553532 : FAILURE in 18m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/62057b1bb4d342c3833822603ae278ec : FAILURE in 16m 42s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/dfda281179a14d2c9591454795913832 : FAILURE in 18m 11s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c28bd877c5ce4261805b50c0d529e0e2 : FAILURE in 13m 32s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/977e8d779f284aada16188abbfb23ddc : FAILURE in 21m 38s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bb7663fa6fa040c3a051257970724397 : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6195a3b8e938f33ae80d9c69d9da9d85a2d4e4d0","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2022-01-05 23:29:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ded510b854263b5f25febf12d6721d2b87797641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 23:52:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1e02c398ba9f403cbd19580744ab1ef1 : FAILURE in 15m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1a4b1c67e8f64ae29ac53021efdeee1a : FAILURE in 17m 31s\n- loci-heat https://zuul.opendev.org/t/openstack/build/12d2e3c33a624890972a4040ce53851a : FAILURE in 15m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f87a463dce744e1da26bc9e2b4fde6b5 : FAILURE in 14m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2567d0bf52e84841a3a23d01a0a4e781 : FAILURE in 18m 36s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2dbe6bb23b5e45bfa09b7e60f463e0b8 : FAILURE in 15m 31s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b09b75679e544d198343c6607337a9c2 : FAILURE in 17m 40s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/198851aa946043aba9190ee3f3dee926 : FAILURE in 17m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/eb87d9b4b1384e389af7168aea48dd85 : FAILURE in 16m 15s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7caa830761fa4b6293b5bfb99fe4dd4b : FAILURE in 15m 26s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/018a71715bf64eaf88ce00baffb675b5 : FAILURE in 15m 55s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0959b77052504aa68b1b757e7877767b : FAILURE in 11m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/876da41fac344b4993f535846d40ea43 : FAILURE in 15m 33s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7ef4891615e24c42b2ba0e6ae7e3c7d2 : FAILURE in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"35cb09e2d5a1535ab4b6b35bb3a5ff96147b088a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2022-01-10 21:45:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b104ab3ad788dd498148769d773be0408e160e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 23:21:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f6f041a57f8f4f81ae395971dc256ac6 : SUCCESS in 30m 38s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4ed6acff35704b4eaa21a01886711a7e : SUCCESS in 28m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/362cd4a66ce14404912f26e8bca7c30b : SUCCESS in 28m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/51db0427593640f1a15402cd3b6e38e3 : SUCCESS in 29m 38s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f784860e70d5417c88d9cef93fa9ceb9 : SUCCESS in 28m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7a2198202e7046899d312d1a40709eda : SUCCESS in 29m 17s\n- loci-manila https://zuul.opendev.org/t/openstack/build/056437982941460e96cd4d270a66e0f4 : SUCCESS in 29m 25s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/65e47f781c744160931f102112670d30 : SUCCESS in 32m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/711024b3ab904218b57ca84bb6a78ea6 : SUCCESS in 29m 38s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6be25655a8ff4b73a0e45d5aeb690115 : TIMED_OUT in 1h 31m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a160b8c7c785480c87ed41b424dae01a : SUCCESS in 29m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a9c6e60ac9564735ba05ce6a4ca08823 : SUCCESS in 26m 26s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/53c7f3dab17048c59d8da73b7bc0f8c5 : SUCCESS in 22m 42s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/09e1f397a33f44b2a3b5ce9e1ffa1727 : FAILURE in 31m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c2062c9b8ec0ce054198780b18920b506a7bf34b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"date":"2022-01-12 14:45:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d2926bed17873f9c8619524c4fc73b166a276ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:42:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2dfaf6c4d7114e9f8fa103d925b07235 : SUCCESS in 35m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4701ec3902c04ea7bb5caef050db8e41 : SUCCESS in 33m 13s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6255e80f24ca48e898802527e86e0797 : SUCCESS in 30m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4bd60bb58d764452a414eac8d2231694 : SUCCESS in 34m 36s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2e35f90cc71546e0b15a427daaa8d899 : SUCCESS in 33m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b1fe132425ca44dea13c209ae1282822 : SUCCESS in 32m 10s\n- loci-manila https://zuul.opendev.org/t/openstack/build/97254df380854d89a05d6835b87ff3a2 : SUCCESS in 32m 08s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d8c78ff3cb784abaa62be5a8bc1606b1 : SUCCESS in 38m 46s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e091f181a02242278d51abbfc16770f6 : SUCCESS in 34m 35s\n- loci-nova https://zuul.opendev.org/t/openstack/build/39db7a691c1f4b38aae3641334267162 : SUCCESS in 36m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d8d930a3fe3345f285b4a969ce9e8a43 : SUCCESS in 33m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/22c7da5ca45a47b78a4f6c2d4cb07df7 : SUCCESS in 26m 33s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/047f10cc28194d30a7dd107d78f06c65 : SUCCESS in 23m 58s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/e95952d8e66745aeac6795be9194e812 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":3},{"id":"5b4866830fee0cdbf98c63d5debfd9c5527b1464","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-14 22:06:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"058d3af4f5450fb9dfd5d0da83554bf5847b2048","author":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"date":"2022-01-17 13:18:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7250fafe8326d7a7fce81dc315fd5032b31541b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 13:19:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06ed85ea95e07407344c0f6f119d2b0943babed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:05:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a0456cf1679b48d38832cc9f5e811f17 : SUCCESS in 33m 38s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6db8f63084a246d58083bb8d1a154a6c : SUCCESS in 31m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/728c2ecc06c5425fb5ed0f452b0b77aa : SUCCESS in 29m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6d37c607303e48609cf74287e9ebfb3f : SUCCESS in 34m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a9cd7ab68c3a412eb369fdd7810854c9 : SUCCESS in 33m 16s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d4533aad648f403d88252868acf1e097 : SUCCESS in 27m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b0edd45f07fe463cab6f0a2bb8c7abd9 : SUCCESS in 38m 30s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/368de7ef6cad4439bc403a63e585a398 : SUCCESS in 26m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ad6ab61b80d6419692c0b60c7452839c : SUCCESS in 39m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/095a2d011ee34e63b9d654bd6a8b1b71 : SUCCESS in 41m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1cd433623d83496c9c2f99719edc8273 : SUCCESS in 38m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/143f31c964ed47308053377723e29e7d : SUCCESS in 27m 05s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c62ecacc923946219b202a3897142645 : SUCCESS in 22m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/437b20a3ba6946be93cb9be4be02c9f8 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c9e100e5057fd088ea4f8eafdfa5f2409cfa21c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:05:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"efccd0a853879ac6af6066eda09792d0d3afe9c0","revisions":{"a5ba0464ced1cde159bfd14ea258e0787ba9f374":{"kind":"REWORK","_number":1,"created":"2022-01-05 20:59:49.000000000","uploader":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"ref":"refs/changes/77/823577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/77/823577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/77/823577/1"}}},"commit":{"parents":[{"commit":"a2a88b8ffbc85683bd5be71ebb9de55289800d7a","subject":"Make py3 default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2a88b8ffbc85683bd5be71ebb9de55289800d7a"}]}],"author":{"name":"Thiago Brito","email":"thiago.brito@windriver.com","date":"2022-01-05 20:55:24.000000000","tz":-180},"committer":{"name":"Thiago Brito","email":"thiago.brito@windriver.com","date":"2022-01-05 20:59:47.000000000","tz":-180},"subject":"Specify a different source for novnc/spice-html5","message":"Specify a different source for novnc/spice-html5\n\nThe StarlingX project is seen some intermittence on the gitlab where\nspice-html5 source is being cloned from and we would like to select a\ndifferent source. This changes gives the option to overwrite the source\nrepos and versions of both spice-html5 and novnc.\n\nThis will also provide the means to easily test new versions of those\nlibraries when in the event of an upversion.\n\nPartial-Bug: #1956173\nSigned-off-by: Thiago Brito \u003cthiago.brito@windriver.com\u003e\nChange-Id: I3001dad62a0caed26347dcb193d101c80901b5d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a5ba0464ced1cde159bfd14ea258e0787ba9f374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a5ba0464ced1cde159bfd14ea258e0787ba9f374"}]},"branch":"refs/heads/master"},"96744d9d462b2561a593647edc91bc5bc40c189f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-10 21:45:04.000000000","uploader":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"ref":"refs/changes/77/823577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/77/823577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/77/823577/2"}}},"commit":{"parents":[{"commit":"460164cc1e42ac5d7faaea30254ddd5d7bbb9ca9","subject":"Fixing job issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/460164cc1e42ac5d7faaea30254ddd5d7bbb9ca9"}]}],"author":{"name":"Thiago Brito","email":"thiago.brito@windriver.com","date":"2022-01-05 20:55:24.000000000","tz":-180},"committer":{"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","date":"2022-01-10 21:45:04.000000000","tz":0},"subject":"Specify a different source for novnc/spice-html5","message":"Specify a different source for novnc/spice-html5\n\nThe StarlingX project is seen some intermittence on the gitlab where\nspice-html5 source is being cloned from and we would like to select a\ndifferent source. This changes gives the option to overwrite the source\nrepos and versions of both spice-html5 and novnc.\n\nThis will also provide the means to easily test new versions of those\nlibraries when in the event of an upversion.\n\nPartial-Bug: #1956173\nSigned-off-by: Thiago Brito \u003cthiago.brito@windriver.com\u003e\nChange-Id: I3001dad62a0caed26347dcb193d101c80901b5d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/96744d9d462b2561a593647edc91bc5bc40c189f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/96744d9d462b2561a593647edc91bc5bc40c189f"}]},"branch":"refs/heads/master","description":"Rebase"},"efccd0a853879ac6af6066eda09792d0d3afe9c0":{"kind":"NO_CHANGE","_number":3,"created":"2022-01-12 14:45:16.000000000","uploader":{"_account_id":32349,"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","username":"outbrito"},"ref":"refs/changes/77/823577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/77/823577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/77/823577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/77/823577/3"}}},"commit":{"parents":[{"commit":"a52d700d86fccb78ad9f92780efd5170c324325e","subject":"Fixing job issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a52d700d86fccb78ad9f92780efd5170c324325e"}]}],"author":{"name":"Thiago Brito","email":"thiago.brito@windriver.com","date":"2022-01-05 20:55:24.000000000","tz":-180},"committer":{"name":"Thiago Paiva Brito","email":"thiago.brito@windriver.com","date":"2022-01-12 14:45:16.000000000","tz":0},"subject":"Specify a different source for novnc/spice-html5","message":"Specify a different source for novnc/spice-html5\n\nThe StarlingX project is seen some intermittence on the gitlab where\nspice-html5 source is being cloned from and we would like to select a\ndifferent source. This changes gives the option to overwrite the source\nrepos and versions of both spice-html5 and novnc.\n\nThis will also provide the means to easily test new versions of those\nlibraries when in the event of an upversion.\n\nPartial-Bug: #1956173\nSigned-off-by: Thiago Brito \u003cthiago.brito@windriver.com\u003e\nChange-Id: I3001dad62a0caed26347dcb193d101c80901b5d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
