)]}'
{"id":"openstack%2Floci~834393","triplet_id":"openstack%2Floci~master~I373f3f1e745d4af19f4266c48949c48d55e715eb","project":"openstack/loci","branch":"master","topic":"placement","attention_set":{},"removed_from_attention_set":{"8898":{"account":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"last_update":"2022-04-27 21:36:32.000000000","reason":"removed on reply"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-04-27 22:13:41.000000000","reason":"Change was submitted"},"33519":{"account":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"last_update":"2022-04-27 22:13:41.000000000","reason":"Change was submitted"},"8863":{"account":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"last_update":"2022-04-27 22:13:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I373f3f1e745d4af19f4266c48949c48d55e715eb","subject":"Remove supported releases for placement","status":"MERGED","created":"2022-03-20 16:33:33.000000000","updated":"2022-04-27 22:13:41.000000000","submitted":"2022-04-27 22:13:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834393-placement","meta_rev_id":"17494768b2cc55a951dced3cdd163af471a00cda","_number":834393,"virtual_id_number":834393,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 22:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-04-27 21:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-04-27 21:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-20 17:17:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 16:29:24.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2022-04-26 16:29:24.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2022-04-26 16:29:24.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"state":"REVIEWER"}],"messages":[{"id":"96b781005bd0728420a7561364e0e58eff699152","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-20 16:33:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c256974175cd85057ddca97ca48d00d9836e2a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 17:17:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9b750f6bed8d40cb9e965fdb2026e063 : SUCCESS in 26m 07s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3ccd16c93afc43f4bab95d1e7d77565e : SUCCESS in 28m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/808d666baabd4db082812db3683da7a6 : SUCCESS in 38m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2dff256b00ac4cd88df6d37b7410c4fb : SUCCESS in 38m 51s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0ed4c1abb54140e6a4263ce88acba594 : SUCCESS in 30m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ff2bdb0a958647cfae66f5cda6f41123 : SUCCESS in 30m 28s\n- loci-manila https://zuul.opendev.org/t/openstack/build/22f7e09069b0418290636790cbbb1ff7 : SUCCESS in 30m 24s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/301fab41832a486989096fae083f461c : SUCCESS in 23m 28s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b96043d546564af9b64e6b2570c48e33 : SUCCESS in 31m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/64d03d525cfc417bbde58b3658292f24 : SUCCESS in 28m 24s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/60d3a218293b48ee9e9c407138a927e0 : SUCCESS in 28m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3792b832a7b14f10977037f2fdfef092 : SUCCESS in 20m 26s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4a72ebdb82f34e42a83c274bc5c92e79 : SUCCESS in 16m 19s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1124a4e395994acc9bce32bce0d1db02 : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"de9c09bb11689e394072775492c4f98aac26da54","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-25 21:33:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"40e4e451421cd0581e51dbcdca18687c0f60c345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-25 21:33:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ba8b5bb7c172a85f57ff76ba55beeee8203072c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 22:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b7ec9b08a32542c988de1117b17849a7 : SUCCESS in 24m 19s\n- loci-glance https://zuul.opendev.org/t/openstack/build/129c56a527d94f1cb4cd6b72296fd01f : SUCCESS in 23m 20s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a22033f704ee48848994151ee890094e : SUCCESS in 24m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/dc843415565b453fa4d0134f02aeb511 : SUCCESS in 25m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d6d43ddba28148358f404bd03eaec684 : SUCCESS in 23m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8c7306d914f942e2abd3627e0740af65 : SUCCESS in 23m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/85770077874f467e985f4e39c9626341 : SUCCESS in 23m 57s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/067d56b246b647d5a80a7902f62743c7 : SUCCESS in 22m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5ffe6593fc4547479c313eef3cb5722b : SUCCESS in 24m 37s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5c5a184111ab48e39eece838ec39f685 : SUCCESS in 26m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/99c90331231147798f1e20eeb7d56360 : SUCCESS in 23m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/11a7545ec2c74d058e693bc9be7bb001 : SUCCESS in 22m 59s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5754c7d478454b83bc4a5017ad3d9224 : SUCCESS in 17m 52s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/197c3c3cb32c4c36aa7167cc994aff02 : SUCCESS in 23m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3be31600a7cc2d138af832c1eb9e9e5f1d43047f","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-04-27 21:36:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6698b9c898ccfd685e1814443322e52e8cc392d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 21:38:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f18086c2abc50141f4bcced88b5e7cb3b452f5ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 22:13:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9580b450c17f4d01948950684c481c3b : SUCCESS in 18m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/70e2ae37b45b4e09b79107de79727281 : SUCCESS in 17m 41s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ac0d1c4b7f35453d87711cebe44e6e89 : SUCCESS in 22m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3c5436309ccc401ea982213aceecfe08 : SUCCESS in 19m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3b557e2634c547d8ad3471d57e9b2c9b : SUCCESS in 20m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cd7419bdddc941beb755cee7e00bf495 : SUCCESS in 22m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2105d2d140d945238e656e330b27f1cf : SUCCESS in 18m 25s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/501d63f8721d48259b329fba90ead6c4 : SUCCESS in 20m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/18b912c936ce4b5697f982e5c7f50e93 : SUCCESS in 26m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7004dc0bb40d4f2b81c71c7cb7aeff61 : SUCCESS in 19m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b53bc6185bcd4865a012cbf588db96e5 : SUCCESS in 24m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d7f7e3817f504340862a8ab382c99bba : SUCCESS in 20m 34s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5f638536c7f74724887094a06bf07c31 : SUCCESS in 14m 58s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d4f4693b2e5f47ecae3941b8b3976355 : SUCCESS in 22m 03s","accounts_in_message":[],"_revision_number":2},{"id":"17494768b2cc55a951dced3cdd163af471a00cda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 22:13:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d56b69f5b50ddf2b2687cb473aae86f43ea7510","revisions":{"8da6d026d599101db67c510ebb986d929ee7b210":{"kind":"REWORK","_number":1,"created":"2022-03-20 16:33:33.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/93/834393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/834393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/834393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/834393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/834393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/834393/1"}}},"commit":{"parents":[{"commit":"279af7f51bf7e32ed4816608eed1c2adc5496fd3","subject":"Merge \"aarch64: stop trying to build confluent-kafka\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/279af7f51bf7e32ed4816608eed1c2adc5496fd3"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-03-20 16:32:17.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-03-20 16:32:17.000000000","tz":-300},"subject":"Remove supported releases for placement","message":"Remove supported releases for placement\n\nPlacement is now ran by default for all latest supported releases\nof openstack. This change removes the supported_releases field\nto make sure each release of openstack we build now correctly\nsupports it.\n\nChange-Id: I373f3f1e745d4af19f4266c48949c48d55e715eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8da6d026d599101db67c510ebb986d929ee7b210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8da6d026d599101db67c510ebb986d929ee7b210"}]},"branch":"refs/heads/master"},"8d56b69f5b50ddf2b2687cb473aae86f43ea7510":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-25 21:33:59.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/93/834393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/834393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/834393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/834393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/834393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/834393/2"}}},"commit":{"parents":[{"commit":"f9676f734ef17a8030f6f0f57a92a0033d948c54","subject":"Merge \"Authenticate with registry on-demand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9676f734ef17a8030f6f0f57a92a0033d948c54"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-03-20 16:32:17.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-03-25 21:33:59.000000000","tz":0},"subject":"Remove supported releases for placement","message":"Remove supported releases for placement\n\nPlacement is now ran by default for all latest supported releases\nof openstack. This change removes the supported_releases field\nto make sure each release of openstack we build now correctly\nsupports it.\n\nChange-Id: I373f3f1e745d4af19f4266c48949c48d55e715eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8d56b69f5b50ddf2b2687cb473aae86f43ea7510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8d56b69f5b50ddf2b2687cb473aae86f43ea7510"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
