)]}'
{"id":"openstack%2Floci~839412","triplet_id":"openstack%2Floci~master~I170d4425ced0633ddbb2287411467eaedbab0051","project":"openstack/loci","branch":"master","topic":"python2","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-04-27 13:20:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I170d4425ced0633ddbb2287411467eaedbab0051","subject":"Remove python 2 references from requirements.sh","status":"ABANDONED","created":"2022-04-26 16:34:07.000000000","updated":"2022-04-27 13:20:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af80ba336e94fdd8cadc10924bf24a8d730af379","_number":839412,"virtual_id_number":839412,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-26 17:14:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 17:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10d678b59d33c45015a90609ba5c9b7c7150d738","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 16:34:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c127004872534656a21b4dda73189635121e5e25","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 16:35:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"b7c708b3dc28b40ba60869a4f55d5344078c7e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 16:37:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"dfe28c0179b8172575bd4ebfd20705fc7b79f9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 17:14:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/221f5eb3680a4b3896d8517e967797f6 : SUCCESS in 27m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ef985a94eb0649f183e70ae0499a495e : SUCCESS in 25m 23s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7cddaffdc4154ab8a414137a3239373f : SUCCESS in 28m 11s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8dc15eae3e0e40f79514ebb27d11b97a : SUCCESS in 28m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ef5fb92d44ac4cdc8c92f080abe1084c : SUCCESS in 25m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a7c5b83098f2408181a7bf9c09b21fb1 : SUCCESS in 25m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1fa68e641c764dd3b3b23165053cf258 : SUCCESS in 27m 26s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/8022140a204646c4922feee9a9dee89f : SUCCESS in 29m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cc24031e8c15407f839a6b295ea5a0f6 : SUCCESS in 28m 30s\n- loci-nova https://zuul.opendev.org/t/openstack/build/eb60bbcf29784e3f8b46e20b0275ab85 : FAILURE in 28m 44s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/aaf277e192d84c9baf7dbba398122f26 : SUCCESS in 26m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/94f87f506cb8414994112371c783eb8b : SUCCESS in 19m 21s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e3a26129c4274838935029fe01742716 : SUCCESS in 20m 47s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/4820405e3e7b4648835f5ca0015d510e : SUCCESS in 26m 06s","accounts_in_message":[],"_revision_number":2},{"id":"49ad375422238c163388fbbc9eda21529d5b8f1f","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 18:20:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"42a3888ef856aaf1c501b05c8bcfea1c5c3f39b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 18:56:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/026ad860dc504e5b86e0163f1b3ab95a : SUCCESS in 26m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d41d4e1d3c67476bb54d664fafcfa3f2 : SUCCESS in 26m 21s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f54814c2124842f2a4f2e482961f728a : SUCCESS in 26m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/afa49975e3e54efc86fb962520d3c265 : SUCCESS in 27m 27s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/513b0377a6864a539287a887304fd180 : SUCCESS in 25m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3991eff36b7d4679a3d9014e27e92be1 : SUCCESS in 26m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1dc09b120e1c4506b7ecaa1ab54d775b : SUCCESS in 26m 30s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6f6598625ebc4ec8a554c8434c1795b8 : SUCCESS in 25m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/85d6932fbc4c4c87aaab5fb5a1d741b3 : SUCCESS in 27m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/720c505b8ad94fe7a0883c9572282b8d : FAILURE in 28m 10s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/db745b4014f44619984f4be9ee224e36 : SUCCESS in 26m 13s\n- loci-placement https://zuul.opendev.org/t/openstack/build/dc7e3953eb424e55b305e6d5251a5bf3 : SUCCESS in 18m 06s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5de1ece722894d758cd1b2bb99de03a2 : SUCCESS in 19m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/17402771e5a14db6879978b3d4203b5e : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":2},{"id":"af80ba336e94fdd8cadc10924bf24a8d730af379","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-27 13:20:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"969b40aad858b106f29c990b02b897f210629f01","revisions":{"f00a581271bc404d92a7280b9d3c013d7c5fd68f":{"kind":"REWORK","_number":1,"created":"2022-04-26 16:34:07.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/12/839412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/12/839412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/839412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/839412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/839412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/12/839412/1"}}},"commit":{"parents":[{"commit":"f9676f734ef17a8030f6f0f57a92a0033d948c54","subject":"Merge \"Authenticate with registry on-demand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9676f734ef17a8030f6f0f57a92a0033d948c54"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:33:24.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:33:24.000000000","tz":-300},"subject":"Remove python 2 references from requirements.sh","message":"Remove python 2 references from requirements.sh\n\nThis change removes the python 2 references and workarounds from\nthe requirement.sh script. This is part of the effort to remove\nthe python 2 uses and hacks from the loci repository.\n\nChange-Id: I170d4425ced0633ddbb2287411467eaedbab0051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f00a581271bc404d92a7280b9d3c013d7c5fd68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f00a581271bc404d92a7280b9d3c013d7c5fd68f"}]},"branch":"refs/heads/master"},"969b40aad858b106f29c990b02b897f210629f01":{"kind":"REWORK","_number":2,"created":"2022-04-26 16:37:56.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/12/839412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/12/839412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/839412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/839412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/12/839412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/12/839412/2"}}},"commit":{"parents":[{"commit":"f9676f734ef17a8030f6f0f57a92a0033d948c54","subject":"Merge \"Authenticate with registry on-demand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9676f734ef17a8030f6f0f57a92a0033d948c54"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:33:24.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:37:56.000000000","tz":0},"subject":"Remove python 2 references from requirements.sh","message":"Remove python 2 references from requirements.sh\n\nThis change removes the python 2 references and workarounds from\nthe requirement.sh script. This is part of the effort to remove\nthe python 2 uses and hacks from the loci repository.\n\nChange-Id: I170d4425ced0633ddbb2287411467eaedbab0051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/969b40aad858b106f29c990b02b897f210629f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/969b40aad858b106f29c990b02b897f210629f01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
