)]}'
{"id":"openstack%2Floci~839414","triplet_id":"openstack%2Floci~master~Icd832edffec2cea907276480fd8c67e44459ad90","project":"openstack/loci","branch":"master","topic":"trollius","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-04-26 23:16:05.000000000","reason":"Change was submitted"},"8863":{"account":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"last_update":"2022-04-26 22:17:25.000000000","reason":"Andrii Ostapenko replied on the change"}},"hashtags":[],"change_id":"Icd832edffec2cea907276480fd8c67e44459ad90","subject":"Remove trollius workaround","status":"MERGED","created":"2022-04-26 16:42:20.000000000","updated":"2022-04-26 23:16:05.000000000","submitted":"2022-04-26 23:16:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839414-trollius","meta_rev_id":"2c0e3b2c89265cf31f0f59743a9e0d8360488eef","_number":839414,"virtual_id_number":839414,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 23:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-04-26 21:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-26 20:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-04-26 21:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 17:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 20:35:09.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2022-04-26 21:15:37.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"64de28924d39fe305a128eea1f248be2c8ac8eb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 16:42:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2550665ee427b2f129efa9eb8e32b6b06decd6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 17:23:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7bf5dd34bcac433cbf68f80c45d0d2b1 : SUCCESS in 28m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8e09bf973a0a43b5ae897fbe54fb79a5 : SUCCESS in 25m 41s\n- loci-heat https://zuul.opendev.org/t/openstack/build/569a7d59f2bf4024a1e1b0a8502251d1 : SUCCESS in 26m 47s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ab326d781d6e47bab2740f1ad8d2baba : SUCCESS in 24m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/14c7e1addb7d496089544c9c5a72e5bd : SUCCESS in 25m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4af3d6a6a8b349b38dda34542fb2ec9c : SUCCESS in 26m 07s\n- loci-manila https://zuul.opendev.org/t/openstack/build/37f350ec68d24938a5f023ea8900a091 : SUCCESS in 30m 08s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ab0b9f733755445d856ff26208912ef1 : SUCCESS in 22m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cb34a551ba594e4d87bf6fa32f2fdd6c : SUCCESS in 27m 03s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7fc429014a1749559f2d1a31f2db8735 : FAILURE in 28m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/6c5ac7fc4aee4f3581fd2df07317f653 : SUCCESS in 26m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2992d39dea17407e8073c7f2edf9396a : SUCCESS in 16m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f73e14e8e3944190994a54ae0dc0308d : SUCCESS in 20m 15s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/9e21331d0b674fad91d459a0750015dc : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4d818e70e0cd5863fdb6a68f048470d6d35db107","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 18:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d13ff21f99d690fa9bd7ae9c4de325f65358cd04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 18:55:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7a099d04c7334dcb82892e534c5cf2ac : SUCCESS in 23m 33s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7dc65ff1d2ee4fa29e409f08178a2f96 : SUCCESS in 26m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/626abff35b4c479fa54280230cf927d1 : SUCCESS in 26m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/af04cc2b02bf412992943236750243cb : SUCCESS in 27m 29s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a40087ee3cb94eb3a298a96a195fc80e : SUCCESS in 25m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/639ca5e021c84be0aa035e3a00cb47b0 : SUCCESS in 26m 00s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f08e977976e54610840019ca2b853133 : SUCCESS in 25m 57s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/95d00ca13f5340969853d10f6269c8cf : SUCCESS in 24m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/223d9cabf40b4e02b0f5c00bbbd5a447 : SUCCESS in 27m 30s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2e781216087e4c5f9621c4b3dea9a817 : FAILURE in 28m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5e413e0601b44514865107a7eef95c92 : SUCCESS in 25m 40s\n- loci-placement https://zuul.opendev.org/t/openstack/build/49b3c8e9daee41e9928b7af5c645cddb : SUCCESS in 18m 10s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/efa3e0a21af64e9496315c09e88f7e46 : SUCCESS in 19m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2d7fb8564b5d4c6283557523870877b3 : SUCCESS in 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5b0e3a70014aebd6ebe294e42eabebcf875b04ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 19:25:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1b542d64217f4c8fb9b94e15243041e5df716b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 19:50:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/fe7ad2fc8c614c2295a7a125c14eaedc : SUCCESS in 19m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/868e0798e99c49d68269ddd09e4a06c8 : SUCCESS in 16m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/50d587f7831d4dbe8f9693796e9dfe39 : SUCCESS in 21m 51s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6df9c0ee1ed44f8fad4e35c4f6e58203 : SUCCESS in 19m 29s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8b02c09c040c4b69b6989702376deb57 : SUCCESS in 21m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9d2b7ce6d0004e55aed6e7884868007c : SUCCESS in 21m 51s\n- loci-manila https://zuul.opendev.org/t/openstack/build/aceb5e8a719c47c6aa05389f0fa1415c : SUCCESS in 19m 03s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d0b6e563977a4d598e42448c27d6dca9 : SUCCESS in 21m 07s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/db5d97eb55b346ebbaedb3225ca88cb1 : SUCCESS in 17m 52s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0969ff4f08804060913949cb02703b2c : FAILURE in 23m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e6627b39d89944779b10434c3d7ee63d : SUCCESS in 21m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cfca1088abda4ff7afbd1879b4517d64 : SUCCESS in 12m 27s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a2f3d6a4a2534ed887a62632103ecf71 : SUCCESS in 17m 39s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/77d66d4a6f664f208f51f757e1fbc3ca : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e7fdd8066b1ece437165a3b222a018ad58c367ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 19:54:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"92625253af784e9a54e38adb2d0d879f94541224","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-04-26 20:31:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b01ffd1fd3a76b7044f6d2b45fc362179eeceafe","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-04-26 20:34:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"8a9423b04c222db071b3cdb3153855dd5cd7d388","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-04-26 20:35:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"db4fa090649e57405dda527ed8670073d2f95339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 21:14:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/73ebe045cd0b42688283a392c07044a2 : SUCCESS in 28m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1e2db3e8634f4f13a7f270f742624f79 : SUCCESS in 28m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4c1c64ec19704e5faf9bdba2f5662be8 : SUCCESS in 36m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a7b6a61d2b724904ace77fa32934feea : SUCCESS in 24m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ae70d1c53f0a45259432f3baa905e84e : SUCCESS in 24m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/644842ca7d8946fb8cf9125db1020cb2 : SUCCESS in 24m 43s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5b8b32677e114fd3b1f305c0608a171f : SUCCESS in 23m 29s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/decc95b30370488a96ab82d75f75c1a6 : SUCCESS in 22m 56s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/432867a0ab86447b81583ea0d641211e : SUCCESS in 25m 02s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5abf849db25f4712a6c91df0abaac83c : SUCCESS in 26m 17s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/368cbc48dfb146ab9c578b744cd16ea9 : SUCCESS in 31m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/40cf7235031d4e6da27275ed7a082941 : SUCCESS in 18m 40s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6d37bd162dd445d4a32468c41354fba9 : SUCCESS in 18m 11s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/88347b39bce649ee93edbcd5d908fe0a : SUCCESS in 28m 09s","accounts_in_message":[],"_revision_number":4},{"id":"868bbc23dc3b8b83bcc79fcfd675eb4c1c628fd9","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-04-26 21:15:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"308a747f86875b4e8a685438c1acc62f9803bf91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 21:16:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8a61283df53d16c31d2396504945f61eeb4f1390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 21:58:26.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/daf92908220443dd92bb5a3258c1e3f5 : SUCCESS in 30m 03s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9690eab2b1fb46de806235f781bb72a1 : SUCCESS in 22m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/300b1323434d46eb970a2d7f56dc022e : SUCCESS in 29m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/954270fecd1e466fac0dce00451c1841 : SUCCESS in 23m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/17571aa7121842ceb499f89d45ecd23b : SUCCESS in 29m 09s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f076158697a0470785c5ae7b4cccfa9a : FAILURE in 16m 03s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2c18f80eab91493da65bd0cddb041fe9 : SUCCESS in 28m 19s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/00a0533051bf448296487b8b7906f9a1 : SUCCESS in 34m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7606823c158340d2ab320d02caf0f21b : SUCCESS in 35m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3c85ba8c7a9b4bb3966f66c24005e159 : SUCCESS in 32m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7ff44f2db0ca4f9a9390f99590e8d552 : SUCCESS in 31m 52s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8296040ec7f549e5b42e97c03bf2e3cc : SUCCESS in 17m 39s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/96a97d3283404f83b5f9c959d0e3c5ed : SUCCESS in 21m 05s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/cb98e1eb629c432c9df5d02167bd4f7a : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4f77d966c152b392d8b8be7209577fc754c443b5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-04-26 22:17:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6986508ce0e6cc6f8a1985ff837549c0bbaab25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 22:45:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/bd3883eeca6d4ddda96dd9fe7b078171 : SUCCESS in 20m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ec13bf8b494f42508194ff9b7bfc333a : SUCCESS in 20m 38s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e2b8f3f293fd4ae2a4bcc300b9f3437b : SUCCESS in 21m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/39b280c4a4a941788b3f6d6cb22b13bd : SUCCESS in 19m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/07e3cecf09274c189638d8fcf5d2cbba : SUCCESS in 22m 16s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c639c2b76a324b1a8a6a5f5c24fc5627 : SUCCESS in 20m 07s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8e279fd32aa24cecb516991a6cfd769d : SUCCESS in 21m 46s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/7ec82a0bb0914912bee1dc43189eb851 : SUCCESS in 17m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/29b6e5c65d3041c69df95cc3db4ad541 : SUCCESS in 18m 09s\n- loci-nova https://zuul.opendev.org/t/openstack/build/81a38459042f4f45a0082b2e0bfc094b : SUCCESS in 18m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d75272f7e36f4adb91ea03c49f2c8de4 : SUCCESS in 17m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ea194f3d12f0457fb924adb6a2a9afd9 : SUCCESS in 9m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/07f6284edc2244408e71b7d6e66e0157 : SUCCESS in 13m 09s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/28b6d8a7034b4d7290d2cf61c25f919c : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":4},{"id":"ab4f917347cb42c4d87c24727bfa2a702f8d90b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 22:46:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"477359526606e6de90736b29bac634e1775be4a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 23:16:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e2ae3f6c3abf463684e7102d4da612cd : SUCCESS in 21m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/233e51ac191641a59a977b3fa399c521 : SUCCESS in 19m 38s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8cdbe0aa98204a99a7d7d4709a23eecd : SUCCESS in 21m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d44c414f539e4f7bbbb18675418784df : SUCCESS in 24m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c6705ab79fb04e86867358984c07889d : SUCCESS in 20m 21s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cb9f8c78e7594ef692fbb80c1b29049a : SUCCESS in 20m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/26554e5200894f00a5a104a17bc71a10 : SUCCESS in 21m 03s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/58cfde306bda483d83d8745408066fa0 : SUCCESS in 21m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/bd6a10d8605b4f4b8c23986a3c876d50 : SUCCESS in 22m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/31d66d737f6d4a17b1eeb95422fae917 : SUCCESS in 23m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5f9fc62710854fc8aced55620b29584c : SUCCESS in 23m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ba81832822fa4ebf9f5fc186cdaea6e8 : SUCCESS in 10m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/014a0d83984f41c5ad9fad39e360251c : SUCCESS in 15m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f04910e061574ddaa822d56773d3ad34 : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":4},{"id":"2c0e3b2c89265cf31f0f59743a9e0d8360488eef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 23:16:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8b514187ccaad6b4040fa0145df11c76e4fe79ef","revisions":{"241a1b90be34576cc626bd97f13820b22375d74f":{"kind":"REWORK","_number":1,"created":"2022-04-26 16:42:20.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/839414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/839414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/839414/1"}}},"commit":{"parents":[{"commit":"f9676f734ef17a8030f6f0f57a92a0033d948c54","subject":"Merge \"Authenticate with registry on-demand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9676f734ef17a8030f6f0f57a92a0033d948c54"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:41:04.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:41:04.000000000","tz":-300},"subject":"Remove trollius workaround","message":"Remove trollius workaround\n\nThe workaround for trollius appears to be no longer needed since\nopenstack has moved away from python 2. This change removes the\nlogic to handle removing trollius since it\u0027s no longer present\nin upper-constraints.\n\nChange-Id: Icd832edffec2cea907276480fd8c67e44459ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/241a1b90be34576cc626bd97f13820b22375d74f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/241a1b90be34576cc626bd97f13820b22375d74f"}]},"branch":"refs/heads/master"},"5563026120499dcaa895105454ede6c4d7024c29":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-26 19:25:09.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/839414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/839414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/839414/2"}}},"commit":{"parents":[{"commit":"f9676f734ef17a8030f6f0f57a92a0033d948c54","subject":"Merge \"Authenticate with registry on-demand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9676f734ef17a8030f6f0f57a92a0033d948c54"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:41:04.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 19:25:09.000000000","tz":0},"subject":"Remove trollius workaround","message":"Remove trollius workaround\n\nThe workaround for trollius appears to be no longer needed since\nopenstack has moved away from python 2. This change removes the\nlogic to handle removing trollius since it\u0027s no longer present\nin upper-constraints.\n\nDepends-on: https://review.opendev.org/c/openstack/loci/+/837412\n\nChange-Id: Icd832edffec2cea907276480fd8c67e44459ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5563026120499dcaa895105454ede6c4d7024c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5563026120499dcaa895105454ede6c4d7024c29"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5fba82841218963d9ac346dc0c9d10fa794ac82e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-26 19:54:04.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/839414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/839414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/839414/3"}}},"commit":{"parents":[{"commit":"f9676f734ef17a8030f6f0f57a92a0033d948c54","subject":"Merge \"Authenticate with registry on-demand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9676f734ef17a8030f6f0f57a92a0033d948c54"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:41:04.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 19:54:04.000000000","tz":0},"subject":"Remove trollius workaround","message":"Remove trollius workaround\n\nThe workaround for trollius appears to be no longer needed since\nopenstack has moved away from python 2. This change removes the\nlogic to handle removing trollius since it\u0027s no longer present\nin upper-constraints.\n\nChange-Id: Icd832edffec2cea907276480fd8c67e44459ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5fba82841218963d9ac346dc0c9d10fa794ac82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5fba82841218963d9ac346dc0c9d10fa794ac82e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8b514187ccaad6b4040fa0145df11c76e4fe79ef":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-26 20:31:51.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/14/839414/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/839414/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/839414/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/839414/4"}}},"commit":{"parents":[{"commit":"3eceae492d63d0d436d11a521fe6b2337c3c6374","subject":"Add required args for ussuri build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3eceae492d63d0d436d11a521fe6b2337c3c6374"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-04-26 16:41:04.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2022-04-26 20:31:51.000000000","tz":0},"subject":"Remove trollius workaround","message":"Remove trollius workaround\n\nThe workaround for trollius appears to be no longer needed since\nopenstack has moved away from python 2. This change removes the\nlogic to handle removing trollius since it\u0027s no longer present\nin upper-constraints.\n\nChange-Id: Icd832edffec2cea907276480fd8c67e44459ad90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8b514187ccaad6b4040fa0145df11c76e4fe79ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8b514187ccaad6b4040fa0145df11c76e4fe79ef"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
