)]}'
{"id":"openstack%2Floci~842906","triplet_id":"openstack%2Floci~master~I9f8a81ed160a5d09b01e846b091d5bbb2bee1656","project":"openstack/loci","branch":"master","topic":"story/2010049","attention_set":{},"removed_from_attention_set":{"34906":{"account":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"last_update":"2022-09-20 18:03:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f8a81ed160a5d09b01e846b091d5bbb2bee1656","subject":"Remove Git repo data from NoVNC and SPICE HTML5 dirs - Story: 2010049","status":"MERGED","created":"2022-05-23 03:25:11.000000000","updated":"2022-09-20 18:03:31.000000000","submitted":"2022-09-20 18:03:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842906-story/2010049","meta_rev_id":"f45d441cafcfb4f544b14b85fe647250da2a282d","_number":842906,"virtual_id_number":842906,"owner":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 18:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-09-20 17:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":2,"date":"2022-09-20 16:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-09-20 17:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 03:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 16:02:11.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2022-09-20 17:35:16.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"784b39e1c754d7ff30e096ea61a4a9bf10ac0aaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"date":"2022-05-23 03:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2da4be0da5962a1ccd6ccaf400b32e3c6bfb42c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"date":"2022-05-23 03:27:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"e02837a2d751c54628d0952baa3aac902df59f43","tag":"autogenerated:gerrit:restore","author":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"date":"2022-05-23 03:27:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a9e13a8b881cfd653e3c42f7cf98ab6cbeee9315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"date":"2022-05-23 03:28:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5111719dc37be54b9ede17fbc475319e6aaca798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 03:51:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ca62604c72c7482096f909db66dc18e7 : SUCCESS in 18m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ddaf4b975647462f96250ef1becb757a : SUCCESS in 17m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f400ae2c63ae4bdfae096c49edebfcb9 : SUCCESS in 21m 02s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ba3c4cb8a28a40a4b529c981b726c030 : SUCCESS in 18m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/84af99877c5d4ef98505ba4781223e56 : SUCCESS in 17m 09s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/13ebfd92f0a842e799ef32098d59a0bd : SUCCESS in 21m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/bd97e9e869ca4ec59945e7ec09d92898 : SUCCESS in 21m 07s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ec9e60e49b4c43be89ee393131aa3576 : SUCCESS in 16m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2a5ba62a0fb94738b5529c3ef3a200db : SUCCESS in 18m 47s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a4b54dcd6bc24212af7757ee647d8d77 : SUCCESS in 18m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c3e4f5f3adc14e958981698a11cd9842 : SUCCESS in 17m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/15eea439b62247cfa6f75df69306b693 : SUCCESS in 17m 17s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b6cd11506f7744f8acf1ab74ff506f0a : SUCCESS in 13m 25s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5c37573cb4b8434dad8f0f703d7ae9db : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7f392a6219de7e4cb3fd58cdcadb3971fdf602e9","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-20 16:02:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e33f368dd424291e6dc897e39b5c3beb24c9cc4","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-09-20 17:35:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c807c1f74685ae1c9d2bcff6ec5b0693298baa1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 17:35:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1aab80fda6b0a9b4dc9ab808879b55c1429c1dc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 18:03:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ba2785b3f1a4693aede794704f7fe86\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4f671417b72640b5934c1dab0574ae38 : SUCCESS in 17m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9532c4ec3fad44589d023dc7a4f438a5 : SUCCESS in 17m 34s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f9fc4b27afb0456b9b724e7ae2222216 : SUCCESS in 20m 25s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9e9a9d7f80274cf9974e5e530aed0b2c : SUCCESS in 19m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/aa586b8eedf74daa84a4950901dbe7aa : SUCCESS in 17m 47s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1bf5a2a731a145729775832c35d4d3cf : SUCCESS in 25m 05s\n- loci-manila https://zuul.opendev.org/t/openstack/build/996ea827414d45139e90775724a07605 : SUCCESS in 17m 47s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ac037d6b851e4784942aa3fdc2660363 : SUCCESS in 23m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/46ee1c6f6a8b4c02a96b034afd6e45c0 : SUCCESS in 19m 19s\n- loci-nova https://zuul.opendev.org/t/openstack/build/10200055df224877bd8eb16805f1d340 : SUCCESS in 19m 43s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e612ec57a253409eb76164ffc55524b1 : SUCCESS in 24m 49s\n- loci-placement https://zuul.opendev.org/t/openstack/build/402d83c031f445ff863179b8c316e16c : SUCCESS in 17m 28s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/cfe156df6e8a4d7893567ee78fcd0ae8 : SUCCESS in 13m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fdbc6620c8e24c25a09b269bc3a32022 : SUCCESS in 18m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f45d441cafcfb4f544b14b85fe647250da2a282d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 18:03:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe9caeebbbfbcf8f241314897c568fbd45bfcbc9","revisions":{"09cc2b800cc4ecde6eea6b3b1548a35b27a6e1f9":{"kind":"REWORK","_number":1,"created":"2022-05-23 03:25:11.000000000","uploader":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"ref":"refs/changes/06/842906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/06/842906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/06/842906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/06/842906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/06/842906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/06/842906/1"}}},"commit":{"parents":[{"commit":"eb95b4adab5b8659ba95131c4bc234776b21cda5","subject":"Merge \"Remove supported releases for placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eb95b4adab5b8659ba95131c4bc234776b21cda5"}]}],"author":{"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","date":"2022-05-23 03:24:17.000000000","tz":720},"committer":{"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","date":"2022-05-23 03:24:17.000000000","tz":720},"subject":"Remove Git repo data from NoVNC and SPICE HTML5 dirs - stoy 2010049","message":"Remove Git repo data from NoVNC and SPICE HTML5 dirs - stoy 2010049\n\nChange-Id: I9f8a81ed160a5d09b01e846b091d5bbb2bee1656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/09cc2b800cc4ecde6eea6b3b1548a35b27a6e1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/09cc2b800cc4ecde6eea6b3b1548a35b27a6e1f9"}]},"branch":"refs/heads/master"},"fe9caeebbbfbcf8f241314897c568fbd45bfcbc9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-23 03:28:09.000000000","uploader":{"_account_id":34906,"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","username":"jeremyl"},"ref":"refs/changes/06/842906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/06/842906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/06/842906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/06/842906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/06/842906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/06/842906/2"}}},"commit":{"parents":[{"commit":"eb95b4adab5b8659ba95131c4bc234776b21cda5","subject":"Merge \"Remove supported releases for placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eb95b4adab5b8659ba95131c4bc234776b21cda5"}]}],"author":{"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","date":"2022-05-23 03:24:17.000000000","tz":720},"committer":{"name":"Jeremy Lamb","email":"jeremy.lamb@catalystcloud.nz","date":"2022-05-23 03:28:09.000000000","tz":0},"subject":"Remove Git repo data from NoVNC and SPICE HTML5 dirs - Story: 2010049","message":"Remove Git repo data from NoVNC and SPICE HTML5 dirs - Story: 2010049\n\nChange-Id: I9f8a81ed160a5d09b01e846b091d5bbb2bee1656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fe9caeebbbfbcf8f241314897c568fbd45bfcbc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fe9caeebbbfbcf8f241314897c568fbd45bfcbc9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[]}
