)]}'
{"id":"openstack%2Floci~850263","triplet_id":"openstack%2Floci~master~Icf42b1083346940c607d6e09c42376b11e03fa0d","project":"openstack/loci","branch":"master","topic":"centos","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-07-25 14:54:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icf42b1083346940c607d6e09c42376b11e03fa0d","subject":"Remove centos building references","status":"ABANDONED","created":"2022-07-19 00:16:06.000000000","updated":"2022-07-25 14:54:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c38e3720bbed6d11b2bd9f9f652955747b24d064","_number":850263,"virtual_id_number":850263,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-19 00:44:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 00:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c21865f9613b52663051db5d5245afc39b069b39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-19 00:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c83034ad188193de33f5b765d72832dbf241c227","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-19 00:17:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"6bc91d0145b881970b0d9b3ecca1167e75c26e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 00:21:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abd735cf10ab47f3905a47afa2c0f599\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/35583a3081f74c708f0c151687255ede : FAILURE in 3m 18s\n- loci-glance https://zuul.opendev.org/t/openstack/build/592f9ebf5b714eb6ad8ebf5d6072766a : FAILURE in 3m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e0d5c689de1a46f0b6abfc9be2da4eef : FAILURE in 3m 14s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/acd9c6c359a2424eb5fd8cd0cb8e9d5c : FAILURE in 3m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e91e7dccec5c416db2f778c3d51f5400 : FAILURE in 3m 14s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c5ef458f6d9a471eaf8c0a38ab4c5b06 : FAILURE in 3m 18s\n- loci-manila https://zuul.opendev.org/t/openstack/build/cd777564aa954a83aa0d6eee107e96ad : FAILURE in 3m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/69d5b6ed181e45c8ba15d6782e3806d9 : FAILURE in 3m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1196b8bf97d54e378b3332712134cc56 : FAILURE in 3m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8735b7daf80547a5aae00e4f62dbebee : FAILURE in 3m 21s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e165970f3c79479b81605fdcb7081fb6 : FAILURE in 3m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a916d09c02bc4e46be8ada54f7160b97 : FAILURE in 3m 17s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b50d68f59c354140a431456b7e0ba9fe : FAILURE in 3m 08s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/54855dc1a65942ffaaddbb7f72f61525 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"08f227ffcc50d9b36de6e1d9a31ed939d7ada108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-19 00:28:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6442349ede16fbc0fa912916e090934d3e641a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 00:44:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78b68f93908c4bcf9d3f8ab103799840\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b197dd50d28248249e727f7ba691f78d : FAILURE in 6m 01s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3978bce572b34240a1bf1dbaf23ed37f : FAILURE in 3m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/239e93a7b7094b989907354dda1616e6 : FAILURE in 6m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6b40664ca5174d3e889a4d16755ae3d7 : FAILURE in 7m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/cc6f4d3a490b422c9e7e7dc212bfbd5e : FAILURE in 6m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/662225f5f47d47edb0f217b4336bb0ff : FAILURE in 6m 05s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3578d51e2fc04632ba23803e03c3aaab : FAILURE in 6m 02s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6dd085c6fa3448d6a1a747a09ed31fb4 : FAILURE in 6m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/95ed8115f93c4d2d8466f109ecbda9a3 : FAILURE in 6m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e13c8ded00dd4957b7364e641021ad58 : FAILURE in 5m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f4a964be2fc347e29a49e9dc41116e78 : FAILURE in 6m 14s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4b61107711ee431cbdf1daf9a862b043 : FAILURE in 5m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/657a6bfcd1f54abeab08231cddd05d2b : FAILURE in 5m 51s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/78be622f8a3b4328a8dc1706b1670386 : FAILURE in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c38e3720bbed6d11b2bd9f9f652955747b24d064","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-25 14:54:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bbc1b171c6e3c176c5bc620dde390b8086d496b","revisions":{"9b899a0aa7e6829eb0a83371354145b21e1f8470":{"kind":"REWORK","_number":1,"created":"2022-07-19 00:16:06.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/63/850263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/850263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/850263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/850263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/850263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/850263/1"}}},"commit":{"parents":[{"commit":"9a3d79d4a40184d260e9708531eb18b74eb6042f","subject":"Merge \"Reduce build targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9a3d79d4a40184d260e9708531eb18b74eb6042f"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-19 00:15:09.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-19 00:15:09.000000000","tz":-300},"subject":"Remove centos building references","message":"Remove centos building references\n\nWe haven\u0027t been building centos images in a long time. Since no\none is currently supporting them, this change removes any of\nthe centos references.\n\nIf someone wishes to support it in the future, these can be added\nback.\n\nChange-Id: Icf42b1083346940c607d6e09c42376b11e03fa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9b899a0aa7e6829eb0a83371354145b21e1f8470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9b899a0aa7e6829eb0a83371354145b21e1f8470"}]},"branch":"refs/heads/master"},"7bbc1b171c6e3c176c5bc620dde390b8086d496b":{"kind":"REWORK","_number":2,"created":"2022-07-19 00:28:33.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/63/850263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/850263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/850263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/850263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/850263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/850263/2"}}},"commit":{"parents":[{"commit":"9a3d79d4a40184d260e9708531eb18b74eb6042f","subject":"Merge \"Reduce build targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9a3d79d4a40184d260e9708531eb18b74eb6042f"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-19 00:15:09.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-19 00:28:27.000000000","tz":-300},"subject":"Remove centos building references","message":"Remove centos building references\n\nWe haven\u0027t been building centos images in a long time. Since no\none is currently supporting them, this change removes any of\nthe centos references.\n\nIf someone wishes to support it in the future, these can be added\nback.\n\nChange-Id: Icf42b1083346940c607d6e09c42376b11e03fa0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7bbc1b171c6e3c176c5bc620dde390b8086d496b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7bbc1b171c6e3c176c5bc620dde390b8086d496b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
