)]}'
{"id":"openstack%2Floci~851919","triplet_id":"openstack%2Floci~master~I9443721d0f2bd78ecee6d29e3d603be08e98e537","project":"openstack/loci","branch":"master","topic":"remove_suse","attention_set":{},"removed_from_attention_set":{"14119":{"account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"last_update":"2022-09-21 17:50:51.000000000","reason":"\u003cGERRIT_ACCOUNT_14119\u003e replied on the change","reason_account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"}},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-09-21 18:32:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9443721d0f2bd78ecee6d29e3d603be08e98e537","subject":"Remove suse references from bindep","status":"MERGED","created":"2022-08-02 16:50:36.000000000","updated":"2022-09-21 18:32:29.000000000","submitted":"2022-09-21 18:32:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851919-remove_suse","meta_rev_id":"078c86304cfb7c2e0befc6068f7db2b11f4e902c","_number":851919,"virtual_id_number":851919,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 18:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-09-20 17:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":2,"date":"2022-09-21 17:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":1,"date":"2022-09-21 17:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 17:25:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 01:50:30.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2022-09-20 15:50:28.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"}],"messages":[{"id":"0017a208c098b2aa1fffbe853af1e87c50041bdb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-08-02 16:50:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a14f385f04c1b4f18d783a7bdbb6e7f097ab401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:25:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309fc873bc0d4006b4f193a7db0461e2\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/098d5267d3cf4a37a1476a2d9a5cb0f9 : SUCCESS in 20m 25s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4a2b568b1dd84ca3b643e5a2f675c08b : SUCCESS in 18m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/62b4b6b29a644defb74b80c8c4f0eb8d : SUCCESS in 18m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3edc0c40a81d46dfba326d0477c65c4a : SUCCESS in 18m 55s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f4c1f7eb3e264ed9a2db3ffd98b38909 : SUCCESS in 18m 09s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b203f2773f614aa38baab4b4d0babc89 : SUCCESS in 18m 34s\n- loci-manila https://zuul.opendev.org/t/openstack/build/213e00e5878e4b98870a70ac999970f6 : SUCCESS in 19m 19s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2ef8308088f54b9f9ac41b76905966c7 : SUCCESS in 17m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9e5e8955da954e2084ea0aa4ea061202 : SUCCESS in 16m 50s\n- loci-nova https://zuul.opendev.org/t/openstack/build/68aa5ae4a2c144438afcfa31383ff9e8 : SUCCESS in 19m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2fbaa2943f84485e93c77021daac6541 : SUCCESS in 32m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a1725234104646638666a864d016ce59 : SUCCESS in 17m 45s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6b33f8bc5ad0477fbbf8d47cfa5460fe : SUCCESS in 18m 55s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c4a625d3e4494cd6b26fc6404ba47f8e : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6845694462cba9b62a305f9751c3e1b5dfe32ca6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-08-02 18:00:05.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"ddef132d925631a6d6fe14c14be3fbdc3252ac64","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-08-04 01:50:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1fd3b666f5e10dbc363c3b89fe4e8971f46c9633","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-20 15:50:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ceba30af90bd5c4f4741b39b8b6b7afd21f87ef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:15:11.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/loci for 851919,1","accounts_in_message":[],"_revision_number":1},{"id":"f5cdaea06b56f7e5086bea561de2b870109c8bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-20 16:44:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f70b0ed363d9dcfcc313ba60f81eb1b39056c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 17:17:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/660f0671cd2d46f0b9ee06e1523e6e3c\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9a1ffac366d04796919e5f88d0a45fca : SUCCESS in 24m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d9e98be5486941109b749237d4f301ab : SUCCESS in 20m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6b96f4836976407e8d773d2cc67bfba3 : SUCCESS in 19m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a9af2b998a8b49e18aa304c4eb11141f : SUCCESS in 28m 07s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3e0ac9813d4b4ddc84856d71016a27ca : SUCCESS in 27m 11s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9eefda0f270246368da6f219e93175de : SUCCESS in 18m 38s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7a4b500ff99c4a29845f45020b79dbf6 : SUCCESS in 26m 29s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6cdf49da6de44b58a9355ef576aca268 : SUCCESS in 18m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/251c78a0555345f8a154ac03f0acb0b1 : SUCCESS in 28m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b478402b54094266a2184fe838c833b1 : SUCCESS in 20m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/28c8259c930048ce85da8153270b42dc : SUCCESS in 17m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/279da4b8dc334d82b8196d26874c370c : SUCCESS in 17m 52s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d2b501c27e8047a98d62511b73caebcd : SUCCESS in 16m 16s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2bda79a7e92145679e723055261887fc : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7b6e920d82a02d47d8eaec59d683fb3450d2bcff","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-09-20 17:35:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e9acd6189a45a9955c110572d298fb8990215d3","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 17:50:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36fb894c3e3efc0aafb16483af79eceadcd20825","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 17:51:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5c0d4e14f9327e76d83f3ef2c6fb240d5114bd51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 18:32:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6736fa15914229b77dee3aeab5e91e\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/acf611f0d26d416a9fb33161f624976a : SUCCESS in 21m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4d38d23767594e0e9644d935268de4e0 : SUCCESS in 16m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/14942ea06aab409782bfa0c4dcb6a72e : SUCCESS in 16m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0726ad0e832e48409ae73fa2ff52d84e : SUCCESS in 15m 36s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ab111b15fe9f432cb08f306e127b3790 : SUCCESS in 20m 30s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/680d6674f53e4951a2806ef3427a59d1 : SUCCESS in 19m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/cf3f7882e4e74e8da1e1148dffd89e81 : SUCCESS in 19m 48s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4f256e1e7176465c87cb7e8e6c4215ad : SUCCESS in 16m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9c45ff1a38d24bca9048c03b0a4888d3 : SUCCESS in 21m 38s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3bca35d2bc7e43898188af4414d5b679 : SUCCESS in 21m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/743fc53d41c340e7b0c90946b63dd355 : SUCCESS in 18m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4b53b91aa3164562ac03d3bd695b5be5 : SUCCESS in 18m 56s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d0594857fbaa470c87ec6c405f9fe501 : SUCCESS in 12m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/12ed9194942b43b3b4a5db1cc11c6033 : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"078c86304cfb7c2e0befc6068f7db2b11f4e902c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 18:32:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd4e6ac1e3ee38365127aa708ec92295b013bded","revisions":{"400b0bbbe6a7eba53e64684228d69e3ded2f113f":{"kind":"REWORK","_number":1,"created":"2022-08-02 16:50:36.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/19/851919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/19/851919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/19/851919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/19/851919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/19/851919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/19/851919/1"}}},"commit":{"parents":[{"commit":"2537db07aa3a3836cd215281e2fe2aa7923706b0","subject":"Refactor docker image task values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2537db07aa3a3836cd215281e2fe2aa7923706b0"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-08-02 16:49:53.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-08-02 16:49:53.000000000","tz":-300},"subject":"Remove suse references from bindep","message":"Remove suse references from bindep\n\nAs part of our effort to clean up unsupported linux distros, this\nchange removes the suse references in loci\u0027s bindep.txt.\n\nAlso removed a few python2 leftover references.\n\nChange-Id: I9443721d0f2bd78ecee6d29e3d603be08e98e537\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/400b0bbbe6a7eba53e64684228d69e3ded2f113f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/400b0bbbe6a7eba53e64684228d69e3ded2f113f"}]},"branch":"refs/heads/master"},"fd4e6ac1e3ee38365127aa708ec92295b013bded":{"kind":"REWORK","_number":2,"created":"2022-09-20 16:44:31.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/19/851919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/19/851919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/19/851919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/19/851919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/19/851919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/19/851919/2"}}},"commit":{"parents":[{"commit":"e77485f8b0d0e72923dbad5e449b1d9ef48b6a22","subject":"Bump centos version up to 9-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e77485f8b0d0e72923dbad5e449b1d9ef48b6a22"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-08-02 16:49:53.000000000","tz":-300},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 16:44:17.000000000","tz":-240},"subject":"Remove suse references from bindep","message":"Remove suse references from bindep\n\nAs part of our effort to clean up unsupported linux distros, this\nchange removes the suse references in loci\u0027s bindep.txt.\n\nAlso removed a few python2 leftover references.\n\nChange-Id: I9443721d0f2bd78ecee6d29e3d603be08e98e537\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fd4e6ac1e3ee38365127aa708ec92295b013bded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fd4e6ac1e3ee38365127aa708ec92295b013bded"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"}}]}],"submit_requirements":[]}
