)]}'
{"id":"openstack%2Floci~853555","triplet_id":"openstack%2Floci~master~I50b92ea36ab2d4157a8ec9e118a7f49fbd8ede2f","project":"openstack/loci","branch":"master","topic":"cleanup/cryptography","attention_set":{},"removed_from_attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-09-27 14:17:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50b92ea36ab2d4157a8ec9e118a7f49fbd8ede2f","subject":"Drop cryptography from dependencies","status":"MERGED","created":"2022-08-17 22:39:25.000000000","updated":"2022-09-27 14:17:32.000000000","submitted":"2022-09-27 14:17:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853555-cleanup/cryptography","meta_rev_id":"03d0fe41aaa4a1984019775faf80a4dfecaa0ff7","_number":853555,"virtual_id_number":853555,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-27 14:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"all":[{"value":2,"date":"2022-09-21 21:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":2,"date":"2022-09-27 13:46:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":1,"date":"2022-09-27 13:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":1,"date":"2022-09-27 13:46:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 22:56:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 16:05:44.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2022-09-27 13:46:28.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-09-27 13:48:28.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"b62b6ea5694b8b262be8c9f09fc3108f1652ba35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2022-08-17 22:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96c456eb3fd54fd4c91eb805b39d599f3e928cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 22:56:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd542d0b853a48c6ba01635bff98dd4b\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3519c6de2cc64d939f9ce111d4f85ea0 : FAILURE in 14m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/26f11c6db3f24d4cb21c9b2b2370b00c : FAILURE in 14m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/82c94797629b4121a00bef98c7e81205 : FAILURE in 14m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f53ee65911294a88943c1f5e7036fe4e : FAILURE in 14m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d8bb8e4631dc48b9a47ae75bbc41ebe7 : FAILURE in 14m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7b3ecf6205e2462883c45af3a0386d8a : FAILURE in 14m 21s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ff61f59a28a64f7086fba11af639d4d5 : FAILURE in 14m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/014357249076493ea93ec37715c7094c : FAILURE in 14m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/65dfc09437504fa09a19df410bd8e659 : FAILURE in 14m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f8282757f9ec4c33a211f4f0f442674a : FAILURE in 14m 07s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/67d286f8f8a64c938c9f5ff611e5b413 : FAILURE in 15m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5d5c6ef350dc4c43953552e4bf3e7c79 : FAILURE in 14m 18s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/15fc4b63afac4d019e113b96dac43dfb : FAILURE in 13m 35s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2e11ff9176694a6d802cfcd58e6d2d76 : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"828406ebdaa89f65c3a487a617698d2cfcf7e3c9","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-20 16:05:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bc5ce4c6f41d55f21f6f38cb0f31733418d9a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e809bc7339ab4fbb956b700157bd4c77\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d1af447270bc40c1a7a1da42153ffbee : SUCCESS in 22m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/153310b2690b4232bcc66a35bff042fa : SUCCESS in 23m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/cb512c0f575f48d1b87669bbbfe1c77f : SUCCESS in 23m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f732a1d95d3e494e9ee6c4823444fad5 : SUCCESS in 23m 29s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a4ca309a36474d9ba99932e160dd35b7 : SUCCESS in 23m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/53df08be3fc445ec926556d46a73982a : SUCCESS in 23m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/168dd9e839ed43e4af6d799f1c21d25c : SUCCESS in 24m 35s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/148fd1edfed04538bdd54bdeb66ad10c : SUCCESS in 20m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1d986663db4f4d4d858a0f1f909df6a8 : SUCCESS in 20m 35s\n- loci-nova https://zuul.opendev.org/t/openstack/build/225215a4c2d34b1aaeea5fe6b6747ed9 : SUCCESS in 16m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/6ffdef0685fa4132a2e484581beb4fb4 : SUCCESS in 16m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c124239f4fa14acdb470b58be41512ec : SUCCESS in 15m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/24fca6d82d494a7c8f09efb96008a6e7 : SUCCESS in 13m 37s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/ea94241491b748549a885a265c6c951a : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"997fd7e8db7c509817c3d111369f8772a83ce19e","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 21:29:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8cd81815263ef8e4f88b4252368312a38cccf073","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-09-27 13:46:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4151e9f0eaf16d11292374c46b8a58ec4e46afaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 13:47:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3850461e37e006e83ef587f47cb364526f6efdde","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-09-27 13:48:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7ba452f28668e30ae2bbc45d9daadd00d269fe45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 14:17:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95c5e8da9ffa4d05866b3efd437cf90a\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8f7cc80e6c1f44f6b1aa43a73336469e : SUCCESS in 24m 49s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d56ae64661a24773b6b689e3be7c7d5f : SUCCESS in 23m 27s\n- loci-heat https://zuul.opendev.org/t/openstack/build/72658d340ba049d1a8e5e8d0940e222a : SUCCESS in 22m 03s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4312fef1b6004c86a26dccc65dbaa46e : SUCCESS in 23m 55s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/816d2eca90f64e93898dd27aca6f97c8 : SUCCESS in 28m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b3252446778a47a7822aca4204fb31f0 : SUCCESS in 26m 47s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a25a389f4c5e43268d8c4b5008cb8d94 : SUCCESS in 21m 51s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/964811c192db4da0b57e2e635b59ebbf : SUCCESS in 27m 39s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6c4f4ab8f6734aefa3f7a4d3559137e7 : SUCCESS in 22m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/1b97c52e532949f2ae0800743c75a3d5 : SUCCESS in 21m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2b0962e3423944928941f88ab437fef5 : SUCCESS in 28m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c9fdcbff34cb4b7ca8b65b9d9247952e : SUCCESS in 24m 10s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8e393f6d97124dbeb950c00b8627f108 : SUCCESS in 22m 36s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a4db8904c82b4923991b42ed3f58db2f : SUCCESS in 24m 32s","accounts_in_message":[],"_revision_number":1},{"id":"03d0fe41aaa4a1984019775faf80a4dfecaa0ff7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 14:17:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f156030190c6518b0c7254231ad5ff886d0d9b4f","revisions":{"f156030190c6518b0c7254231ad5ff886d0d9b4f":{"kind":"REWORK","_number":1,"created":"2022-08-17 22:39:25.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/55/853555/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/55/853555/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/853555/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/853555/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/853555/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/55/853555/1"}}},"commit":{"parents":[{"commit":"2537db07aa3a3836cd215281e2fe2aa7923706b0","subject":"Refactor docker image task values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2537db07aa3a3836cd215281e2fe2aa7923706b0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-17 22:39:21.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2022-08-17 22:39:21.000000000","tz":-240},"subject":"Drop cryptography from dependencies","message":"Drop cryptography from dependencies\n\nThere is no need to always install it, projects will usually\nhave it listed in their requirements.txt anyways.\n\nChange-Id: I50b92ea36ab2d4157a8ec9e118a7f49fbd8ede2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f156030190c6518b0c7254231ad5ff886d0d9b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f156030190c6518b0c7254231ad5ff886d0d9b4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[]}
