)]}'
{"id":"openstack%2Floci~858551","triplet_id":"openstack%2Floci~master~I7482e9097a7c0a5b332907c9e5849f2fe48af473","project":"openstack/loci","branch":"master","topic":"dnm_all_fixes","attention_set":{},"removed_from_attention_set":{"14119":{"account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"last_update":"2022-10-06 04:10:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7482e9097a7c0a5b332907c9e5849f2fe48af473","subject":"Update fetch_wheels to add Accept header","status":"MERGED","created":"2022-09-20 16:15:28.000000000","updated":"2022-10-06 04:10:38.000000000","submitted":"2022-10-06 04:10:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858551-dnm_all_fixes","meta_rev_id":"0b8f2f74de922cbb0cc0a178b4406b6928dfd8bb","_number":858551,"virtual_id_number":858551,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 04:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-10-06 03:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-22 15:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-10-06 03:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 16:53:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 15:52:47.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-10-06 03:28:43.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"bc355ba8982000723ac28d6bd13ed3bfdd5b6701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-20 16:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb2c88b04895ee0195c66b7aedea333a285f6499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 16:53:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff4ded82fbf04e9a86549f87d6af2bc1\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/679233e8aa4c483892b7ff62439ed5e0 : SUCCESS in 20m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/013e8266cb774273a34d03439cead800 : SUCCESS in 20m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ac1b7fccc1aa488ab833be2307edcb4a : SUCCESS in 20m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/240704b709a849f3a5126aaaf431aa5a : SUCCESS in 20m 11s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4998dd31a9e6402b8d4c56135b10bcfd : SUCCESS in 21m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5f011719d3944fc887e93bf8482347f7 : SUCCESS in 19m 51s\n- loci-manila https://zuul.opendev.org/t/openstack/build/52ff230b5f1a460e84b481facc261ee9 : SUCCESS in 27m 05s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/a66b89d6b38343d2bf5de63c8b0dfef7 : SUCCESS in 27m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8578ead6d63c41d399ab7432067c7c89 : SUCCESS in 21m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5991331c9ef14e0a8fe4ac45902adf2f : SUCCESS in 29m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a8d8439793a449d7986969b53449f0a2 : SUCCESS in 23m 53s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c821b634f52f42069ab19a16c9c4e542 : SUCCESS in 27m 36s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4e80982f02bc426fbafe5213d179ed61 : SUCCESS in 18m 41s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/dacaa8556b284aed9c99a4697238156a : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"58f0c403989e252ec6bb8b70ab4a57fefda88890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 18:58:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ea981e732bae7878120df8529b0956afa7f9656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 19:24:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d5584b778ca4c0eb3e45e1876e16468\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c7a4b2901b6b4ee6bae3862c4bb75b14 : FAILURE in 15m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/351f631ad37c4db3b46f94bf5d6cf005 : FAILURE in 15m 20s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a574919cf7b74676bcfdacb189ccf8a6 : FAILURE in 15m 16s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/40f51be911864a789b42eee5d2c80ee1 : FAILURE in 16m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/14cced616b8d4edd999a39a80de9f2e9 : FAILURE in 15m 19s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c8b9e9a42d604922921fcc1995c0b338 : FAILURE in 15m 32s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4791fe2e165541be921277b892669ad1 : FAILURE in 15m 22s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/a20b27bb6bf540fc8d097eaad4cd1dcb : FAILURE in 13m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/21af8574e78d4051a0d741e0ed78ec99 : FAILURE in 15m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8e830c4605c048919292790e1be6478e : FAILURE in 15m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7ce0c7befa694e0e9152a7e368bcc33d : FAILURE in 22m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4535b8ccb8f14b689aa8325476178ccc : FAILURE in 16m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/56b241b8c6084ec9bd479c7990ffc14e : SUCCESS in 13m 47s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/0eef64cc25ad4200bd956a98b5cb0b8c : FAILURE in 19m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ee673b3d1fdcfffb6658ac76ab1857187da39992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 19:36:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48a6cbff964ca7522b74f1d7053fc68a41549722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 20:15:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eab4a1c6be54fae9cff28aed453c947\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6faca6d7ef7b42dba5051da1681a0d91 : SUCCESS in 31m 22s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3f0fa500baf644258e9b24981b48d9b0 : SUCCESS in 16m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/58c8be7a13f843479c66c94395454e4d : SUCCESS in 28m 20s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/91768498a8c440f3a4ca85ab71ea738e : SUCCESS in 17m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b8f8d1f7c4894966806677369dfd4558 : SUCCESS in 28m 24s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b480e7c878ff43538fa3bb43a03d29a7 : SUCCESS in 15m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f089e92b2be148768ff668ddcacd0ded : SUCCESS in 16m 16s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d0dabb3e0bed460fb08b21d2cf055508 : SUCCESS in 29m 06s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8fa53cf408f74e1ba546d16391a3e32d : SUCCESS in 29m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0290084981314f0fa2ad993fcb5a69ef : SUCCESS in 30m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/289155ea63a049f5a49cf1be6a205d07 : SUCCESS in 27m 52s\n- loci-placement https://zuul.opendev.org/t/openstack/build/126df9d014a44bf993ec078ca7055472 : SUCCESS in 18m 21s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b914bfecf9e445b5a75ef77b92fb7258 : SUCCESS in 18m 23s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/0721415139e14a108a069ceb1ee98191 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":3},{"id":"a911a7e6b7ffc26af870b46ad972c04b781f23ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 23:30:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"20a2329eadd3f9b4e43b2dbadd97ff1753003382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 23:56:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfc186a19ec84f0e8206ab85a2768181\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/04cac298a998401bbe3619358fe164ae : FAILURE in 10m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0397518387f0409a929e4e034cb9e5e1 : FAILURE in 10m 39s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b016daa134e940f6ad903d394658631f : FAILURE in 12m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f669b44f4cdb44b8a98635df73dddc66 : FAILURE in 12m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f9c78e19695b473f8eb7cfcf9d31c986 : FAILURE in 11m 08s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a1bef37745f94415826e1797c009348d : FAILURE in 10m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5b0c600941e04a8eaae938b2b65b04f8 : FAILURE in 10m 00s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/16d0e1c970ba4a8eadb6766c170c6a5f : FAILURE in 9m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7d860372aaf24590bf95c674852f580f : FAILURE in 10m 38s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5f086d61692d4605b6abf9fc973d8029 : FAILURE in 10m 32s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/84a8d991abc44d8d992f11c55d87160f : FAILURE in 9m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/427dd53354b74f2b9da8e74fff631d1a : FAILURE in 9m 41s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/810e324e90a44756b30d87326a95bf59 : SUCCESS in 7m 19s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2c1c6d69a40649329852ae2a2cff9f5f : FAILURE in 9m 55s","accounts_in_message":[],"_revision_number":4},{"id":"bd7bb220e8148c29a92ec9421fb17327a1be1281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-22 01:28:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1e6a48c8d891e48336707a711a33aa5cd86ba4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 01:49:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68783b9d90b54f0aab4e4380a8b005d0\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/12a5d026d4014e04bdbc252e5f41717c : SUCCESS in 18m 48s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a8fff989e6af46a48a80192563c7ee46 : SUCCESS in 17m 27s\n- loci-heat https://zuul.opendev.org/t/openstack/build/dbdd7be307d141b0aef1fe11711090fd : SUCCESS in 17m 58s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/81d56ea132534e96b6209ce583e79fdf : SUCCESS in 18m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2eca569a80594f23af61acb2700a6cef : SUCCESS in 17m 45s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/52cf45eac15946f3b70867d5f0ef9cdd : SUCCESS in 17m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/63cf1bf3218144f3ac000fc668fab136 : SUCCESS in 17m 33s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/68147ba4045a46d2a532e2b1d9e9555c : SUCCESS in 17m 06s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/141bc6e0bfee4f9580e84e2fa421d65b : SUCCESS in 19m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d4d4737b5f864257bf2f802cc6300e91 : SUCCESS in 19m 11s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/617ca771345c418bb8e0c590a1357757 : SUCCESS in 17m 46s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7cf1cfdcbe0b4adca856b47e21fe9891 : SUCCESS in 17m 20s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6e56535c46d34555844f7e245552dcfa : SUCCESS in 13m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bfc70ca32c5c47f0be55cac2b04bcdc3 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":5},{"id":"df210192b99b7bc0cf3ec784da1f707847ed1dcc","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-09-22 15:52:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"da7732c601efbe13e588b1119e4fae3ef9703f9e","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-10-06 03:28:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"edd1d7b3f6d858bcbff8504ce7aeb3f5ccf961a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 03:29:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"17fb4e28cb65acbb025ccf1bef9248938b883873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 04:10:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f49db3a3a43462a82bb755730266149\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8aea075d73b74b938e6fd590e6d4fc45 : SUCCESS in 33m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/deb59f21dcd1457b835f43f8c17535c0 : SUCCESS in 32m 46s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f28a8b62ea634734a0a2896e3b236b78 : SUCCESS in 25m 37s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f39393e55e424988ab22f0f8a5482835 : SUCCESS in 33m 40s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/776b800696dc4a32b219e33ed1912c4e : SUCCESS in 30m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a26d6cb5765141a19f13584c735daecf : SUCCESS in 30m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/034de9aacb694733beda22733fdcb3ba : SUCCESS in 32m 49s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/9d40f6cfac984871b1a4345c66fb68e1 : SUCCESS in 21m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/eab940ffa79a444ca0ace74ffa55b77c : SUCCESS in 25m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8f1c1e1bad6e449d9ed19e7d5f59e3cf : SUCCESS in 31m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/494eb2c370144894a11e8fc2a4a95271 : SUCCESS in 22m 22s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b0708345f1e74a0a86e26cf1d8091a29 : SUCCESS in 22m 05s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7f5e87ef4bfe45c7ab9f75c36352e817 : SUCCESS in 19m 18s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b4e857cfd6a04f6a9f299628c69a6de7 : SUCCESS in 24m 24s","accounts_in_message":[],"_revision_number":5},{"id":"0b8f2f74de922cbb0cc0a178b4406b6928dfd8bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 04:10:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2e4f6d555a249d81dc141a7f3cdd5f42f2e0f5ae","revisions":{"141c43d9c6f4fd86ae6aae46d6a2aa0ccf182669":{"kind":"REWORK","_number":1,"created":"2022-09-20 16:15:28.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/51/858551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/858551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/858551/1"}}},"commit":{"parents":[{"commit":"2537db07aa3a3836cd215281e2fe2aa7923706b0","subject":"Refactor docker image task values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2537db07aa3a3836cd215281e2fe2aa7923706b0"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:36:46.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 16:15:13.000000000","tz":-240},"subject":"Update fetch_wheels to add Accept header","message":"Update fetch_wheels to add Accept header\n\nA private registry run from docker.io/registry:2 will use the Accept\ncontent type added. This allows private repos to work:\n\n  WHEELS\u003d192.168.1.50:5000/privateloci/requirements:ubuntu\n\nChange-Id: I7482e9097a7c0a5b332907c9e5849f2fe48af473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/141c43d9c6f4fd86ae6aae46d6a2aa0ccf182669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/141c43d9c6f4fd86ae6aae46d6a2aa0ccf182669"}]},"branch":"refs/heads/master"},"debb55449417d84eb6bfaabab277c85e535ff659":{"kind":"REWORK","_number":2,"created":"2022-09-21 18:58:56.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/51/858551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/858551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/858551/2"}}},"commit":{"parents":[{"commit":"2537db07aa3a3836cd215281e2fe2aa7923706b0","subject":"Refactor docker image task values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2537db07aa3a3836cd215281e2fe2aa7923706b0"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:36:46.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 18:58:17.000000000","tz":-240},"subject":"Update fetch_wheels to add Accept header","message":"Update fetch_wheels to add Accept header\n\nA private registry run from docker.io/registry:2 will use the Accept\ncontent type added. This allows private repos to work:\n\n  WHEELS\u003d192.168.1.50:5000/privateloci/requirements:ubuntu\n\nAlso remove the simple distutils function we used ahead of its removal\nin python3.12\n\nChange-Id: I7482e9097a7c0a5b332907c9e5849f2fe48af473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/debb55449417d84eb6bfaabab277c85e535ff659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/debb55449417d84eb6bfaabab277c85e535ff659"}]},"branch":"refs/heads/master"},"512dc82b7a791a3499c2ae2e657aa394acefc63f":{"kind":"REWORK","_number":3,"created":"2022-09-21 19:36:55.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/51/858551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/858551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/858551/3"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:36:46.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:36:42.000000000","tz":-240},"subject":"Update fetch_wheels to add Accept header","message":"Update fetch_wheels to add Accept header\n\nA private registry run from docker.io/registry:2 will use the Accept\ncontent type added. This allows private repos to work:\n\n  WHEELS\u003d192.168.1.50:5000/privateloci/requirements:ubuntu\n\nAlso remove the simple distutils function we used ahead of its removal\nin python3.12\n\nChange-Id: I7482e9097a7c0a5b332907c9e5849f2fe48af473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/512dc82b7a791a3499c2ae2e657aa394acefc63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/512dc82b7a791a3499c2ae2e657aa394acefc63f"}]},"branch":"refs/heads/master"},"a6522a7348bf127af1246b61172d97144fa26ad8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-09-21 23:30:16.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/51/858551/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/858551/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/858551/4"}}},"commit":{"parents":[{"commit":"7c86a8d52b8c08735f96498b679bc50c0eac1ca2","subject":"Fix requirements build failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c86a8d52b8c08735f96498b679bc50c0eac1ca2"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:36:46.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 23:29:57.000000000","tz":-240},"subject":"Update fetch_wheels to add Accept header","message":"Update fetch_wheels to add Accept header\n\nA private registry run from docker.io/registry:2 will use the Accept\ncontent type added. This allows private repos to work:\n\n  WHEELS\u003d192.168.1.50:5000/privateloci/requirements:ubuntu\n\nAlso remove the simple distutils function we used ahead of its removal\nin python3.12\n\nChange-Id: I7482e9097a7c0a5b332907c9e5849f2fe48af473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a6522a7348bf127af1246b61172d97144fa26ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a6522a7348bf127af1246b61172d97144fa26ad8"}]},"branch":"refs/heads/master"},"2e4f6d555a249d81dc141a7f3cdd5f42f2e0f5ae":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-09-22 01:28:27.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/51/858551/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/858551/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/858551/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/858551/5"}}},"commit":{"parents":[{"commit":"4ff506ae5cd88fd551945da3c7f9e0e99f5dd468","subject":"Remove PYTHON3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4ff506ae5cd88fd551945da3c7f9e0e99f5dd468"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:36:46.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-22 01:28:12.000000000","tz":-240},"subject":"Update fetch_wheels to add Accept header","message":"Update fetch_wheels to add Accept header\n\nA private registry run from docker.io/registry:2 will use the Accept\ncontent type added. This allows private repos to work:\n\n  WHEELS\u003d192.168.1.50:5000/privateloci/requirements:ubuntu\n\nAlso remove the simple distutils function we used ahead of its removal\nin python3.12\n\nChange-Id: I7482e9097a7c0a5b332907c9e5849f2fe48af473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e4f6d555a249d81dc141a7f3cdd5f42f2e0f5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e4f6d555a249d81dc141a7f3cdd5f42f2e0f5ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[]}
