)]}'
{"id":"openstack%2Floci~858810","triplet_id":"openstack%2Floci~master~Ia7be53e84090cc8b80e1dd5753ba1460ff454851","project":"openstack/loci","branch":"master","topic":"dnm_all_fixes","attention_set":{},"removed_from_attention_set":{"14119":{"account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"last_update":"2022-10-06 03:59:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7be53e84090cc8b80e1dd5753ba1460ff454851","subject":"Remove PYTHON3","status":"MERGED","created":"2022-09-21 18:52:22.000000000","updated":"2022-10-06 03:59:13.000000000","submitted":"2022-10-06 03:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858810-dnm_all_fixes","meta_rev_id":"6d145a6c298ea6ac9b928904a113a3ecf3399684","_number":858810,"virtual_id_number":858810,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 03:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-10-06 03:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-22 15:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-10-06 03:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 19:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 15:52:35.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-10-06 03:28:44.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"06ed6d4076581d874e20b03e419389cd8863f55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 18:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eafeaaeee1abd03fbffbcf05e42f6869230cf2c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 19:25:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83dcfda044a44a7b9dfe7db439035b0\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a7af2c48574545489c6371cba71b38cc : SUCCESS in 23m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e960820b84004fb7ae0ef3ece7020404 : SUCCESS in 19m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/51791dfbe34b45f6af8fcbfaab32f6c8 : SUCCESS in 27m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3609310986fb4e24a884d3b2e1e82587 : SUCCESS in 26m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/68dc4b1a8965449d8ff12c3b07d6415b : SUCCESS in 26m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7d1088d1df604bad96135ea52fb35062 : SUCCESS in 27m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9ebf41bc727243ac8613cb463b85cc52 : SUCCESS in 28m 49s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/c7559dbb20104059893e279e3497f9a3 : SUCCESS in 27m 04s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c88e5516d1144e709d5a05ad55bebf1e : SUCCESS in 29m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/89fddef72c4445478429b2c660269f68 : SUCCESS in 30m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2412f7304fcb42df8dd07614aa98cf55 : SUCCESS in 26m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/68c9b66751e44103ae69d468dd6b90e2 : SUCCESS in 27m 12s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a272e59f3af0486fb8c7349dada30185 : SUCCESS in 21m 34s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/bd72d91c98d347d08306cf74e189e1f0 : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"094f018a7809bf313056fa8b923362f6de849735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 23:30:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b8ffbe01f7ab10e5bfc3d4223fbd0f01f9c3f2d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 00:05:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4560dada883a4b3e81ccc9d58c43a7b4\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c92795542cba4664b72198d313037141 : SUCCESS in 18m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/39d81d905c154ac7a629204c96ea6b54 : SUCCESS in 22m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a2ec88cdd99642fdb23a435b5a1922a0 : SUCCESS in 25m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9e4d2ef4c949484796d70f3282a72c4c : SUCCESS in 23m 07s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ec4de5d1afab463dba0d2ae9d4239cc9 : SUCCESS in 24m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/44a3a84b2dcc45c5a7280022f950cb8e : SUCCESS in 19m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5f86e34e45bd494b8510a7f36f205eb3 : SUCCESS in 19m 50s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/8f53cd40e93c41dca5413abe888d7d05 : SUCCESS in 20m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/34d37307606f49369912451b891fef0f : SUCCESS in 24m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e54ec32f4b994dfaac49a346c3d54658 : POST_FAILURE in 22m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2d9b9a960c9647cd90c8150acdfc98e9 : SUCCESS in 24m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c85827acb6aa4137861712b262cb7fe0 : SUCCESS in 26m 42s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/07534fbc54564c6ab8f560d9dc88429d : SUCCESS in 20m 25s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/792f43fcfa344d82bbb23ffc7b8900fb : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1a243a099b21e809276ece9245f5138c3cd366f6","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-22 01:21:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"032cc12f25cd0c54ccdedc9bfbdcd1a958856ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-22 01:28:27.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2ebab5a47bc46554636acdc1981a32e6af892cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 01:50:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a0efb5ca46f444cb86b7a5eed9479f1\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4fabf76c57694761b4eac8320cf66fc6 : SUCCESS in 18m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/812ccff77b1d46308e2859cf3da7f66f : SUCCESS in 18m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a05c2f8c6d704b6895c6e457e41c989b : SUCCESS in 18m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7ed53a0a219d43f8a57378ee3f2e39ef : SUCCESS in 18m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/dd5d69d3aaa446f7a9da4817aa438e74 : SUCCESS in 18m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1c996fa55f6240389c9d09275f082a90 : SUCCESS in 18m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/82b579c64bbf4521a7403de491d289a3 : SUCCESS in 18m 53s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/119c2e5538e04e46a6bccaeef302a391 : SUCCESS in 17m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/fdc7ca339afd4517bb97f979cd3e2d42 : SUCCESS in 18m 24s\n- loci-nova https://zuul.opendev.org/t/openstack/build/55a1d10313b045418b90b010db3b75ce : SUCCESS in 20m 22s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/53620ded82024a21838094f386c80b15 : SUCCESS in 17m 46s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f47f27e41c384b188ce7967f041720dc : SUCCESS in 17m 16s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/dfebf648b70a48bdae34cc8b6acfd813 : SUCCESS in 13m 53s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/9da112a326274cea9445e1be17035f2f : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c4143d9d260fcdd08c3ce82ec2e91ca1df0f7b6a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-09-22 15:52:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4cbbccf9f3baad1519c84da4d82b46a59175f562","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-10-06 03:28:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"44d571075917f3397fbbe76c40e0394cecadded2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 03:29:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2bb33a17fdfc5ce332abcbae7ae7e548e5147987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 03:59:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cb5230980c41d8ae62195f1fa6b9c2\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dd3519c3587f4a4987bc052ff796a54e : SUCCESS in 21m 34s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e5fb267bb2064824b024144654d2af66 : SUCCESS in 19m 03s\n- loci-heat https://zuul.opendev.org/t/openstack/build/eb91e543cadc4a4a85c6c6aa21c82e7d : SUCCESS in 19m 19s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0633fc4931464ff59862c38dc44e4c38 : SUCCESS in 18m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/165235b470de49cabf71ae8b40560487 : SUCCESS in 21m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/632fb3c8c5d94e4bbc2ef91ad93d068d : SUCCESS in 17m 51s\n- loci-manila https://zuul.opendev.org/t/openstack/build/80240dd8c5e448edbd26d54f1cba39ce : SUCCESS in 16m 30s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/45358b8b90134197b3a3fc43b52b4e11 : SUCCESS in 17m 40s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b6d1260d55274cdea7a04057ba2dc19d : SUCCESS in 21m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/73fbb0f73890467196a6892494cd6619 : SUCCESS in 22m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4a67090626304c8899864b50334cdc78 : SUCCESS in 17m 38s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c4e45cff261c470998cfcc8e4bfeeee1 : SUCCESS in 19m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7989f78da2f34c0bb8e955f33c320976 : SUCCESS in 13m 40s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3c38888820bc4c00bbe76bcbb370d4cf : SUCCESS in 20m 16s","accounts_in_message":[],"_revision_number":3},{"id":"6d145a6c298ea6ac9b928904a113a3ecf3399684","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 03:59:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ff506ae5cd88fd551945da3c7f9e0e99f5dd468","revisions":{"1ad95860db0ee74525a5d627cb0ecba57fe938e0":{"kind":"REWORK","_number":1,"created":"2022-09-21 18:52:22.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/858810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/858810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/858810/1"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 17:57:59.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 18:51:33.000000000","tz":-240},"subject":"Remove PYTHON3","message":"Remove PYTHON3\n\nThe python2 is long dead and the PYTHON3 flag has been noop for a while.\n\nChange-Id: Ia7be53e84090cc8b80e1dd5753ba1460ff454851\nRelated-Id: I3f4f60756887ad07229cf522803b30d4e16463cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1ad95860db0ee74525a5d627cb0ecba57fe938e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1ad95860db0ee74525a5d627cb0ecba57fe938e0"}]},"branch":"refs/heads/master"},"b7f4e7974f063367bb99a116846364f32466c1e0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-21 23:30:16.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/858810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/858810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/858810/2"}}},"commit":{"parents":[{"commit":"b5e4eafc9549e43f6630edd1ad3101743cc0a0b7","subject":"Do not always install libapache2-mod-oauth2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b5e4eafc9549e43f6630edd1ad3101743cc0a0b7"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 17:57:59.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 23:29:57.000000000","tz":-240},"subject":"Remove PYTHON3","message":"Remove PYTHON3\n\nThe python2 is long dead and the PYTHON3 flag has been noop for a while.\n\nChange-Id: Ia7be53e84090cc8b80e1dd5753ba1460ff454851\nRelated-Id: I3f4f60756887ad07229cf522803b30d4e16463cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b7f4e7974f063367bb99a116846364f32466c1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b7f4e7974f063367bb99a116846364f32466c1e0"}]},"branch":"refs/heads/master"},"4ff506ae5cd88fd551945da3c7f9e0e99f5dd468":{"kind":"NO_CHANGE","_number":3,"created":"2022-09-22 01:28:27.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/10/858810/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/858810/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/858810/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/858810/3"}}},"commit":{"parents":[{"commit":"361fd6e1045a13c793b84335790e30b828eea7c2","subject":"Do not always install libapache2-mod-oauth2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/361fd6e1045a13c793b84335790e30b828eea7c2"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 17:57:59.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-22 01:28:12.000000000","tz":-240},"subject":"Remove PYTHON3","message":"Remove PYTHON3\n\nThe python2 is long dead and the PYTHON3 flag has been noop for a while.\n\nChange-Id: Ia7be53e84090cc8b80e1dd5753ba1460ff454851\nRelated-Id: I3f4f60756887ad07229cf522803b30d4e16463cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4ff506ae5cd88fd551945da3c7f9e0e99f5dd468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4ff506ae5cd88fd551945da3c7f9e0e99f5dd468"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[]}
