)]}'
{"id":"openstack%2Floci~858820","triplet_id":"openstack%2Floci~master~Id92d1d6bd11acfbdd748073931b7b8f63d698368","project":"openstack/loci","branch":"master","topic":"dnm_all_fixes","attention_set":{},"removed_from_attention_set":{"8898":{"account":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"last_update":"2023-07-12 23:08:21.000000000","reason":"Change was abandoned"},"14119":{"account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"last_update":"2023-07-12 23:08:21.000000000","reason":"Change was abandoned"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2023-07-12 23:08:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id92d1d6bd11acfbdd748073931b7b8f63d698368","subject":"Fix requirements build failures","status":"ABANDONED","created":"2022-09-21 21:20:44.000000000","updated":"2023-07-12 23:08:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"49b3f9ea735d0c50498fcafe11394e973c379d25","_number":858820,"virtual_id_number":858820,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-18 18:50:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 21:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 18:17:10.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"CC"},{"updated":"2022-11-18 16:47:07.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"cea080146519a6432ca8e1ebad80c0ed8d71952c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 21:20:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f105d81d435d9357159a364c134013221b6da08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 21:22:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dca884eb9bf941180888b4d6428f0ae87b99634e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 21:40:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5aca4c351d9449eb65bc4549d739fe8\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/79ecce0b744f4927b6393b05d5a12f57 : FAILURE in 11m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7a41b8adf81c420a8b1e6cbca28af940 : FAILURE in 11m 17s\n- loci-heat https://zuul.opendev.org/t/openstack/build/163ef528f11b434ab8f5380aae795d7e : FAILURE in 10m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/460da10a17b94b11af2c372f1de4a28c : FAILURE in 11m 13s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7c7c6dfb43494ef2a777ec5f78e5f458 : FAILURE in 10m 22s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8405892e9c0b4345a101db6db830f4c4 : FAILURE in 8m 50s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2e1359dccdb54f9f80623be0f5117cbd : FAILURE in 14m 39s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6426a4cc168c417189cef755d43282f7 : FAILURE in 8m 35s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b03490ac756246f1b4a37ebcdd04bca7 : FAILURE in 9m 07s\n- loci-nova https://zuul.opendev.org/t/openstack/build/92eadab744eb4588b45ba52c8a912abc : FAILURE in 13m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3e36dad051cb49c18bfacff9e2f42af7 : FAILURE in 9m 23s\n- loci-placement https://zuul.opendev.org/t/openstack/build/62f8586d3186458e9ffe8ba4368d18fd : FAILURE in 16m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/dc5c0e7c215f49538ea2fb0eedc616c5 : SUCCESS in 6m 25s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/07de85f123834e239eef68d2b159b13b : FAILURE in 8m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8ec9a62a7cd7b4cd6e86261483fbecb38b13e83f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 23:21:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f3b3ee1397803fb4b98f0f8c06417768924b1099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 23:30:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0335057c01500d0cbf5d6543253dc60e8b31d069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 23:49:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cacdd203cbd342329a125db894245668\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/67399dce3d5a4003b2faed2cc7860848 : FAILURE in 7m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/307a156199024994955c86421e2841ec : FAILURE in 10m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d1e175854d1b471bbf390bcea135e38a : FAILURE in 9m 47s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5c357e39e84d4bfabe1a15f608608095 : FAILURE in 8m 39s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/494d865b34594df0b36f1544fe7a3558 : FAILURE in 9m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/88c9e91426c84727bb56ad448ccb959a : FAILURE in 10m 23s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4e6cba88ea50464fa23eb810a010ddd4 : FAILURE in 10m 33s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/9ad98d956b3e4d24a4878cbb65270d38 : FAILURE in 7m 26s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4ebd123a544f450db6050dcce36326ac : FAILURE in 13m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d7aa79aec7b34225aad6391a76484c14 : FAILURE in 9m 31s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d3cd487d2c2c4eedb12561abafef8717 : FAILURE in 9m 21s\n- loci-placement https://zuul.opendev.org/t/openstack/build/230cd1df2d864ef0b62ce498a2fe24fc : FAILURE in 9m 16s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7137c29297de493eab22e8e803d66821 : SUCCESS in 6m 16s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/4b6894ffaaaa47d2b3b4706e09579fd4 : FAILURE in 9m 55s","accounts_in_message":[],"_revision_number":4},{"id":"a7733d161559c391365323d298393a9b3742e189","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-22 01:28:27.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8681996ab955773a6b2f298fe2fa1d3987a4768a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 01:45:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7aa74242ddd4529b3af16e0f06eff4d\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c60f025ca0ac46329e18f5efb11e7a78 : FAILURE in 11m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/553b209cf3b548e3bd08126f0ca7d083 : FAILURE in 9m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c4f0b660ae8f4e488da5af2021cbd8f8 : FAILURE in 10m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2b094f1728f6481db143560058bc277b : FAILURE in 9m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9675434c56b64dabb0e84c8db003c24f : FAILURE in 9m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4cadc93c04d8498a91c43028ea36b8bb : FAILURE in 10m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/01ca600a0b774ebe9ef9e518a1271174 : FAILURE in 9m 57s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d4a277fad53d47d187d0fac4a92edf36 : FAILURE in 9m 05s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cfb9a4699355441d9201a5c47bf9f9d4 : FAILURE in 10m 43s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5b2fc6ce6d44402eba4d3bd0736493ce : FAILURE in 9m 29s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/be5580e00e314988b2a323fc8946a4db : FAILURE in 9m 53s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1531edff5cc245fbbb3ca45582e91070 : FAILURE in 10m 15s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4aa2668d4e8a4105a25b24a1c192c150 : SUCCESS in 6m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fa715b0f72a048b4bacc87bec6367242 : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":5},{"id":"266d56657ca1a13d0fc25de3f343f1145e9c23db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-06 17:53:11.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3e9d09e24c6b0f3dbf845f4806de3da6fb50f495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-06 18:18:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1daf95fee8ac4934bcebb0e0d2b959ad\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1e1cdec39a384af5bb2eab53eef37e9e : FAILURE in 7m 52s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c7ced6467aeb4e40bf70569e7f2664d1 : FAILURE in 8m 12s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5d0fbbce673e4e23b45030226307d34a : FAILURE in 7m 35s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3e43f5c791ad4137b77051bf8232c9be : FAILURE in 7m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3e3e566c83fc4417abded24e5259e00c : FAILURE in 7m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4663a9dae080438ead8e144a3d7d8fbe : FAILURE in 7m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ef3de246cf554615b1fcb1c82b145460 : FAILURE in 10m 02s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/70ae70a6355b4dc683d4b8e5f3604e39 : FAILURE in 7m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/fecfc4fe24ba427b954db2cf27e458fe : FAILURE in 10m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b299b0d9627f436faa004f3bffcb2461 : FAILURE in 7m 11s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5e7abffce7924c0fafe09ee83ed86aa8 : FAILURE in 8m 27s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e2f23062d91f4be983dc717427e96b78 : FAILURE in 7m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/dabe995aac18444f80a92cceff0b644c : SUCCESS in 7m 28s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8fecbf2060ca4f6e9928272c0b7724b1 : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":6},{"id":"8a5d4dc8b877d8a6589dbb513e1c8daa7fd15c8a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-08 18:17:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"60f34c1392dfda73fd486681a399fc9ea07ec721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 18:34:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0023bb80e7f340d8866189a0aca2eae6\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1fcd99c9d6ca4edc998ad9abde882263 : FAILURE in 7m 34s\n- loci-glance https://zuul.opendev.org/t/openstack/build/45d49ab328334bd5bfc71de0fddebc1f : FAILURE in 10m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/42a7fdc3790749dc8e6994379d1b3692 : FAILURE in 7m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8b98f8e167904529a7fd6941a7cdc196 : FAILURE in 10m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b61c4a46c920423083176ef147327234 : FAILURE in 7m 15s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fd01688e4dba4d42b166abb3a407a880 : FAILURE in 10m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b2f50d7213114e58aa26b842cf07f5e0 : FAILURE in 7m 27s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/03dfbf6f22bb46b99193635daff28cfe : FAILURE in 7m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/781d098c518c466d90c2bfddf5539959 : FAILURE in 11m 08s\n- loci-nova https://zuul.opendev.org/t/openstack/build/be24a80fbfbd4a71bc4dd3bfe6e3a2bd : FAILURE in 6m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1ec2490583bd4874be8301464252204c : FAILURE in 9m 49s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b4152af82e09481b8ce9f9e5c8a3a11f : FAILURE in 9m 42s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/132251a296ce40ada2343a06e8c5eaa2 : SUCCESS in 5m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7c9b76f289394fce89b67081407165b0 : FAILURE in 9m 43s","accounts_in_message":[],"_revision_number":6},{"id":"7be054647fb413560bcdf13f79969712b8b8e750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-18 16:46:51.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"72d9b3e4169a0771205cfc8171c3430b67dd9d06","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-18 16:47:07.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"41974ab03c36e8d262478ba82e70d6815684fb07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 17:10:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7244ad0e39048d9865681e6471fcaca\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c816186f70214ab6bc4c261ecc666892 : FAILURE in 9m 26s\n- loci-glance https://zuul.opendev.org/t/openstack/build/33b0c153a5654472be4453946cf42cf7 : FAILURE in 11m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/16f557cf5ffd4855a23303fffd30a9b6 : FAILURE in 12m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/96bfe6ab4eb74100b23c20926529a47c : FAILURE in 11m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f4518f2510934e44a69ae8a1e0f6b146 : FAILURE in 10m 14s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9a07ab487f7a4817a10218563c3d261c : FAILURE in 11m 34s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4aae2d09cb9743649045b39061f09caa : FAILURE in 9m 50s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f68ef22f67434d9babf26d8a2802fcdf : FAILURE in 8m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/da9d32cad57447b1a61aadf23901bd48 : FAILURE in 10m 07s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6624e7a6a8ab43128d1ced1bad02d0ae : FAILURE in 8m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/de20b411703047fca9cf05bd03f72694 : FAILURE in 9m 25s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b4ac9326d84f48aa929d7f752e72b4ac : FAILURE in 8m 48s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/fadc8b84d87941d5ac6b80de4932c3db : SUCCESS in 7m 40s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f5f6c362e7864947b55a10700b8f93da : FAILURE in 10m 40s","accounts_in_message":[],"_revision_number":7},{"id":"8c02c8910e69884d1d7b7d0642609615c40928ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-19 19:44:55.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"0457567ee37e714054a14dc90b1d4b01edfc860a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-19 20:02:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7da4f5d7f01147d2a48a5ea7fdcf89cd\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f11ec29472d042deac5a6ac8206d5584 : FAILURE in 9m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b44bd593c3104647bf5a4fc8fcca5b07 : FAILURE in 8m 39s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3992c340fc244f6a959b4e32cc43c0f3 : FAILURE in 9m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9f06a01aaf404c17a04b33fc9fba551b : FAILURE in 9m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8751d539d3ca4270b7d80cad814d2152 : FAILURE in 10m 11s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3c73b8f36a4b46558aee11613cbc77d6 : FAILURE in 9m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c9e4742257014e45b3906d7ad6f56f63 : FAILURE in 8m 45s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/3efa88bcdcd54b7bb9db4a16d1f66570 : FAILURE in 8m 30s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c18c375747f34d628a39979ea17b78ab : FAILURE in 9m 26s\n- loci-nova https://zuul.opendev.org/t/openstack/build/27f7beb0afa548b1b0abc6b9aa1023ff : FAILURE in 8m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5fdd6eb849fc44d5a8f85a2d22b8f5cf : FAILURE in 8m 43s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d2dcdf68a5c44956ad9401c6598dd2ad : FAILURE in 11m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a2502394c1234e2c899cd93a9443c53d : SUCCESS in 6m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2aa20ea5790e462990c1f6c6f8ea5289 : FAILURE in 8m 28s","accounts_in_message":[],"_revision_number":8},{"id":"8d0635a611c66d11ef90a77c21ae4f9b2b9dc63b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-22 22:19:17.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ed54743ca5281a934281d362251bdd962f0490f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 22:31:36.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/115a12e1f618430faf911ec75f641601\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/716615844e8946c9a625f713d4f14eb9 : FAILURE in 7m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c1dcdeb929f84d74a67adbde953f01a8 : FAILURE in 9m 10s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e76f087278ee421897c5726330d97aa8 : FAILURE in 7m 28s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8e670b11a8e047e892aa0e965afd6027 : FAILURE in 7m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1fff3013200741d98150e3400a61cca1 : FAILURE in 7m 57s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/65a642c68def461da993101653dfcb5f : FAILURE in 8m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/fff93f93bab14c238fcd89539286bfa7 : FAILURE in 6m 19s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6a020a2b9e39428e8cb2d4ded6a01bc9 : FAILURE in 7m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8de825111dee483fb311d915eb8cb291 : FAILURE in 8m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7f9c0fc3c4ff4f98bfa095ab03f41d56 : FAILURE in 7m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8f5b65aa30524ae28536411ff4d0c438 : FAILURE in 8m 07s\n- loci-placement https://zuul.opendev.org/t/openstack/build/eb624dc27ae84c1ab369a5d93064eda7 : FAILURE in 7m 40s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/448b976279e74f8fb974d50961c6e8ea : SUCCESS in 5m 26s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/89deac455e32429e86a12989f515610e : FAILURE in 6m 56s","accounts_in_message":[],"_revision_number":9},{"id":"825b03affd52c3ee573e0003f3182d95a1058b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-12-18 18:32:36.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"2360e0acd5f6c01deca994c7ae6c538a10390e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-18 18:50:22.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93345530d05645e48f8476af47a194b6\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/315014c7201949c7a3d8399ea85a918a : FAILURE in 9m 26s\n- loci-glance https://zuul.opendev.org/t/openstack/build/288c1441c6d34d1794e04430988c9072 : FAILURE in 7m 43s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1c859f2262bd46bf9b8ee1092012acc8 : FAILURE in 8m 20s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d88cefd2c41342ca882785dd33c7d3d6 : FAILURE in 8m 46s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/537e709b8be24539a941caca854d20cc : FAILURE in 7m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/053711ea14bf4164b4e50df946c45368 : FAILURE in 7m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/11684ebf903845df94aefa6c43cbf451 : FAILURE in 8m 07s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d05e45f714954eafb31285dc44337b28 : FAILURE in 7m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2ae80028b0bc4734b1f5ccfd1f5f8ea2 : FAILURE in 8m 36s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c3aa0c59480644cbbec2a62b0fa0d23a : FAILURE in 7m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/652df793d94943cfa01c05b27dbce642 : FAILURE in 7m 27s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cff571dfd55d4c4292b430f51c7e32d0 : FAILURE in 8m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/679e02732f774b6d82dc01e361eab37a : SUCCESS in 5m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8806cc535b1240168185be86142d021d : FAILURE in 7m 53s","accounts_in_message":[],"_revision_number":10},{"id":"49b3f9ea735d0c50498fcafe11394e973c379d25","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2023-07-12 23:08:21.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 206 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8beb554ad2a8c74ad24f66ddcb135ffcaa4bc5ba","revisions":{"f3fe0197d7d88e01a703ca5e6341f3e8f111b173":{"kind":"REWORK","_number":1,"created":"2022-09-21 21:20:44.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/20/858820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/1"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 21:19:30.000000000","tz":-240},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for arm64 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f3fe0197d7d88e01a703ca5e6341f3e8f111b173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f3fe0197d7d88e01a703ca5e6341f3e8f111b173"}]},"branch":"refs/heads/master"},"30b8f3bcbfa7794915cfd5ae3253140e369203fd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-21 21:22:54.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/20/858820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/2"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 21:21:39.000000000","tz":-240},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/30b8f3bcbfa7794915cfd5ae3253140e369203fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/30b8f3bcbfa7794915cfd5ae3253140e369203fd"}]},"branch":"refs/heads/master"},"8b6493fef77bc667740af51f45d780832863aa1e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-21 23:21:53.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/20/858820/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/3"}}},"commit":{"parents":[{"commit":"42a58623f6ebf09085e35a939cba68b065a81971","subject":"Redo venv setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/42a58623f6ebf09085e35a939cba68b065a81971"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 23:21:37.000000000","tz":-240},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8b6493fef77bc667740af51f45d780832863aa1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8b6493fef77bc667740af51f45d780832863aa1e"}]},"branch":"refs/heads/master"},"7c86a8d52b8c08735f96498b679bc50c0eac1ca2":{"kind":"REWORK","_number":4,"created":"2022-09-21 23:30:16.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/20/858820/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/4"}}},"commit":{"parents":[{"commit":"aca325bd09e4ca51fff1d6bde7d2c08a09dcdfe2","subject":"Redo venv setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/aca325bd09e4ca51fff1d6bde7d2c08a09dcdfe2"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 23:29:57.000000000","tz":-240},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c86a8d52b8c08735f96498b679bc50c0eac1ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c86a8d52b8c08735f96498b679bc50c0eac1ca2"}]},"branch":"refs/heads/master"},"3b60213769173b336333e440cf8bc218f33cd50b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-09-22 01:28:27.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/20/858820/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/5"}}},"commit":{"parents":[{"commit":"2e4f6d555a249d81dc141a7f3cdd5f42f2e0f5ae","subject":"Update fetch_wheels to add Accept header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e4f6d555a249d81dc141a7f3cdd5f42f2e0f5ae"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-22 01:28:12.000000000","tz":-240},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b60213769173b336333e440cf8bc218f33cd50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b60213769173b336333e440cf8bc218f33cd50b"}]},"branch":"refs/heads/master"},"4b31248d922ffb07934ddbdacb7eca145eb5d0a3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-11-06 17:53:11.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/20/858820/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/6"}}},"commit":{"parents":[{"commit":"5688c6ffa366f268782ea53add4b7630449eee8a","subject":"Merge \"Update fetch_wheels to add Accept header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5688c6ffa366f268782ea53add4b7630449eee8a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-06 17:53:11.000000000","tz":0},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4b31248d922ffb07934ddbdacb7eca145eb5d0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4b31248d922ffb07934ddbdacb7eca145eb5d0a3"}]},"branch":"refs/heads/master","description":"Rebase"},"f1644e5328dc0b767d53056f696c273aae40c5da":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-11-18 16:46:51.000000000","uploader":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"ref":"refs/changes/20/858820/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/7"}}},"commit":{"parents":[{"commit":"7dd55124c3d4a1368d027575332eb0e045b426ce","subject":"Add cryptography back for placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7dd55124c3d4a1368d027575332eb0e045b426ce"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Chris Wedgwood","email":"cw@f00f.org","date":"2022-11-18 16:46:51.000000000","tz":0},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f1644e5328dc0b767d53056f696c273aae40c5da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f1644e5328dc0b767d53056f696c273aae40c5da"}]},"branch":"refs/heads/master","description":"Rebase"},"4d14cbd458329736fae974840258f7f1c1e4bb43":{"kind":"REWORK","_number":8,"created":"2022-11-19 19:44:55.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/20/858820/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/8"}}},"commit":{"parents":[{"commit":"7dd55124c3d4a1368d027575332eb0e045b426ce","subject":"Add cryptography back for placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7dd55124c3d4a1368d027575332eb0e045b426ce"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-19 19:44:54.000000000","tz":0},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4d14cbd458329736fae974840258f7f1c1e4bb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4d14cbd458329736fae974840258f7f1c1e4bb43"}]},"branch":"refs/heads/master"},"06c6b8652875c88e06a8f265a711fd4185c29ea6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-11-22 22:19:17.000000000","uploader":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"ref":"refs/changes/20/858820/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/9"}}},"commit":{"parents":[{"commit":"39aaeb3f329993046703233d2f52c9f7470a8de1","subject":"Remove uwsgi by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39aaeb3f329993046703233d2f52c9f7470a8de1"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Chris Wedgwood","email":"cw@f00f.org","date":"2022-11-22 22:19:17.000000000","tz":0},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06c6b8652875c88e06a8f265a711fd4185c29ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06c6b8652875c88e06a8f265a711fd4185c29ea6"}]},"branch":"refs/heads/master","description":"Rebase"},"8beb554ad2a8c74ad24f66ddcb135ffcaa4bc5ba":{"kind":"NO_CHANGE","_number":10,"created":"2022-12-18 18:32:36.000000000","uploader":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"ref":"refs/changes/20/858820/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/858820/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/858820/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/858820/10"}}},"commit":{"parents":[{"commit":"5806885d68c6e15251168183245976a26c0da81a","subject":"Merge \"Drop UWSGI from always being installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5806885d68c6e15251168183245976a26c0da81a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 19:20:51.000000000","tz":-240},"committer":{"name":"Chris Wedgwood","email":"cw@f00f.org","date":"2022-12-18 18:32:36.000000000","tz":0},"subject":"Fix requirements build failures","message":"Fix requirements build failures\n\nChange contains the following cleanup:\n  * remove build variable for cassandra-driver\n    - this is a whl in pypi for arm64 and aarch64 since Xena\n  * allow scipy packages\n    - these packages are whl in pypi for amd4 and aarch64 since Wallaby\n  * remove suds-jurko and anyjson\n    - these packages use 2to3 which is now removed\n  * remove python-nss\n    - this breaks with python3.10 and libnss\u003e\u003d3.58\n  * remove enum-compat\n    - this was a python2/3 compat lib. It is noop on python\u003e\u003d3.4\n  * readd python-qpid-proton\n    - this has pypi whls for amd64 and aarch64 since Victoria\n\nChange-Id: Id92d1d6bd11acfbdd748073931b7b8f63d698368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8beb554ad2a8c74ad24f66ddcb135ffcaa4bc5ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8beb554ad2a8c74ad24f66ddcb135ffcaa4bc5ba"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
