)]}'
{"id":"openstack%2Floci~858829","triplet_id":"openstack%2Floci~master~I5ae0c6340e40a21c31a244839830a7c519eee3ec","project":"openstack/loci","branch":"master","topic":"remove_virtualenv","attention_set":{},"removed_from_attention_set":{"14119":{"account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"last_update":"2022-09-21 22:32:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5ae0c6340e40a21c31a244839830a7c519eee3ec","subject":"Redo venv setup","status":"ABANDONED","created":"2022-09-21 22:10:55.000000000","updated":"2022-09-21 22:32:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bbabcde7470cfcfc3ebd0d0bfbcfb4c407e517df","_number":858829,"virtual_id_number":858829,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 22:17:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e2fc4171c7cdefe27a73449210010d354411c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 22:10:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3961b197142773ae1017741d4e4a31166c71992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 22:17:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8776e0f42fdc4451ba7ba8f836b48570\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ed7f4ac4af7c409b893679b2d0b080fc : FAILURE in 5m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4f7471d5736840499e5e354f04e258ea : FAILURE in 4m 04s\n- loci-heat https://zuul.opendev.org/t/openstack/build/44681f9e8c6c4b65be90be77d3dae490 : FAILURE in 5m 15s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7be4b48966324b7a933550ea3759b383 : FAILURE in 4m 54s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ca1c540f604e4745a2a13753b215cddc : FAILURE in 5m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0f20a6fe74664a73b9d09f1d3f1f834d : FAILURE in 4m 00s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8b6b0e164a3647c7a7bb0af36b041610 : FAILURE in 4m 47s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/97bf6586b6be464ba4dad02a9d97c16a : FAILURE in 3m 56s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e10124e10dd44da3bf16d9b7f89ce0ef : FAILURE in 4m 56s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d2bb8b55bf7c4017880ef8093a0931e6 : FAILURE in 4m 10s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/378c13fc1ca9402fbe6eadcaa622669c : FAILURE in 5m 13s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1e03a766a8ac4e49b3c6668a37c403f4 : FAILURE in 3m 53s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0a4b405f4556478e8213b8e296979fc4 : FAILURE in 4m 42s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b760c3090fbb4f458173881ed353ae6a : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c11d2e8569b4458234721d71ce563bbb2d4d77f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 22:22:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"753f41e897a02ef1b4e35a6b11f635963b749d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 22:29:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bbabcde7470cfcfc3ebd0d0bfbcfb4c407e517df","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 22:32:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"32a7849794d3fcd4cd2efb91fbbb1f298a4ca9a0","revisions":{"80efe515ef1179f485e4b52859a670e572d5b5fb":{"kind":"REWORK","_number":1,"created":"2022-09-21 22:10:55.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/29/858829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/858829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/858829/1"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:40:32.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 22:10:32.000000000","tz":-240},"subject":"Redo venv setup","message":"Redo venv setup\n\nInstalling `virtualenv` also pulls in all the build tooling (including a\ncompiler). We can use venv with `--upgrade-deps` to download the latest\npip and setuptools, quickly followed by wheel which we leverage heavily.\n\nAdditionally, using `--symlinks` we ensure we dont copy the python3\nbinary and the venv points directly to the host. No PIP_BOOTSTRAP\nanymore.\n\nThe `--upgrade-deps` option is only available in python\u003e\u003d3.9\n\nChange-Id: I5ae0c6340e40a21c31a244839830a7c519eee3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/80efe515ef1179f485e4b52859a670e572d5b5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/80efe515ef1179f485e4b52859a670e572d5b5fb"}]},"branch":"refs/heads/master"},"b2ddcd723340c0f814b6b16c7eb8593c35a35375":{"kind":"REWORK","_number":2,"created":"2022-09-21 22:22:30.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/29/858829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/858829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/858829/2"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:40:32.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 22:22:10.000000000","tz":-240},"subject":"Redo venv setup","message":"Redo venv setup\n\nInstalling `virtualenv` also pulls in all the build tooling (including a\ncompiler). We can use venv with `--upgrade-deps` to download the latest\npip and setuptools, quickly followed by wheel which we leverage heavily.\n\nAdditionally, using `--symlinks` we ensure we dont copy the python3\nbinary and the venv points directly to the host. No PIP_BOOTSTRAP\nanymore.\n\nThe `--upgrade-deps` option is only available in python\u003e\u003d3.9\n\nChange-Id: I5ae0c6340e40a21c31a244839830a7c519eee3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b2ddcd723340c0f814b6b16c7eb8593c35a35375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b2ddcd723340c0f814b6b16c7eb8593c35a35375"}]},"branch":"refs/heads/master"},"32a7849794d3fcd4cd2efb91fbbb1f298a4ca9a0":{"kind":"REWORK","_number":3,"created":"2022-09-21 22:29:34.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/29/858829/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/858829/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/858829/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/858829/3"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:40:32.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 22:29:20.000000000","tz":-240},"subject":"Redo venv setup","message":"Redo venv setup\n\nInstalling `virtualenv` also pulls in all the build tooling (including a\ncompiler). We can use venv with `--upgrade-deps` to download the latest\npip and setuptools, quickly followed by wheel which we leverage heavily.\n\nAdditionally, using `--symlinks` we ensure we dont copy the python3\nbinary and the venv points directly to the host. No PIP_BOOTSTRAP\nanymore.\n\nThe `--upgrade-deps` option is only available in python\u003e\u003d3.9\n\nChange-Id: I5ae0c6340e40a21c31a244839830a7c519eee3ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/32a7849794d3fcd4cd2efb91fbbb1f298a4ca9a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/32a7849794d3fcd4cd2efb91fbbb1f298a4ca9a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
