)]}'
{"id":"openstack%2Floci~858830","triplet_id":"openstack%2Floci~master~I267875117d8e4f18eef0bfc21b6122fe37b02dbd","project":"openstack/loci","branch":"master","topic":"dnm_all_fixes","attention_set":{},"removed_from_attention_set":{"14119":{"account":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"last_update":"2022-11-10 17:53:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I267875117d8e4f18eef0bfc21b6122fe37b02dbd","subject":"DNM","status":"ABANDONED","created":"2022-09-21 22:25:03.000000000","updated":"2022-11-10 17:53:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7b15bfc1a54faefb870458192ac8ffe72c4d85c","_number":858830,"virtual_id_number":858830,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-21 23:16:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 22:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a06d9165fb1668171f2ce71e3abd4f54fb053a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 22:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c61f50ad85d2ec20e91eb2981fba0c0c1d46dd1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 22:30:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30326f5f965051bfced647c41ff5a3d3ebb79b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 22:39:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9323ab13fb7b4a2c9a3295038ca94bba\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ac748738710f4bd88deabaf37a1a39c3 : FAILURE in 4m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/50ba183b4955455ba2a7fe9ffa2901ab : FAILURE in 4m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0764afb3f0984f10af37ea2db194b942 : FAILURE in 3m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9e1fdbad73694d8498131ba485c67925 : FAILURE in 4m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/877ea6f73d334f0e9a7b73235adf1b0c : FAILURE in 5m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/69fb03a0ce314ea4b32332af1d3c718a : FAILURE in 3m 49s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1c69897d5396401480df191cb11672e6 : FAILURE in 4m 50s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f5065b319d7f49359c8ae6aac8367400 : FAILURE in 3m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/16b2a655258441b38652bfff7575d834 : FAILURE in 4m 56s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bdff781f30774c478f301d85789856be : FAILURE in 3m 54s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0551dad066524bb48b7cb26ff5c90fcc : FAILURE in 4m 51s\n- loci-placement https://zuul.opendev.org/t/openstack/build/23427364a1904424ad5984f8a80bc148 : FAILURE in 3m 40s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3b0535e895fd4aeaa4b5122133a69228 : FAILURE in 6m 53s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/54d89625204844eb9918eb7dd478b031 : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"cff232aa2b7f5ba6fcd11ac54784cb2112827f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2022-09-21 23:00:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b004cbc0c4126c2694aa06413871a8cd48945e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 23:16:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a5d28b4b0dd45cea60af669eb32c864\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7f05f7dea7c948229e9355ae54eed522 : FAILURE in 8m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/42610e95d53c4e9a9dd81b9fe4d27c8e : FAILURE in 8m 04s\n- loci-heat https://zuul.opendev.org/t/openstack/build/43ef677a130a40ad8d47465666bf5a30 : FAILURE in 8m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6e979efcc17d40b0bc74fb07f96d515a : FAILURE in 9m 14s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/83a16492532f46bbbcabd0942ad8884d : FAILURE in 7m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/26d3db6d32954dbebb793997a0be37dc : FAILURE in 7m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5b787376e0bc4240be02734cef17039c : FAILURE in 7m 45s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6a9ca32128114a2e9944dcd735c1040d : FAILURE in 8m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e50d416b1b904d7a862d8031937c9af8 : FAILURE in 9m 17s\n- loci-nova https://zuul.opendev.org/t/openstack/build/044d1a1f5226407799527e0240005e5e : FAILURE in 7m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1ea095bae4cb4797b2ac19c535639ea8 : FAILURE in 8m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d471f8b74d9845f7bfdce9b64fe7f384 : FAILURE in 7m 52s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/2112298d9ebd4381bc99d1b3b672235a : SUCCESS in 5m 35s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fbf78fbb779e4b7c91e9d7f9377f9588 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c7b15bfc1a54faefb870458192ac8ffe72c4d85c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-10 17:53:38.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 49 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"719ef4ec90871303234990a50d5370be6bc9cdd9","revisions":{"7d81d6c76697c34729e745435398348976642756":{"kind":"REWORK","_number":1,"created":"2022-09-21 22:25:03.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/30/858830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/30/858830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/30/858830/1"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:40:32.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 22:24:40.000000000","tz":-240},"subject":"DNM","message":"DNM\n\ntesting gate\n\nChange-Id: I267875117d8e4f18eef0bfc21b6122fe37b02dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7d81d6c76697c34729e745435398348976642756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7d81d6c76697c34729e745435398348976642756"}]},"branch":"refs/heads/master"},"3b65e23569a41c042864d0f2869f3378a2dd6b39":{"kind":"REWORK","_number":2,"created":"2022-09-21 22:30:56.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/30/858830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/30/858830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/30/858830/2"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:40:32.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 22:30:35.000000000","tz":-240},"subject":"DNM","message":"DNM\n\ntesting gate\n\nChange-Id: I267875117d8e4f18eef0bfc21b6122fe37b02dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b65e23569a41c042864d0f2869f3378a2dd6b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b65e23569a41c042864d0f2869f3378a2dd6b39"}]},"branch":"refs/heads/master"},"719ef4ec90871303234990a50d5370be6bc9cdd9":{"kind":"REWORK","_number":3,"created":"2022-09-21 23:00:12.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/30/858830/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/30/858830/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/30/858830/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/30/858830/3"}}},"commit":{"parents":[{"commit":"1f2b4cb18f89716ae655cc2cd29f6864eef4136a","subject":"Merge \"Remove suse references from bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f2b4cb18f89716ae655cc2cd29f6864eef4136a"}]}],"author":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-20 15:40:32.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam@yaple.net","date":"2022-09-21 22:59:55.000000000","tz":-240},"subject":"DNM","message":"DNM\n\ntesting gate\n\nChange-Id: I267875117d8e4f18eef0bfc21b6122fe37b02dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/719ef4ec90871303234990a50d5370be6bc9cdd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/719ef4ec90871303234990a50d5370be6bc9cdd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
