)]}'
{"id":"openstack%2Floci~864025","triplet_id":"openstack%2Floci~master~I4e7d8c81d3793d4d3b85b450ded673ad122dc266","project":"openstack/loci","branch":"master","topic":"cryptography","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-11-08 18:15:03.000000000","reason":"\u003cGERRIT_ACCOUNT_21420\u003e replied on the change","reason_account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}},"hashtags":[],"change_id":"I4e7d8c81d3793d4d3b85b450ded673ad122dc266","subject":"Add cryptography back for placement","status":"MERGED","created":"2022-11-08 16:50:27.000000000","updated":"2022-11-08 19:23:02.000000000","submitted":"2022-11-08 19:23:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864025-cryptography","meta_rev_id":"7af0ba0fc03258638ddd34529fbef21d631e2067","_number":864025,"virtual_id_number":864025,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-08 19:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-11-08 17:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-08 16:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2022-11-08 17:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 16:57:31.000000000","updated_by":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"reviewer":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"state":"REVIEWER"},{"updated":"2022-11-08 17:20:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 17:32:45.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"253fbda4495963877ad9dde3459d0e7645b64f45","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-08 16:50:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb29e4dd33440cdea5be233c32f245445e2bc83","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-08 16:55:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"399f131afd791f60aa344d9c9a1e00843e825a7c","author":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"date":"2022-11-08 16:57:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22e01a178f96de9143ec9305b4394c0d33c9260d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 17:20:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc9e9f96282e4cbe88220a6e9c571bbd\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3735c5f90bc445fa8bc304a0272832e2 : SUCCESS in 20m 05s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8b926e3ca90b40939f9dc67050cd1b5b : SUCCESS in 20m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0c55e9dd840d4b7d8273e621517a88fa : SUCCESS in 24m 35s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/95141e81d5c647b7ba1aa62a1c09dae7 : SUCCESS in 19m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1bc618c4380b45098f5a6bcce084559e : SUCCESS in 19m 43s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d35cc403a4ea40c1ae379037f0fbb902 : SUCCESS in 17m 21s\n- loci-manila https://zuul.opendev.org/t/openstack/build/919d2b657a0145e5b021666eeb29d963 : SUCCESS in 17m 07s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/12f2ffca1e7145789d087651f5e38fd7 : SUCCESS in 15m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/185199c399324d73a4ea685a9ca65049 : SUCCESS in 18m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/73eb86a6de4c4ad4a72554fed424a2e3 : SUCCESS in 18m 25s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/27df6ea2bb944d0f9a1d1f6a2d556e1b : SUCCESS in 19m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8b0be043a89642f19799fd5e83c51f6e : SUCCESS in 18m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d19767e79b6742b48d32338529260470 : SUCCESS in 12m 34s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2b72658021694d9e8117210c9a9fb26a : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ea30be9ae2b1082b034c131108e9ff8ee7867faa","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-08 17:32:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb4e9fb188197494450f4cffb32fe486bb0b4132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 17:33:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25f5a739eeae2d5a3093ab768796e99b19601e6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 18:06:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3cc04fe2680419eb74c6d3b123f86fa\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6b78d75570234683a01918b465236fef : SUCCESS in 15m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a4f4e78c7977484d88889469b87e524a : SUCCESS in 16m 15s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b4a6cf55a4994062a226d245f6da1b7d : SUCCESS in 21m 42s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/266bced4344e42ff8ce5572bbd9b2bb8 : SUCCESS in 17m 13s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6550cdfd1ea04a66ae75db717e153974 : SUCCESS in 19m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3a77a9b0af7c46b78fe11ba9d4e14ea4 : FAILURE in 15m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5eb0deb40bf140c6a60853bb2bc58bd1 : SUCCESS in 19m 05s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b842cdf9ecf64b5dad96e4eedbcf95a3 : FAILURE in 17m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a0d6ed132bf54f2bb1ca3b7cbcf469e6 : SUCCESS in 17m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d3982c49f0504e1c9ab33bc9cc98d53e : SUCCESS in 23m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/32d985d890774d6b85dd85b50a1e0569 : SUCCESS in 20m 00s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0199bdb5be7a4510a590c9c411339604 : SUCCESS in 15m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d7470f02742641168e726bc9d55e07c2 : FAILURE in 16m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3cf75608862645dca9f280a907f9bfee : FAILURE in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6684f00d0061092ac2244014544e0c58adfa8f19","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-08 18:15:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1d9f9fffe7fa4598e37293914ca2cd6254c1241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 18:46:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98cd68466a040078bc0930da94a93c4\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/98c7500e535f4ea092694c31df7e84d1 : SUCCESS in 27m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9c25c9094f8947e1a096b500d46b3b77 : SUCCESS in 17m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ec6654be056f486d80b3a9d475c67ef5 : SUCCESS in 27m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3a4f263e353b43efb2ed2bfd8246a8b4 : SUCCESS in 21m 24s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2519fd94f0ba4b3885890f01c2699481 : SUCCESS in 26m 28s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6b0dff6a2ac34990b7c3b523239adf81 : SUCCESS in 17m 24s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4aea96b6eca0464f9ede0751e9c1b9c5 : SUCCESS in 26m 55s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/869ad727f8a64157aaf09c54f2d43925 : SUCCESS in 16m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5adbacaf38714b349457343d318901bf : SUCCESS in 20m 15s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a52011439a44471d8b6e98fcb2bed2ee : SUCCESS in 28m 17s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b0e08b059e5c4db2a09195a5a7c9fc3c : SUCCESS in 20m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3bbc61bd15994bf88cf77347f643f497 : SUCCESS in 16m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/93085e6d3605497391efd8f54bf7ab9f : SUCCESS in 15m 02s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/33049cef27b94f0ca54b62388e1def25 : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":1},{"id":"64025fc49eb86a73492ca616cb781dbfa0c0c4ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 18:47:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4c77f1939ad671968e94735ce60a9900fece7b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 19:23:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/076624afd13d460f8c2d62041da3405a\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/025b52a2711c461d9b86c777b649a9ce : SUCCESS in 21m 54s\n- loci-glance https://zuul.opendev.org/t/openstack/build/273a87669680423c9a92eb63a2303c49 : SUCCESS in 19m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/bfbdc645883e4b94bf7d8d20dddaf524 : SUCCESS in 20m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c363df44aa0142a88ae06dc62366e0f3 : SUCCESS in 20m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/93ed890dd24f4972a00bf510bf2ea91c : SUCCESS in 20m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ea1823ef583a4b7e92c51edb686c6302 : SUCCESS in 19m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/006c64d0329d4e5c8e9e2fb760a0bbb4 : SUCCESS in 18m 05s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f631d2ed13224578ae96588138ed6c43 : SUCCESS in 18m 28s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f565e2d6df6c44539b678c9542811fb1 : SUCCESS in 20m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7aa0ca972751439ca7061d089e9e7c2b : SUCCESS in 22m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/93a09ee752ae414da51074b31be48468 : SUCCESS in 23m 23s\n- loci-placement https://zuul.opendev.org/t/openstack/build/21458b34ad7a4784a835551a00aca201 : SUCCESS in 20m 53s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6f352417ebbd41378449da173b060ac3 : SUCCESS in 16m 22s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/17ce934516ed43eb916455ed62586465 : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7af0ba0fc03258638ddd34529fbef21d631e2067","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 19:23:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7dd55124c3d4a1368d027575332eb0e045b426ce","revisions":{"7dd55124c3d4a1368d027575332eb0e045b426ce":{"kind":"REWORK","_number":1,"created":"2022-11-08 16:50:27.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/25/864025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/25/864025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/25/864025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/25/864025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/25/864025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/25/864025/1"}}},"commit":{"parents":[{"commit":"5688c6ffa366f268782ea53add4b7630449eee8a","subject":"Merge \"Update fetch_wheels to add Accept header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5688c6ffa366f268782ea53add4b7630449eee8a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:49:18.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:49:18.000000000","tz":-360},"subject":"Add cryptography back for placement","message":"Add cryptography back for placement\n\nThe compute-kit jobs are currently failing in osh-images due to\nplacement not having cryptography available. At this time the\nplacement project does not specify cryptography in their\nrequirements, so in the meantime this change adds it back but\nspecifically for the placement project.\n\nChange-Id: I4e7d8c81d3793d4d3b85b450ded673ad122dc266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7dd55124c3d4a1368d027575332eb0e045b426ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7dd55124c3d4a1368d027575332eb0e045b426ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[]}
