)]}'
{"id":"openstack%2Floci~864026","triplet_id":"openstack%2Floci~master~Id387d4d0bd901682d0aec87ef5e5ee784b9af10c","project":"openstack/loci","branch":"master","topic":"pyldap","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2023-11-02 18:25:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id387d4d0bd901682d0aec87ef5e5ee784b9af10c","subject":"Remove pyldap from pydep","status":"MERGED","created":"2022-11-08 16:52:08.000000000","updated":"2023-11-02 18:25:06.000000000","submitted":"2023-11-02 18:25:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864026-pyldap","meta_rev_id":"d5e72cc302c54f2a7df957e8fa6a3db536b08e6f","_number":864026,"virtual_id_number":864026,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-02 18:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2023-11-02 17:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2023-02-24 17:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2023-11-02 17:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"}],"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 17:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-18 16:34:36.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2023-10-13 08:21:13.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"CC"},{"updated":"2023-11-02 17:54:13.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"}],"messages":[{"id":"711488bc0997d001c26c99cbbc4e880ceeabc51e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-08 16:52:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e849ea44bb27d1d40cc5ce76c1b041e28bb9bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 17:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a076f88d58994d15b787ce62c1c93993\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/184c080d0dd64bbf98ef72ee01667b64 : SUCCESS in 21m 35s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4aab47be95984def86b4cf48df54f241 : SUCCESS in 24m 04s\n- loci-heat https://zuul.opendev.org/t/openstack/build/63ccbf96f0ca44a1a8878ea950e0c6b7 : SUCCESS in 21m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/12e0b701c5054b669b33a77b845159b0 : SUCCESS in 25m 28s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b3e74f4c2ba245db97c7b320deed818b : SUCCESS in 16m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/73f9e6d6b1f14a45b22819a1ee56fc88 : SUCCESS in 20m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/82c0902320b84ae18496d5d06548f39e : SUCCESS in 16m 21s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5dabc27578b045e6b2159ed525759d11 : SUCCESS in 24m 07s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2cb7522bcb644320830a9f1212644981 : SUCCESS in 17m 15s\n- loci-nova https://zuul.opendev.org/t/openstack/build/16dc59fa60b541498e2e4a337a6b3820 : SUCCESS in 23m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8291accec29940809a527ed8fc8e02eb : SUCCESS in 23m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/13fc61252f7143e991b8e85ac085c5d1 : SUCCESS in 15m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/064e03d694c44665887cf2c95f491a3f : SUCCESS in 21m 03s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5f798b9c738549e78002827c34354bc1 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0277f1d7abce05308f3432d72da4121be168037a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-16 15:49:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1714105369cf8e52884a022b3643fa88f4e94b15","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-11-16 15:49:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"54bf32f81af6225c78610e43c79ec1c07db08f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:13:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d1c8583ef34ea08adaef0fee995dc2\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/03320c1c317a4233b544280283513449 : SUCCESS in 17m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/77ece7f276cd460da8a5b86f2e69bd06 : SUCCESS in 15m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/533f87b4c5c844bbac99b793a7bccdc3 : SUCCESS in 19m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3dd2bdd29f544165b2381de1e6b7100e : SUCCESS in 16m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c0f50a2250ad46e5956b92debe324663 : SUCCESS in 16m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5bcff24bbc2b4d1c9c8d87ffd2da7d47 : SUCCESS in 17m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b25e979798624a53927421d6dbe4dac8 : SUCCESS in 16m 46s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d990b1779c9c40cd99533787262cc227 : SUCCESS in 16m 36s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/87ac8782f5f24b39acfcf43fdebcbb1a : SUCCESS in 17m 37s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f6079eaea94d4d6784f020b376c863b6 : SUCCESS in 21m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fad550eb08d34128a140d32c3f5ef37c : SUCCESS in 17m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3ad0513910d641538f13580f775e65f2 : SUCCESS in 19m 20s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3f3de61a65ce44c0961e90f9086257ee : SUCCESS in 13m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2100e365e04a4b3c9af4932638c13cf5 : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":2},{"id":"24c21afa83601b2814552b35a6c98089d82d644d","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-18 16:34:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"524de3cb542d6dca21168b97708c981420b620aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-22 22:18:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0fae8132c7ffb55f3f8984f0f876a7c224a80d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 22:51:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397474a48ca649959aaa091fc9d82192\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5f9da378d7884136855ea5855ba60318 : SUCCESS in 21m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/eb6f8b33a9dc4d35aa40c3849a2e3e77 : SUCCESS in 17m 26s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0153829fafea49d090c43fbfcecc7dfe : SUCCESS in 26m 06s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5dd35a9044e646cf9db4a94c475ae868 : SUCCESS in 18m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4cc3b37ea7774fa7baa12bcd18c734a6 : SUCCESS in 25m 34s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/038496eec3224abaaa2626b71c96e19a : SUCCESS in 17m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/06e4229c75a64a0ea76bfabc13e5b4dd : SUCCESS in 19m 46s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6eff4c13da4e452f92a3cd017e355c19 : SUCCESS in 17m 50s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d0174c49661042a4ad111148f2f5c308 : SUCCESS in 17m 32s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b506d51fa06841349ae21d3f50175a8a : SUCCESS in 23m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2a6889cd0a7244d1be2899b593745be5 : SUCCESS in 15m 58s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5cf4ed6e512c42ffaf1c458c5f3444b9 : SUCCESS in 29m 32s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f307e8e5147e4239ad73d7860bec0c24 : SUCCESS in 15m 06s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3862baca544c409cb7197de129caab36 : SUCCESS in 22m 13s","accounts_in_message":[],"_revision_number":3},{"id":"728ed69563f1f318c1879266178dde9c249cdd2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-12-12 22:51:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"94ac6e0f8f548a196e5f8439e023c1e95662c288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 23:30:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/901336cd81bf4f9c8e3325f47dd9dcb7\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e47f62474d504067ac5e04b2d5b2e342 : SUCCESS in 21m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/fcd75606906f4a40b97cc84458bef137 : SUCCESS in 21m 41s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e13fd4b5b51142609eb9a2ddb531cbd8 : SUCCESS in 23m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6604507ab15b45ff90c0d5b302c4a6a0 : SUCCESS in 21m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b9d776f92f364d2b85b876d79c304fb2 : SUCCESS in 19m 34s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/073d31d0e31e47debf1c40f1bb416516 : SUCCESS in 25m 03s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7c8b3ade87794d82b22c9ec0aa7b3c2a : SUCCESS in 25m 24s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/6f8c79b4fbaf44a7986dd091ffde98f0 : SUCCESS in 22m 19s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0cead1daeea94898afaa244913690a4b : SUCCESS in 19m 54s\n- loci-nova https://zuul.opendev.org/t/openstack/build/45def2cebcdd483f9349e0f2a7345d2e : SUCCESS in 22m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0033367baaef4b6c9a4b479431c85756 : SUCCESS in 25m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8a46c68b1a9541e99f4ee6626d08b616 : SUCCESS in 17m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/df6b1ffc6a32410ab8587f2909ab4906 : SUCCESS in 14m 34s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f2b93bce952d431b9c2b791cfdc76207 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":4},{"id":"eae1da873640c9c06e1b444b19d091466307326f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2023-02-16 06:03:04.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3d66063ec2275023e2bcac62189115b3ff2f3186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:41:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cf98c892d0f4e6591e9642acdf05641\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a8434877e23941bf9ebc7742ff2bd088 : RETRY_LIMIT in 2m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3a168c6f2a8d485bba97fa9590fca7aa : RETRY_LIMIT in 2m 38s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e9a3461e6b954bd2b10c9ce2481ce6f4 : RETRY_LIMIT in 3m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ca6a9ef516a245c5aca22028a16aa9bd : RETRY_LIMIT in 2m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3ccef25780554a0c92267b08cf23f398 : RETRY_LIMIT in 2m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4719c788fcfa48b996542f803de439fa : RETRY_LIMIT in 2m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/66516455b92343b4b3798ec14f3604b4 : RETRY_LIMIT in 2m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/35fcebe5eb4c4c25b346c0d4eb5abc60 : RETRY_LIMIT in 1m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2a026cc6347448328aaa498036817f16 : RETRY_LIMIT in 2m 13s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a1afe5c34b4a4422bf908b94a68bc131 : RETRY_LIMIT in 2m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a0c23a0212334fd08f8c9541c70e2678 : RETRY_LIMIT in 2m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/94ab197ae24e4a0b92ee1bf70cf7fe6f : RETRY_LIMIT in 1m 53s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ef627b7fbb44424a9fd5d89d9374b676 : RETRY_LIMIT in 1m 51s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3839ab8384d64c5480ae19dbf761ad79 : RETRY_LIMIT in 1m 51s","accounts_in_message":[],"_revision_number":5},{"id":"73b3a95572487b43e9fa8766c338998880c90a19","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2023-02-24 17:48:38.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5a9b4f07ffb5936afc2d23a18dcab5be2ea00938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 18:19:20.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/088ecde02db24a85915475615f2fed57\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/083f88697eef454c99c5c0acdb15742d : RETRY_LIMIT in 2m 12s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1bd4ca922df64a33b70c30b62c50d565 : RETRY_LIMIT in 2m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e34beeaab6d44039bf0072e146ab621f : RETRY_LIMIT in 2m 22s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/30af816000474686b6274bba5ba299d7 : RETRY_LIMIT in 3m 01s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d7497ad5c8d547adab89c513d6a9fba8 : RETRY_LIMIT in 2m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8b4e3a180df342289f546fa602ed71a8 : RETRY_LIMIT in 2m 50s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e7e674100d7c459a8bdbcda13b77184c : RETRY_LIMIT in 2m 48s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/60dce5e15b5e4e00966a49e457cdf356 : RETRY_LIMIT in 2m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4ea806259f554e2abccb972950bae991 : RETRY_LIMIT in 2m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7edb7e9e46e540098fc7d76557b9a1bb : RETRY_LIMIT in 2m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/779cc4d246934906b52d25944c3b149c : RETRY_LIMIT in 2m 13s\n- loci-placement https://zuul.opendev.org/t/openstack/build/81082ec931554d8a80fea43f6a2e2be8 : RETRY_LIMIT in 2m 11s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7e8d1861323f40c59c89990b99c81998 : RETRY_LIMIT in 2m 00s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6add1ffe994b4e58bc854e39658d875a : RETRY_LIMIT in 2m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b60538e22590d29a00c33abf87472da7dc0db4eb","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2023-03-17 05:03:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c040bbc16eae75bb496604237f85a49fe6e06e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 05:30:15.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f555a96354342718fed0fb407b2d12d\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7984f1479d74465fa51cc427f8b1b82e : RETRY_LIMIT in 2m 01s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f33c32f7258748ab8a5c282af2e34210 : RETRY_LIMIT in 2m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4e840a4678924b26a298245992d5bb29 : RETRY_LIMIT in 2m 42s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/bf16d22f1e2f4d18a77c3b8ed3833a25 : RETRY_LIMIT in 2m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1f7058a0ce154d10b1ff2bcc8cf7d8bc : RETRY_LIMIT in 2m 15s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3c1b4e0804b34825b38465e0779c24d5 : RETRY_LIMIT in 2m 17s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4e6b54e8a5f44dc3a5c5a41b2c72ff39 : RETRY_LIMIT in 2m 08s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4412d759c32149f39e1ad30714f36453 : RETRY_LIMIT in 2m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ede61f9856884c41b45e5585189ddd8a : RETRY_LIMIT in 2m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cde3072d2717407e89c53b75eb5f7f5d : RETRY_LIMIT in 3m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7cdc6ec4831c49938b8b3280668ac13a : RETRY_LIMIT in 2m 07s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0d43948d401b4374928f34b905b47466 : RETRY_LIMIT in 1m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9d487e0399f4444eb9b36ac35434a6a5 : RETRY_LIMIT in 2m 02s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/ec170b64789b48908e1fe5baf31f2b8b : RETRY_LIMIT in 2m 14s","accounts_in_message":[],"_revision_number":5},{"id":"17b9a4c189054ba4996f703f44800ea0532c8773","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2023-10-13 08:21:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f5f6ca381f9c22c34cbcd74835d33c642a24af5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 08:50:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b1b2b5054a54ed499a496c074dc7ea3\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a4de5be76c1745699e8009922d8a8341 : SUCCESS in 17m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6899664d86a64cd0b8ca48ff5d83fdeb : SUCCESS in 19m 55s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fb0f92a9975c4c82b26e9617afc37d62 : SUCCESS in 17m 00s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/624dbf5876cb4c769c02d8c33a6105e4 : SUCCESS in 19m 17s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e3a8903c1f114d679a99920923e34089 : SUCCESS in 16m 50s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/175489c863484c50a5588eb4f29e25b8 : SUCCESS in 19m 54s\n- loci-manila https://zuul.opendev.org/t/openstack/build/db8494f86e994863adaaab6996b4850e : SUCCESS in 19m 01s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b69fbbc7e3f147748d67ff411b1988ee : SUCCESS in 19m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e2d1e4626b424752b8ec1bc2c33e53f7 : SUCCESS in 20m 08s\n- loci-nova https://zuul.opendev.org/t/openstack/build/88b3224aa523434e8cc61df3d1ed79f4 : SUCCESS in 22m 07s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3fea087a8c804f33a18edeb185a036db : SUCCESS in 17m 58s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0b4c637699ab4de48fc9c60b7636e306 : SUCCESS in 19m 28s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8ab2f2f8c45f4931a5cb841a13ea77a3 : SUCCESS in 13m 29s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/9a128dface0144c8aa216ace19dadf41 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":5},{"id":"057f386ce30eef7ba6bbddb6770f49367b9658b4","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-11-02 17:54:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5e16201641500ae1c28d82e4cf0b76637a8b020c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 17:54:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f06914c9a1b37fdd64aa40ca3b4e90b462212121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 18:25:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3658c8f70a054a0c82fc51e33a844df3\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3686e8c62db64e25b2fdd537a1d6ebb9 : SUCCESS in 20m 04s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8211ffad11074a21aa32a55898c830b4 : SUCCESS in 20m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4c11a6a97e95446baf4bc657f7ff6b42 : SUCCESS in 18m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/643d28283d8145318cb16b89c097782d : SUCCESS in 17m 48s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0c19247af13d469fb666e1c7b0552d27 : SUCCESS in 20m 07s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4bf01e4b04eb430abae08ce061c97fd5 : SUCCESS in 18m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5496b5946afa48f4a43d8e30c938335d : SUCCESS in 15m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/420705cfe52f473cb37fa6c3ef6e7781 : SUCCESS in 18m 04s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/444bd197539d48c7a8564eb1a9b677a0 : SUCCESS in 17m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b0da14d48bcc42cab4ba3a9dd903f274 : SUCCESS in 18m 45s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c368d2d1345547e59cf8fe72e62f97a8 : SUCCESS in 15m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8ad36d57e5034a168b2f979e974fa27a : SUCCESS in 18m 45s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b90aad839a9e4b0d8805630d4a8fbf6b : SUCCESS in 10m 37s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7669d5af8dad4effaea58bd8566525fd : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":5},{"id":"d5e72cc302c54f2a7df957e8fa6a3db536b08e6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 18:25:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c4281d2291cc57fcac3b42096d1b40fb1a5143cb","revisions":{"e6b5597df76dc08e631297d402d61166939915ab":{"kind":"REWORK","_number":1,"created":"2022-11-08 16:52:08.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/864026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/864026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/864026/1"}}},"commit":{"parents":[{"commit":"5688c6ffa366f268782ea53add4b7630449eee8a","subject":"Merge \"Update fetch_wheels to add Accept header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5688c6ffa366f268782ea53add4b7630449eee8a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:50:59.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:50:59.000000000","tz":-360},"subject":"Remove pyldap from pydep","message":"Remove pyldap from pydep\n\nThe pyldap library was needed only through the queens release, which\nwe no longer support. This change removes the entry as well as\nremoves the checks for python-ldap for these older releases.\n\nChange-Id: Id387d4d0bd901682d0aec87ef5e5ee784b9af10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e6b5597df76dc08e631297d402d61166939915ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e6b5597df76dc08e631297d402d61166939915ab"}]},"branch":"refs/heads/master"},"6e09ee69d31b731ed1eae3752c79d571dc06dc30":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-16 15:49:44.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/864026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/864026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/864026/2"}}},"commit":{"parents":[{"commit":"7dd55124c3d4a1368d027575332eb0e045b426ce","subject":"Add cryptography back for placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7dd55124c3d4a1368d027575332eb0e045b426ce"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:50:59.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-16 15:49:44.000000000","tz":0},"subject":"Remove pyldap from pydep","message":"Remove pyldap from pydep\n\nThe pyldap library was needed only through the queens release, which\nwe no longer support. This change removes the entry as well as\nremoves the checks for python-ldap for these older releases.\n\nChange-Id: Id387d4d0bd901682d0aec87ef5e5ee784b9af10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6e09ee69d31b731ed1eae3752c79d571dc06dc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6e09ee69d31b731ed1eae3752c79d571dc06dc30"}]},"branch":"refs/heads/master","description":"Rebase"},"6e5952f27c3f038b454109a5e17be13d36575981":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-22 22:18:57.000000000","uploader":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"ref":"refs/changes/26/864026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/864026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/864026/3"}}},"commit":{"parents":[{"commit":"39aaeb3f329993046703233d2f52c9f7470a8de1","subject":"Remove uwsgi by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39aaeb3f329993046703233d2f52c9f7470a8de1"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:50:59.000000000","tz":-360},"committer":{"name":"Chris Wedgwood","email":"cw@f00f.org","date":"2022-11-22 22:18:57.000000000","tz":0},"subject":"Remove pyldap from pydep","message":"Remove pyldap from pydep\n\nThe pyldap library was needed only through the queens release, which\nwe no longer support. This change removes the entry as well as\nremoves the checks for python-ldap for these older releases.\n\nChange-Id: Id387d4d0bd901682d0aec87ef5e5ee784b9af10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6e5952f27c3f038b454109a5e17be13d36575981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6e5952f27c3f038b454109a5e17be13d36575981"}]},"branch":"refs/heads/master","description":"Rebase"},"8e42aa89f93707bca1ba7ea754d1774c51fbf21a":{"kind":"NO_CHANGE","_number":4,"created":"2022-12-12 22:51:09.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/864026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/864026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/864026/4"}}},"commit":{"parents":[{"commit":"5806885d68c6e15251168183245976a26c0da81a","subject":"Merge \"Drop UWSGI from always being installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5806885d68c6e15251168183245976a26c0da81a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:50:59.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-12-12 22:51:09.000000000","tz":0},"subject":"Remove pyldap from pydep","message":"Remove pyldap from pydep\n\nThe pyldap library was needed only through the queens release, which\nwe no longer support. This change removes the entry as well as\nremoves the checks for python-ldap for these older releases.\n\nChange-Id: Id387d4d0bd901682d0aec87ef5e5ee784b9af10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e42aa89f93707bca1ba7ea754d1774c51fbf21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e42aa89f93707bca1ba7ea754d1774c51fbf21a"}]},"branch":"refs/heads/master","description":"Rebase"},"c4281d2291cc57fcac3b42096d1b40fb1a5143cb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-16 06:03:04.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/26/864026/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/26/864026/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/26/864026/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/26/864026/5"}}},"commit":{"parents":[{"commit":"f725a1256a2413fdada29b4d9d88acae63970ae9","subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f725a1256a2413fdada29b4d9d88acae63970ae9"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-11-08 16:50:59.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2023-02-16 06:03:04.000000000","tz":0},"subject":"Remove pyldap from pydep","message":"Remove pyldap from pydep\n\nThe pyldap library was needed only through the queens release, which\nwe no longer support. This change removes the entry as well as\nremoves the checks for python-ldap for these older releases.\n\nChange-Id: Id387d4d0bd901682d0aec87ef5e5ee784b9af10c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c4281d2291cc57fcac3b42096d1b40fb1a5143cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c4281d2291cc57fcac3b42096d1b40fb1a5143cb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
