)]}'
{"id":"openstack%2Floci~880460","triplet_id":"openstack%2Floci~master~I3e9c74fc6afc4f11b14d5c9834f585fd57f51cbf","project":"openstack/loci","branch":"master","topic":"PRODX-31565","attention_set":{},"removed_from_attention_set":{"35392":{"account":{"_account_id":35392,"name":"Oleksandr Klymenko","display_name":"Oleksandr Klymenko","email":"oklymenko@mirantis.com","username":"oklymenko"},"last_update":"2024-03-06 23:13:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3e9c74fc6afc4f11b14d5c9834f585fd57f51cbf","subject":"CADF audit support for Placement API","status":"ABANDONED","created":"2023-04-14 08:30:22.000000000","updated":"2024-03-06 23:13:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b01e04962bed13138bd2dc1b19d78cbb991690f","_number":880460,"virtual_id_number":880460,"owner":{"_account_id":35392,"name":"Oleksandr Klymenko","display_name":"Oleksandr Klymenko","email":"oklymenko@mirantis.com","username":"oklymenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-14 08:43:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 08:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6aa45fd7ed6a1fb07af33cb512c2f32698b5bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35392,"name":"Oleksandr Klymenko","display_name":"Oleksandr Klymenko","email":"oklymenko@mirantis.com","username":"oklymenko"},"date":"2023-04-14 08:30:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fef353c73ae24b65d74a90b735f79610c713cb16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 08:43:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f12d29eabda14bdfa0f5533aa818c4f0\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/41e6d9f83c4941948d84fa51ef232b1b : FAILURE in 2m 15s\n- loci-glance https://zuul.opendev.org/t/openstack/build/92b2dbda9a8a4e9e97401ad1f4f3c18a : FAILURE in 2m 30s\n- loci-heat https://zuul.opendev.org/t/openstack/build/80fbadff0c434c79b5ee02d0b7fefa14 : FAILURE in 2m 19s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e78462529d744caa8ef421ee228178d6 : FAILURE in 2m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ab687320eb9a4887a44afcd5e9153056 : FAILURE in 2m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bd2ce20a0e9e42e3ba92ddf448ff20ed : FAILURE in 2m 22s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d2f6c094e00e4380bce727006d3780ff : FAILURE in 2m 07s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/61cb8153c37449ca9b98d573337384d2 : FAILURE in 2m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a8592fbd459f4a83ab673e7fa32f4eec : FAILURE in 2m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/248b778d7009405e8d2abd13db121959 : FAILURE in 2m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7b7d320d7da54ca7b4c7ff10f602c4ad : FAILURE in 2m 27s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e52b1b6f7e6746d19a17cc61937b50f3 : FAILURE in 2m 26s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5b93a5c0383d4640a0e5d5797c4f02e3 : FAILURE in 2m 11s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/30b0bd477a7b4b75a6223793e98278fb : FAILURE in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5b01e04962bed13138bd2dc1b19d78cbb991690f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2024-03-06 23:13:44.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 327 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2083b10fef7303b632737e2557e44b26a8908d8","revisions":{"e2083b10fef7303b632737e2557e44b26a8908d8":{"kind":"REWORK","_number":1,"created":"2023-04-14 08:30:22.000000000","uploader":{"_account_id":35392,"name":"Oleksandr Klymenko","display_name":"Oleksandr Klymenko","email":"oklymenko@mirantis.com","username":"oklymenko"},"ref":"refs/changes/60/880460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/60/880460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/60/880460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/60/880460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/60/880460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/60/880460/1"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"oklymenko","email":"oklymenko@mirantis.com","date":"2023-04-14 08:30:13.000000000","tz":180},"committer":{"name":"oklymenko","email":"oklymenko@mirantis.com","date":"2023-04-14 08:30:13.000000000","tz":180},"subject":"CADF audit support for Placement API","message":"CADF audit support for Placement API\n\nChange-Id: I3e9c74fc6afc4f11b14d5c9834f585fd57f51cbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e2083b10fef7303b632737e2557e44b26a8908d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e2083b10fef7303b632737e2557e44b26a8908d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
