)]}'
{"id":"openstack%2Floci~886115","triplet_id":"openstack%2Floci~master~I956985410f55293c998e75c9fa8960e2b9106d45","project":"openstack/loci","branch":"master","topic":"more_image_info","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-06-29 16:45:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I956985410f55293c998e75c9fa8960e2b9106d45","subject":"Add more info to the /etc/image_info","status":"MERGED","created":"2023-06-14 17:45:06.000000000","updated":"2023-06-29 16:45:17.000000000","submitted":"2023-06-29 16:45:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886115-more_image_info","meta_rev_id":"0246dfa45c9b69ea850ab8181423b2bbd74c6ab7","_number":886115,"virtual_id_number":886115,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 16:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2023-06-29 16:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-29 14:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":1,"date":"2023-06-29 16:23:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 18:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 14:22:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2023-06-29 16:23:09.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"0ebdbd6598ae3c9eaacfce4d0b9b6f8fef8c7e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-14 17:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6fcc1f2174d757a0616dc0196f1a64f75e4a75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 18:07:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de9288c06c6d4e489fd2291b6585d1bf\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/30907e5a11b24e04b96511e468d58394 : RETRY_LIMIT in 2m 18s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7ba3cc36718441e5b48fe3898c58ceea : RETRY_LIMIT in 2m 12s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e283316d401e48f79c87e2f8a553e5c1 : RETRY_LIMIT in 2m 46s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c6d3bd4fc40940cfabf22e4787003f24 : RETRY_LIMIT in 2m 17s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e3c9f3e9cd3d4362a018c5a66282564f : RETRY_LIMIT in 2m 12s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/400bec99519042a79e711d7df77f6881 : RETRY_LIMIT in 2m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1485ae9414a0419192c4f7852c87c295 : RETRY_LIMIT in 2m 10s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/57c11d2d6b034276b08c8071a234ea9d : RETRY_LIMIT in 2m 04s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2373b4455356495dbfc61abbcff18b87 : RETRY_LIMIT in 2m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ed4a90475f274bc3b70101fbaa785824 : RETRY_LIMIT in 1m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/eb95c9fc803e4316bd1e2121642a146f : RETRY_LIMIT in 2m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e76e5751f87348cc9496c433e17c35de : RETRY_LIMIT in 2m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/dc3ed680b9a04f76a2bf66208c78b75a : RETRY_LIMIT in 2m 10s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b9e264e4703748c79e9fdb87df0a4941 : RETRY_LIMIT in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"205e4aac6792b1e0c45c210964644a0649252f88","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-14 21:30:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"800e806aeff6f5260534fd754451ff97de8a1378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 22:01:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9f418b03ed841d3ab27babc28cd101e\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4bdaef4f7f604c1daaa057c015599ab6 : RETRY_LIMIT in 1m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b37fa60af6734077bca0404fba791e93 : RETRY_LIMIT in 2m 50s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1c295ddaab7e456ebd99a097a5403e16 : RETRY_LIMIT in 2m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b4b6acd32d6a4452b0405ac739be5023 : RETRY_LIMIT in 2m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3fc81e76a400470bbdb342ee642594de : RETRY_LIMIT in 2m 06s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/152b74329bf04ff78324d8b6e1fe2a9b : RETRY_LIMIT in 2m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c189cec5ddba4b15a9b143c2080093f2 : RETRY_LIMIT in 2m 06s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/67ace174888f423db8d1e9f1da0f1b87 : RETRY_LIMIT in 2m 06s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e333ba251aa545708b42d315b75bc5ea : RETRY_LIMIT in 1m 57s\n- loci-nova https://zuul.opendev.org/t/openstack/build/43e5da6e218e486b8a32dea096522c40 : RETRY_LIMIT in 2m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c1f024d3b66b4caaad321a967f7cff45 : RETRY_LIMIT in 2m 13s\n- loci-placement https://zuul.opendev.org/t/openstack/build/06b79bc8863342a5800405ebc0d2a931 : RETRY_LIMIT in 2m 02s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6e8e93196aa54a42810174ed847d8f36 : RETRY_LIMIT in 2m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8ebc1bfc870846e0ab84be965103fb54 : RETRY_LIMIT in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c025315c153655bf4978757d5e17b6037fd6850c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 01:29:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"00db2787b780b1ce50946b6ad305888565115a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 01:39:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"23e03cc1f400d7efe8a17ae50986d1314020d14a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 02:08:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc78dca3aea14d619d6cad8e9639c457\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d98787a524524b4b87ae399db55e7171 : SUCCESS in 22m 59s\n- loci-glance https://zuul.opendev.org/t/openstack/build/70850dc2ebe04f3ebc69fe372a63598f : SUCCESS in 19m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5ee2d05587364875966ce89910b91c17 : FAILURE in 19m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2d1d17805d2f4abab9ebb0834203c80f : SUCCESS in 21m 32s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6a7d7dec28d5435db86e043aa44025a9 : SUCCESS in 20m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4e9f8249ba56465490ba9c9383d80cf1 : SUCCESS in 19m 32s\n- loci-manila https://zuul.opendev.org/t/openstack/build/35188ea67c9c4c6a80e863a14e89a059 : SUCCESS in 18m 37s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4d2b774816ce43fb95ac3034a831fd93 : SUCCESS in 17m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/bf28a81194694595bf5c7c5f05963054 : SUCCESS in 21m 17s\n- loci-nova https://zuul.opendev.org/t/openstack/build/745299792acc4a358d485d7ca5e0a67c : SUCCESS in 21m 31s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1820fe21a0ce4440a972e5ae3af1eeef : SUCCESS in 19m 45s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f4e5bf8e9f4e4349b5689e5a8bab102b : SUCCESS in 14m 27s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b85915780d60495a93a9d73cee646550 : SUCCESS in 10m 32s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/0770a5d120ca4189abcbd3bff25d8da2 : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d8eba038ef293ef0b9ad58e87ea2c9a71d1c0969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 04:00:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a620503272054ba1856cff8ae6f6fbba4d4ddb6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 04:20:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc26a877bfdc407e8eb956102d02763e\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/83ebf19d298f4ff8884b60068fc970ae : FAILURE in 9m 44s\n- loci-glance https://zuul.opendev.org/t/openstack/build/987aa0b828c44928958f44f65a2d8c13 : FAILURE in 7m 10s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e17a1a4cdb2f4631bbea3804476c738c : SUCCESS in 11m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f92b8915756541ac8da194a9b82a7543 : SUCCESS in 12m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/63f1f44dbbec4adaa03cfac33c942aa8 : SUCCESS in 9m 14s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7923c69c4c814ccea25e84c32f7167d8 : SUCCESS in 9m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/71dd9f2c13464cafb4cecf47b0ff46ea : SUCCESS in 8m 32s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/80d8c5cfe5ec42ea8c2c2ca5302d2da1 : SUCCESS in 9m 55s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/989b646dcf5d468cabb1c1a4fdfbecae : SUCCESS in 7m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fb325e3dc5194d598fedc0895ebedd17 : SUCCESS in 11m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0c54fabf9be44d1db869f78819bbfcc9 : SUCCESS in 9m 10s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f28fb1db2510446489ebe9f94c54b9a2 : SUCCESS in 9m 06s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ce703837d99d489587567d1c964bcb51 : SUCCESS in 6m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d8b24cb9ea874decafc6e703fc2da842 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":4},{"id":"a692fd6b631ada1092a87dbd3e1c3016ff1e3ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-16 17:29:35.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a55534639ac4e0aba5fe5e266008939b689dbed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-16 17:30:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"960e2942c4a52ffabb31c37eac50d0494a445cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 17:52:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc09c2cf3b4479083f8b26b0f9c2477\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8e5fb796182c45ca9e65e629a2e6e835 : SUCCESS in 8m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bdcae40195914273a96e1a30066cd5d3 : SUCCESS in 8m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e2d2a9fe35a645bca2fd2d37eafc5b17 : SUCCESS in 10m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/99cc0b5c36da45248a4aea12d56fd6b8 : SUCCESS in 10m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5a474ff6c4fd43e39a0252c5443d7d22 : SUCCESS in 9m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/551f0cc47bd54df7895a8a666a1e1795 : SUCCESS in 9m 27s\n- loci-manila https://zuul.opendev.org/t/openstack/build/70b7670e0e014a77a6b22fb29ae04e2a : SUCCESS in 9m 18s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5f8dcafc90ed459fbe7f40dc2e72b349 : SUCCESS in 8m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7ef36fd8397849eeb137d9697193741f : SUCCESS in 8m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/36a3eb00ffb84ab9ab885cddd1709163 : SUCCESS in 11m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/30f43669db8344de905c198f58ffee5a : SUCCESS in 9m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3d403584306a4b979fcbc40d2d681c6b : SUCCESS in 8m 55s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/24a16b31d90d48a58140159dc709f6a1 : SUCCESS in 7m 02s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d8fd94f8ef55419395c91529a2239378 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":6},{"id":"6fe134ca33f8972a12a5736cf65effdd70852c44","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-06-29 14:22:31.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"396e513868b96b717647bbc6141705f722cad0f2","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2023-06-29 16:23:09.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"92df45b98d8b4cf51ec01bd962a815693d225787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:23:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"194d8eefa8d7f0d05b439774c05019759b423492","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:45:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe9760d43ac4479195f2cfb3e5219018\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/bfe4f8452a594f1f8e39ef7cfb854762 : SUCCESS in 14m 13s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8aa639671e874dec93b4fe949517b255 : SUCCESS in 9m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/979e9815402a4b10aa935048e04ebf83 : SUCCESS in 10m 03s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d87b358f54294a1d816d3bb706d513f1 : SUCCESS in 11m 11s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/702155e4fb704b3c8a50fd260b386cc8 : SUCCESS in 14m 16s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a495df58290d435ebb52eddfa3f2ab0d : SUCCESS in 8m 51s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a28bd2eb0f724a47847c6e99397e3742 : SUCCESS in 10m 00s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f7900231b7fc4e6bbb81a72d24d99cde : SUCCESS in 10m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2cd55018e55d465496d342904d902284 : SUCCESS in 11m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/916430644af04a04a3f6616c43126214 : SUCCESS in 11m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2f953a0b0fd048c19654c5631fcedd41 : SUCCESS in 10m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b5f5d80346e14f169e81cfc64fe7b30e : SUCCESS in 10m 06s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f685f67967db441fb9422963f3c46663 : SUCCESS in 7m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a97ca889492842b089fab39aa1a81ac3 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":6},{"id":"0246dfa45c9b69ea850ab8181423b2bbd74c6ab7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 16:45:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7c3c53c375777909664c17b760493cbf8a19e24f","revisions":{"a02274e9ed574bb09ef8917d1ec4255aac650bee":{"kind":"REWORK","_number":1,"created":"2023-06-14 17:45:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/15/886115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/15/886115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/15/886115/1"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"subject":"Add more info to the /etc/image_info","message":"Add more info to the /etc/image_info\n\nChange-Id: I956985410f55293c998e75c9fa8960e2b9106d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a02274e9ed574bb09ef8917d1ec4255aac650bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a02274e9ed574bb09ef8917d1ec4255aac650bee"}]},"branch":"refs/heads/master"},"08dd23b65fd1b9db04f6ca7d6c8b597175f30f5b":{"kind":"REWORK","_number":2,"created":"2023-06-15 01:29:33.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/15/886115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/15/886115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/15/886115/2"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:29:28.000000000","tz":180},"subject":"Add more info to the /etc/image_info","message":"Add more info to the /etc/image_info\n\nChange-Id: I956985410f55293c998e75c9fa8960e2b9106d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/08dd23b65fd1b9db04f6ca7d6c8b597175f30f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/08dd23b65fd1b9db04f6ca7d6c8b597175f30f5b"}]},"branch":"refs/heads/master"},"81a3efc54e89a8028c96a47ea54eccf57785fae3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-15 01:39:19.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/15/886115/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/15/886115/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/15/886115/3"}}},"commit":{"parents":[{"commit":"d77a63a798b076630a8df19ffaabdc64bf7a130e","subject":"Unpin pip docker package version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d77a63a798b076630a8df19ffaabdc64bf7a130e"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:39:07.000000000","tz":180},"subject":"Add more info to the /etc/image_info","message":"Add more info to the /etc/image_info\n\nChange-Id: I956985410f55293c998e75c9fa8960e2b9106d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/81a3efc54e89a8028c96a47ea54eccf57785fae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/81a3efc54e89a8028c96a47ea54eccf57785fae3"}]},"branch":"refs/heads/master"},"45bee07f023e2cd131bfacf7fb082c7b3a4537c8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-15 04:00:15.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/15/886115/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/15/886115/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/15/886115/4"}}},"commit":{"parents":[{"commit":"8e8aab50f9642507b9446fb544a25b1d975be69f","subject":"Unpin pip docker package version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e8aab50f9642507b9446fb544a25b1d975be69f"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 03:59:59.000000000","tz":180},"subject":"Add more info to the /etc/image_info","message":"Add more info to the /etc/image_info\n\nChange-Id: I956985410f55293c998e75c9fa8960e2b9106d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/45bee07f023e2cd131bfacf7fb082c7b3a4537c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/45bee07f023e2cd131bfacf7fb082c7b3a4537c8"}]},"branch":"refs/heads/master"},"95e279990e6e84a25ded6072c5f7d4e98fe1e6f2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-06-16 17:29:35.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/15/886115/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/15/886115/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/15/886115/5"}}},"commit":{"parents":[{"commit":"e09a032257651effdf6a4f6dd28d4ecbe08ba59d","subject":"Fix setup-gate.yaml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e09a032257651effdf6a4f6dd28d4ecbe08ba59d"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-16 17:29:26.000000000","tz":180},"subject":"Add more info to the /etc/image_info","message":"Add more info to the /etc/image_info\n\nChange-Id: I956985410f55293c998e75c9fa8960e2b9106d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95e279990e6e84a25ded6072c5f7d4e98fe1e6f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95e279990e6e84a25ded6072c5f7d4e98fe1e6f2"}]},"branch":"refs/heads/master"},"7c3c53c375777909664c17b760493cbf8a19e24f":{"kind":"REWORK","_number":6,"created":"2023-06-16 17:30:57.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/15/886115/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/15/886115/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/15/886115/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/15/886115/6"}}},"commit":{"parents":[{"commit":"e09a032257651effdf6a4f6dd28d4ecbe08ba59d","subject":"Fix setup-gate.yaml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e09a032257651effdf6a4f6dd28d4ecbe08ba59d"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-14 17:43:56.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-16 17:30:51.000000000","tz":180},"subject":"Add more info to the /etc/image_info","message":"Add more info to the /etc/image_info\n\nChange-Id: I956985410f55293c998e75c9fa8960e2b9106d45\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c3c53c375777909664c17b760493cbf8a19e24f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c3c53c375777909664c17b760493cbf8a19e24f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
