)]}'
{"id":"openstack%2Floci~886154","triplet_id":"openstack%2Floci~master~I958677e3df88028d00ac0df812c0030419d517a7","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-06-16 17:25:14.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},"34520":{"account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"last_update":"2023-06-15 18:37:51.000000000","reason":"\u003cGERRIT_ACCOUNT_34520\u003e replied on the change","reason_account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}},"hashtags":[],"change_id":"I958677e3df88028d00ac0df812c0030419d517a7","subject":"Fix setup-gate.yaml playbook","status":"MERGED","created":"2023-06-15 01:36:05.000000000","updated":"2023-06-16 17:44:31.000000000","submitted":"2023-06-16 17:44:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886154","meta_rev_id":"4a01fbe6aaf13eb2194b0441fbb454389282bca8","_number":886154,"virtual_id_number":886154,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-16 17:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2023-06-16 17:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2023-06-16 15:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":1,"date":"2023-06-15 18:37:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2023-06-16 17:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 02:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 14:11:53.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"CC"},{"updated":"2023-06-15 14:46:38.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2023-06-16 15:30:16.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"a87a118abcf14e44c7d6f0d4e1bf1d60ad7c260b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 01:36:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"802ccda12bd57b1a24c684e9f406421d7500b28b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 02:06:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bec6128a0e7f44ca97ec8662221c9bca\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/fac1d1c96a6646f59ab679d392a9802c : SUCCESS in 18m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/151b9c8e7ee14a3ea525ca29f5cdbdef : SUCCESS in 17m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3baf906d906945d3a398b2f37d45f1d1 : FAILURE in 14m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a521cd4e28df4f40b41bdab7c3976c76 : SUCCESS in 16m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5fa33f7c7c0943b4b8625594630ca294 : SUCCESS in 17m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9959a58575f54adbb62555e2235a1d16 : SUCCESS in 17m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b049a25060ec48e2a3a71d303c64829a : SUCCESS in 17m 29s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/db7cc154cf9d426d8e119357a48d0fed : SUCCESS in 13m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3b9acef8cb1a4803b9bc034052347ab0 : SUCCESS in 15m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8f9d25ddc7c24ef3b90c72559ae246d3 : SUCCESS in 20m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f4542de718304b9da70bc2d9dc5c8560 : SUCCESS in 20m 14s\n- loci-placement https://zuul.opendev.org/t/openstack/build/43eb1a839eb94f4da798249c152885fd : SUCCESS in 19m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/68121f23631e4c4c91254acdb3ba61d1 : SUCCESS in 14m 11s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/0c5c32a40a1c4f288bf9509793969c14 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ae44a1b00f1b2e1383761d3c2f4b530e54c501b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 02:18:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"997e2c13331149dece5780589eb4a5f439f0329f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 02:36:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9534adc5e526b434e96f60cc3c3f2c60791c6942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 02:45:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"efd9d549ddae23265414bd0125334db0f63a96fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 03:11:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77d9fce11b7d4d7d834e2496fb00db13\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1a8b55e89713401faef7e880e00cc642 : SUCCESS in 10m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e3adccf9db204640a0872e53a530db7a : SUCCESS in 9m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8f923d2023414c86b6df2d387cc77524 : SUCCESS in 10m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/881a543b878e4159b456e6191a9bfe83 : FAILURE in 10m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3989996bd9ef4bada3092d526b94d9a4 : SUCCESS in 9m 55s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a0480080e95b462fbe1afd3b6084dc0f : FAILURE in 9m 32s\n- loci-manila https://zuul.opendev.org/t/openstack/build/60b39f18f0b846569a287d7fb717bdf5 : FAILURE in 8m 36s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/fb015d5d6fde4a198101c91c5318df0d : SUCCESS in 11m 28s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/196dd0f06cec48e9851145f10db553b0 : FAILURE in 12m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7d53767630f34bae8cd7066e393ec098 : FAILURE in 10m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3952ffcd72e04e18a8d41a5ab464caf9 : FAILURE in 11m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fbfa2ae975cd4b96bb0c51e4ab258988 : SUCCESS in 13m 18s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/378d00e09bfd46a297b8176c73df6610 : SUCCESS in 10m 53s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c48fb9cc9f5d47319d2e5d6c73185e4e : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":4},{"id":"be489685a421d3365b29cc6b0ae840f6268f31fb","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 03:12:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8660fa4df73cf78a9625437215fb4c8b0da30c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 03:42:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/252ecd281eeb4b609ffb61629dc05238\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9c4f0ee8af774e9795751dfec81280d2 : FAILURE in 22m 19s\n- loci-glance https://zuul.opendev.org/t/openstack/build/478abce49cb346c5b851d5ce63722ef1 : FAILURE in 20m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9d970485a3a34ba3b9e4075520347529 : SUCCESS in 11m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b2b777b3ea094495ab9058ef80bf7303 : FAILURE in 5m 11s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3cf727aa175b49cca9794a0f1d0a3890 : SUCCESS in 10m 56s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e69e27ad902d424482899d5ee1a75a51 : SUCCESS in 9m 23s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ca69fb607d374625bf88f6790f918615 : SUCCESS in 9m 22s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/1d37ece5b14549ef8e8875873518d17e : SUCCESS in 9m 42s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c0136adda699423f82b00c227777edce : SUCCESS in 10m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ee5fed918f84464dad685775fa4d655b : SUCCESS in 10m 46s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2967c086af284507b0a66ad26a268c71 : SUCCESS in 8m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9d0a6a63aeb64af493825661ede824e9 : SUCCESS in 9m 24s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/fbb119fdd7c743a684878911cb8e5be7 : SUCCESS in 6m 43s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6ba9791c62e24c4db29109205d9502b1 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"eb15fb3a45d7c47fada598ae5392fddee67d7c3d","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 03:44:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c3e278955f2056fb5b2f5356b000d79bbcb8838e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 03:57:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d098d8045744178410e799b9ddcf37\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1f420fbebe724a398e80f80f27807d9c : SUCCESS in 7m 56s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2bd1540b40ad456a97c0ee3485c622be : SUCCESS in 9m 49s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6fb20c1873bc4f3f9adf59a8737056b9 : SUCCESS in 9m 47s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0f44edd1674048a3ad07d4e921059a82 : SUCCESS in 9m 44s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b25663c0c6d74975a53c6eb62601081d : SUCCESS in 9m 08s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d940bd84ab664308a260370ea0f14251 : SUCCESS in 9m 03s\n- loci-manila https://zuul.opendev.org/t/openstack/build/78c8e9e6b4e94d498609a83486c715a0 : SUCCESS in 7m 36s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5c8626a6559f4d0599c0b70f4b9de8b8 : SUCCESS in 8m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9caa547b4fb547a7aa53ebb6817ad386 : SUCCESS in 11m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7ebe5066b81440cfab0ee7d150c8a39d : SUCCESS in 9m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0d65b06cc2f7433ab301f5e1cf72e430 : SUCCESS in 9m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b614f6ec5909465abcd607689f62b19c : SUCCESS in 9m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9d8a3427b2514f06b203e14af848fedd : SUCCESS in 7m 01s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/4a7f845cf64440a4b217f11d30aaedcf : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":4},{"id":"31b76e856458dbd752e2d0e73b0cdd0061342cc2","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2023-06-15 14:11:53.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e5be6dc218c7d6468cfcc4ae8a76a67e35c111e2","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2023-06-15 14:46:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"722c329ad942763171222260631edd6c2d1cb5eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 16:58:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6526f3d699dfe032fa79fb8d4de6eddfd763943c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 16:58:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"04e693e474e89a322dcfe0360cade5c804c78e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 17:27:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/753d7a0f0e1a4ff6a73897dcace1b822\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/cb7b6997787540cea6da1663068038c7 : SUCCESS in 10m 54s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bb964adc8ca04b1284f19ca0bc4987fa : SUCCESS in 8m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d6c89397c4db4074be2b39336e14a89d : FAILURE in 8m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e17b036a9829433fbec8051062530d37 : SUCCESS in 10m 38s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d2f4c93f7d244678ac60159cbf0ddffd : SUCCESS in 8m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/25b21b71ceca4c3a9a5a231a9d7368b5 : SUCCESS in 10m 35s\n- loci-manila https://zuul.opendev.org/t/openstack/build/49d30a86d66d475aa8929eef5f24f170 : SUCCESS in 9m 52s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/946fe6ad3de14b3dbf8c8553b519ada3 : SUCCESS in 10m 18s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6e68d0a58cab48d09f7d1473da0bcca6 : SUCCESS in 10m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/da4ad657f32b401caff4fbfc80337f4b : SUCCESS in 11m 44s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b1285eebc90c4e1ca72215978df80ce0 : SUCCESS in 10m 40s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8c5502f792ad4a4f8b01de5a1b5bef38 : SUCCESS in 9m 45s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/45ab892aca834570afdfe84d5fe1fb01 : SUCCESS in 8m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b98a47b96ada41a88a81f839f449020f : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":6},{"id":"96516d86d8118e5cd2287cefa1d1fd23355d5fd1","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-15 17:28:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b95886dbca3e3fae20181861442929318d7ed61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 18:37:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d889854f19d4cd9827713b2eadaad82\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c444065cbd3e43beb768d26255c45b64 : SUCCESS in 10m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2f3488472856445f8d34f829fa26c884 : SUCCESS in 9m 39s\n- loci-heat https://zuul.opendev.org/t/openstack/build/98264cdcdb1042d781894fa9cb9eac11 : SUCCESS in 9m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d086472ac04e43b0a0f245c1a520cab8 : SUCCESS in 9m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/058b4a4ca877462895e90d5e5c8d6b83 : SUCCESS in 10m 02s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4c946c0cc7b840d88fc1c7085e84af1f : SUCCESS in 9m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/62e4e6e11f1040c8b57ea2f6b57627d8 : SUCCESS in 9m 03s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5aa9771d56a34fe39c90258d1ef469d9 : SUCCESS in 8m 46s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7049b424de6c497c860f9c6c849a6187 : SUCCESS in 10m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fc40102c256a4dea9bad6cfb69912e56 : SUCCESS in 11m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/32e0fc44033f4f7b84e25d9aa1d6912e : SUCCESS in 11m 02s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c1e06e0cbdf64555927acf430757f097 : SUCCESS in 1h 03m 29s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/866e186374124fcba021cb63b96a3078 : SUCCESS in 8m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/eb71e188d3d84af98c58531e11d9df3f : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":6},{"id":"ed2c3ce3f9c1be0852bcd1ff833647890599a13b","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2023-06-15 18:37:51.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"463ce1889b7994120158cf7fc4cfa1bef0d3ebbb","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2023-06-16 15:30:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ec4f7df69c8ddd93cab0dfa32592aac9168cc53e","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2023-06-16 15:31:03.000000000","message":"Patch Set 6:\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"834c3eed9a3e829a30cd153e10958ff12cfaf15d","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2023-06-16 15:31:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7e4b70640cf8372d64be8d5c2c49b659e99bd7f5","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-06-16 17:25:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e051ef9d6ed99cac1d4181344b657745bf53d0e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 17:25:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"03ef9d3746bef2ae920ed4af6333c84393b3017e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 17:44:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19c43a654a04d82b0664c370178b4aa\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c1c7e08c27704b07a909ee4e8b8bc4d4 : SUCCESS in 9m 38s\n- loci-glance https://zuul.opendev.org/t/openstack/build/dff449faf6ef4d6ab5a585fb3bbf3577 : SUCCESS in 10m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a3e6af16a26043a5b42c5eef55a15c41 : SUCCESS in 7m 51s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/95d93c31feea40c18cb011923abcaea1 : SUCCESS in 10m 21s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/48890105fb1b4685b223290904ca011d : SUCCESS in 9m 11s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e0a8815ad4d740c38090ee4590e79101 : SUCCESS in 10m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/85d8352a1b814cab9a613f2e605869a6 : SUCCESS in 10m 11s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/cce64823262f4ebd80f92cbd6d757e94 : SUCCESS in 8m 05s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b5ab178ea0b147438c95750adbec71ca : SUCCESS in 11m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7a42d175205a44c68816dec62a9b4113 : SUCCESS in 11m 29s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a287bf723af94617832cca51972ee104 : SUCCESS in 8m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4da726eb47d3424cbc7bb78171d1dbd9 : SUCCESS in 8m 25s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/024685c71c554a8e917f950982a405d5 : SUCCESS in 8m 18s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d53740fd6c0c4f50a65e69d71bb64bd3 : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":6},{"id":"4a01fbe6aaf13eb2194b0441fbb454389282bca8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 17:44:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e09a032257651effdf6a4f6dd28d4ecbe08ba59d","revisions":{"d77a63a798b076630a8df19ffaabdc64bf7a130e":{"kind":"REWORK","_number":1,"created":"2023-06-15 01:36:05.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/886154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/886154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/886154/1"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"subject":"Unpin pip docker package version","message":"Unpin pip docker package version\n\nIt used to be pinned to 2.7 and\nit is incompatible with the latest\ndocker api.\n\nChange-Id: I958677e3df88028d00ac0df812c0030419d517a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d77a63a798b076630a8df19ffaabdc64bf7a130e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d77a63a798b076630a8df19ffaabdc64bf7a130e"}]},"branch":"refs/heads/master"},"4e51c12dbf96fd84c2f74bb4258ce91d880769f9":{"kind":"REWORK","_number":2,"created":"2023-06-15 02:18:55.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/886154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/886154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/886154/2"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 02:18:49.000000000","tz":180},"subject":"Unpin pip docker package version","message":"Unpin pip docker package version\n\nIt used to be pinned to 2.7 and\nit is incompatible with the latest\ndocker api.\n\nChange-Id: I958677e3df88028d00ac0df812c0030419d517a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4e51c12dbf96fd84c2f74bb4258ce91d880769f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4e51c12dbf96fd84c2f74bb4258ce91d880769f9"}]},"branch":"refs/heads/master"},"ffe4759caedbd20b54e51fe1d9979a46f09ec3b1":{"kind":"REWORK","_number":3,"created":"2023-06-15 02:36:33.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/886154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/886154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/886154/3"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 02:33:54.000000000","tz":180},"subject":"Unpin pip docker package version","message":"Unpin pip docker package version\n\nIt used to be pinned to 2.7 and it is incompatible\nwith the latest docker api.\n\nAlso disable build of Ubuntu Bionic images which\nby default used Ussuri release (not supported).\n\nChange-Id: I958677e3df88028d00ac0df812c0030419d517a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ffe4759caedbd20b54e51fe1d9979a46f09ec3b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ffe4759caedbd20b54e51fe1d9979a46f09ec3b1"}]},"branch":"refs/heads/master"},"8e8aab50f9642507b9446fb544a25b1d975be69f":{"kind":"REWORK","_number":4,"created":"2023-06-15 02:45:25.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/886154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/886154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/886154/4"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 02:44:35.000000000","tz":180},"subject":"Unpin pip docker package version","message":"Unpin pip docker package version\n\nIt used to be pinned to 2.7 and it is incompatible\nwith the latest docker api.\n\nAlso do not build Ubuntu Bionic images which\nby default use Ussuri release (not supported).\n\nChange-Id: I958677e3df88028d00ac0df812c0030419d517a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e8aab50f9642507b9446fb544a25b1d975be69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e8aab50f9642507b9446fb544a25b1d975be69f"}]},"branch":"refs/heads/master"},"6b5e6ff5a951643f4c393d8ae2371f00260b9046":{"kind":"REWORK","_number":5,"created":"2023-06-15 16:58:17.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/886154/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/886154/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/886154/5"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 16:55:41.000000000","tz":180},"subject":"Fix setup-gate.yaml playbook","message":"Fix setup-gate.yaml playbook\n\n- Use roles from zuul-jobs\n- Unpin docker python package version. It used\n  to be pinned to 2.7 and it is incompatible\n  with the latest docker api.\n- Also do not build Ubuntu Bionic images which\n  by default use Ussuri release (not supported).\n\nChange-Id: I958677e3df88028d00ac0df812c0030419d517a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b5e6ff5a951643f4c393d8ae2371f00260b9046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b5e6ff5a951643f4c393d8ae2371f00260b9046"}]},"branch":"refs/heads/master"},"e09a032257651effdf6a4f6dd28d4ecbe08ba59d":{"kind":"REWORK","_number":6,"created":"2023-06-15 16:58:52.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/886154/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/886154/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/886154/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/886154/6"}}},"commit":{"parents":[{"commit":"c152ea081b6fb523de22ae1ba914a6cf21bac9dd","subject":"Install dist packages before pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c152ea081b6fb523de22ae1ba914a6cf21bac9dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 01:34:42.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-06-15 16:58:47.000000000","tz":180},"subject":"Fix setup-gate.yaml playbook","message":"Fix setup-gate.yaml playbook\n\n- Use roles from zuul-jobs\n- Unpin docker python package version. It used\n  to be pinned to 2.7 and it is incompatible\n  with the latest docker api.\n- Also do not build Ubuntu Bionic images which\n  by default use Ussuri release (not supported).\n\nChange-Id: I958677e3df88028d00ac0df812c0030419d517a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e09a032257651effdf6a4f6dd28d4ecbe08ba59d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e09a032257651effdf6a4f6dd28d4ecbe08ba59d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
