)]}'
{"id":"openstack%2Floci~888351","triplet_id":"openstack%2Floci~master~I5211aac7b6a3bbc1f1b74e8662170dc8932525f9","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-07-12 21:52:34.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I5211aac7b6a3bbc1f1b74e8662170dc8932525f9","subject":"Fix building python-nss on Ubuntu Jammy","status":"MERGED","created":"2023-07-12 19:27:59.000000000","updated":"2023-07-12 22:13:53.000000000","submitted":"2023-07-12 22:13:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"888351","meta_rev_id":"f88e4c4bccaf5f000a28425b9e1ab14344c399ce","_number":888351,"virtual_id_number":888351,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 22:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},"all":[{"value":2,"date":"2023-07-12 21:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},"all":[{"value":1,"date":"2023-07-12 21:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 20:11:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 21:50:23.000000000","updated_by":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},"reviewer":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},"state":"REVIEWER"}],"messages":[{"id":"e4009f1deb924b1a3709800d6338af1db1f8a789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-12 19:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"911f4258d81a264dfd99f50d0832e35f5666ab91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-12 19:45:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"573ae5fcdbcc050cb5e836836486ffff047ba491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 20:11:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3256e83be93b464b83e4ee26fe650b02\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/277346514cf64c4786c8d771255af697 : SUCCESS in 17m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/19e56b2df49e4ec8b277c214ea594cfa : SUCCESS in 15m 17s\n- loci-heat https://zuul.opendev.org/t/openstack/build/78776c4c1a7746f29657280dffc13b11 : SUCCESS in 14m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/079da937738640b0bd65d14485928496 : SUCCESS in 16m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/caf4e36b5f2643ecbb22040ee3c1a88d : SUCCESS in 18m 15s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/64fbc1423da34052a175d5783024763f : SUCCESS in 16m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0fbb94e5c03040cf9fbbeb2ae15581a9 : SUCCESS in 18m 23s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/b31a19dd6aaf4a29a15810688ac3abaf : SUCCESS in 17m 56s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/af348c28f22543749bbf36a804829bb9 : SUCCESS in 13m 06s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6a7652323dbb45de9726c282742d5011 : SUCCESS in 15m 43s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/df897f622f4647f480cde948a395dacc : SUCCESS in 13m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8f0e6153dee3441bafa986cc228fb0ad : SUCCESS in 13m 13s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5863db82d4e54f9bb6748313bdca507d : SUCCESS in 12m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d9b42dc1e0c14ca9bbd25589595b7c42 : SUCCESS in 16m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4f2bdd08b15868d197761a8970325edbc8fd4863","author":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"},"date":"2023-07-12 21:50:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6111b0972f919bd861ebc6a4ebb4b1a96b91f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 21:50:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4330f69dee23acd881bc79f58978486593bd236","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-12 21:52:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26a9c140dbbad835a2440ef13ac5cc0fd23aecc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 22:13:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977a380fb4d7410d9393810d3201051b\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e5da97ee12844557a76a19a85ec01654 : SUCCESS in 15m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2db9f536f48d4084a93e953186931a00 : SUCCESS in 13m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1c1c01efd8274117866a367359ecb164 : SUCCESS in 13m 41s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/224ab3dd6c8f43e19af47a1f6389ec93 : SUCCESS in 16m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3fc9a7e2871b457c8e32ecfbd2856862 : SUCCESS in 16m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5d51052ac9734ff6a2e92a8a2f36a468 : SUCCESS in 12m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ebecbbdb19f7410fab25f35cd042ada9 : SUCCESS in 17m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/0127e2110be4420cb496b917111c3731 : SUCCESS in 14m 16s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5d04d135383148b6860ace0980adc481 : SUCCESS in 18m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d9572d2bd3a04f7aa0d6044781f984aa : SUCCESS in 17m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3afdb4f075264c87a3f2f0385eed9470 : SUCCESS in 17m 57s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9fc4f5434aba4b3e97d7ce00a3dd51f7 : SUCCESS in 15m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/871c206f724c450fadf00372ff44ada6 : SUCCESS in 15m 05s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6348065c65434023adcc33d71effb7c8 : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f88e4c4bccaf5f000a28425b9e1ab14344c399ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 22:13:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce9fea8fe62180dab0a778dfcc4c89116a945be5","revisions":{"4d6f09506d9b5624e420d05fedef952930ab469b":{"kind":"REWORK","_number":1,"created":"2023-07-12 19:27:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/51/888351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/888351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/888351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/888351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/888351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/888351/1"}}},"commit":{"parents":[{"commit":"ec70407d68c433863aadcd2b19a65862e0976b88","subject":"Build jammy images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec70407d68c433863aadcd2b19a65862e0976b88"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-12 19:22:07.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-12 19:22:07.000000000","tz":180},"subject":"Fix building python-nss on Ubuntu Jammy","message":"Fix building python-nss on Ubuntu Jammy\n\nThe libnss3 headers in Ubuntu Jammy are compatible\nwith python-nss\u003d\u003d\u003d1.0.1. Ubuntu Jammy itself\nprovides the binary package python-nss and\nthey apply the patch that renames RSAPublicKey/DSAPublicKey\ntypes into PyRSAPublicKey/PyDSAPublicKey.\n\nThis change applies the same patch before\nbuilding python-nss wheel.\n\nChange-Id: I5211aac7b6a3bbc1f1b74e8662170dc8932525f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4d6f09506d9b5624e420d05fedef952930ab469b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4d6f09506d9b5624e420d05fedef952930ab469b"}]},"branch":"refs/heads/master"},"ce9fea8fe62180dab0a778dfcc4c89116a945be5":{"kind":"REWORK","_number":2,"created":"2023-07-12 19:45:30.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/51/888351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/888351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/888351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/888351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/888351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/888351/2"}}},"commit":{"parents":[{"commit":"ec70407d68c433863aadcd2b19a65862e0976b88","subject":"Build jammy images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec70407d68c433863aadcd2b19a65862e0976b88"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-12 19:22:07.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-12 19:45:24.000000000","tz":180},"subject":"Fix building python-nss on Ubuntu Jammy","message":"Fix building python-nss on Ubuntu Jammy\n\nThe libnss3 headers in Ubuntu Jammy are compatible\nwith python-nss\u003d\u003d\u003d1.0.1. Ubuntu Jammy itself\nprovides the binary package python-nss and\nthey apply the patch that renames RSAPublicKey/DSAPublicKey\ntypes into PyRSAPublicKey/PyDSAPublicKey.\n\nThis change applies the same patch before\nbuilding python-nss wheel.\n\nChange-Id: I5211aac7b6a3bbc1f1b74e8662170dc8932525f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce9fea8fe62180dab0a778dfcc4c89116a945be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce9fea8fe62180dab0a778dfcc4c89116a945be5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":33330,"name":"Karl Kloppenborg","email":"k@rl.ag","username":"karl_kloppenborg"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
