)]}'
{"id":"openstack%2Floci~890531","triplet_id":"openstack%2Floci~master~I244c29844d3ceb29b73e9a60b284d900bbeeb4ec","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-08-15 22:49:12.000000000","reason":"Change was submitted"},"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2023-08-15 22:49:12.000000000","reason":"Change was submitted"},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2023-08-15 22:49:12.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2023-08-15 22:49:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I244c29844d3ceb29b73e9a60b284d900bbeeb4ec","subject":"Copy default configuration files to /etc/","status":"MERGED","created":"2023-08-04 14:54:33.000000000","updated":"2023-08-15 22:49:12.000000000","submitted":"2023-08-15 22:49:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890531","meta_rev_id":"591b79d873f01c70c87eeabc703af8f4e93621a7","_number":890531,"virtual_id_number":890531,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-15 22:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2023-08-15 22:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2023-08-15 22:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2023-08-15 22:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 15:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 16:33:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2023-08-04 17:20:38.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"CC"},{"updated":"2023-08-09 09:21:55.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2023-08-15 22:21:11.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"fde82cf84584a79804f7403c8719a5f3f30a1e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-08-04 14:54:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3decfc52d88432b9ffc4ac8505ef9f4b10ef7da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 15:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53e7725d732245b7a2634ab36d3c8e4e\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3ac99cd43e5f480f8aeb57cb5e09d52a : SUCCESS in 15m 18s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4f938bd6f5e64e7abfd26e0cac773281 : SUCCESS in 17m 05s\n- loci-heat https://zuul.opendev.org/t/openstack/build/09a49ee3f2a048278960f8eb4c714c6b : SUCCESS in 23m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/17187ac96e2a4209a75ca20f777f2eab : SUCCESS in 20m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/bc85d0e69bf949e3828690b6d3badc32 : SUCCESS in 15m 11s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/77db639d0be549f8a1e147b9a90193d7 : SUCCESS in 16m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/79f429352c2e4eff89a9d7bb15ffe3c5 : SUCCESS in 19m 51s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/e335b0f2674b4b8c9011d924fc4ad770 : SUCCESS in 16m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2368633f3dc84e8a8ee64dc0c931d074 : SUCCESS in 20m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3acb1f5017334cf383639bb57347169a : SUCCESS in 20m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8d9658414ae84968986235b174edbb59 : SUCCESS in 16m 11s\n- loci-placement https://zuul.opendev.org/t/openstack/build/831fd338ef0446279e269f530d798c58 : SUCCESS in 14m 28s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4a0d653c02d3400c81e224dd6333215d : SUCCESS in 12m 47s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6cebed0036b9432599fef1f78380f14c : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"707016d5ab3cf9fb28387f2a79b80ba2fa67b068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-08-04 16:17:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b78e022370ab38ac2f8b31695c89376d74a6d42f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-08-04 16:33:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c0a9a1cb3d8c9f951da149a3c09d046d1b4ee89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:44:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0eeba3613cb4b09993f4740d3b7fd65\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8e981ccd7c044d06816dd28836a2e741 : SUCCESS in 18m 05s\n- loci-glance https://zuul.opendev.org/t/openstack/build/05e2bde2a980448891097f6a3c72a4fb : SUCCESS in 14m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a4b6ebb9a1d0429994cc051b08ed6e3f : SUCCESS in 17m 14s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/453bc5eec47a4176a6210fa4c4b896ee : SUCCESS in 14m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a280422fb0ae4e41b02619606e9eb512 : SUCCESS in 15m 03s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3412362108a54b0ab2ff56002926fe87 : SUCCESS in 17m 37s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f38fbb05bd1b4d3d8c54460917321984 : SUCCESS in 14m 44s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/1c41395f158042d7af2316b2854d7861 : SUCCESS in 14m 42s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1e753feccad74c288be27d8a1ca7f8f4 : SUCCESS in 15m 08s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b005fb241fb34bfe94fdac789093cbd7 : SUCCESS in 21m 06s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b2a23d00d48640d7932ae2a1fff67f15 : SUCCESS in 19m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6f2a4e0c74f14641a252dd92d89b8d25 : SUCCESS in 17m 10s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/32e28ff2f6ef413bab3a2a164d4461e2 : SUCCESS in 11m 40s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/53cd573f74d64c13ad762cbbd4d8d3e0 : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":2},{"id":"95af72d84639c15c738a9657c5e180c01947f94c","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-08-04 17:20:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3fb1a8acbfc7f174f26972235f93182a73a798a0","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-08-07 14:48:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bf08ed24b24c4750adead3b335af76049822d560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-08-07 14:48:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c4747d06ca48861f73f088aabb9aeb0d109721d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 15:19:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f449f53bed4e4a2993d1accd5fc13553\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e06ca3e98e8245f0afe16b9185c68e8a : SUCCESS in 19m 00s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7955dbbda003464e9d3e4577a49f4363 : SUCCESS in 17m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7451c28ba0194054af21cef735eab8f3 : SUCCESS in 23m 00s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b7c9fd3842984b488d15e2bea2d02482 : SUCCESS in 17m 24s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0dde6268c38e455196004f6b9f753223 : SUCCESS in 15m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/589e73f86fd2403fab464aa2bbad3368 : SUCCESS in 16m 39s\n- loci-manila https://zuul.opendev.org/t/openstack/build/746fda50aa664934be8743cb74704c86 : SUCCESS in 16m 01s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/c1b7a92234ad4427afefbb065aa3bf0d : SUCCESS in 17m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/eb30fc9807d24477b4f5ab59a96231a7 : SUCCESS in 15m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a20f640c51484b1b8dc702d14c62229d : SUCCESS in 17m 21s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/429f1a58899744b2a7707a11dba6dbc0 : FAILURE in 13m 39s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c317b2ab1ac44bc58174edd7027baf7f : SUCCESS in 17m 18s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5c8ca923c7fb41bdb5efac2894c93679 : SUCCESS in 13m 55s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b60679ed4073489481d9f524c8e7bd75 : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":3},{"id":"151c4ee0d7b526f4729bdb1df0a5d68d66b5e149","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-08-08 08:43:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c54b5c7209172bed05958fb4b8841d634d51b303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 09:16:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b310dae4e245ceb263198eeacdbdad\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f85fa35ae5ef459096874cdc8bc929da : SUCCESS in 19m 47s\n- loci-glance https://zuul.opendev.org/t/openstack/build/936d270f94de4d4cac0ee9f3a832162b : SUCCESS in 19m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a82b0f6e716b49b3b3998bc686df0477 : SUCCESS in 20m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ac8a53c2804b4422a86f3af3365746fc : SUCCESS in 18m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1b355f715f454cffb8040ff0294efe38 : SUCCESS in 19m 51s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3d941bcb2c93434bb5362700ade00bd6 : SUCCESS in 19m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/883324622b6c4f8a8443384cb8d4c3e1 : SUCCESS in 17m 39s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/3cffada4634448beb07f6f74297bf0fd : SUCCESS in 18m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/875f0a837a9b4f37bc889203bb00f240 : SUCCESS in 19m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3a18394b9b824af594e4651cbbfdd31f : SUCCESS in 21m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5f198dabfc6444ae83ccb8707f82a1d6 : SUCCESS in 19m 09s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c0ef93b384394aee9532704e6a7e1773 : SUCCESS in 16m 14s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d1ac6e30227742d187eabdb11961f695 : SUCCESS in 12m 54s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/a05080b5e8af4ba78fe9cac2e3a1dae6 : SUCCESS in 16m 55s","accounts_in_message":[],"_revision_number":3},{"id":"7c9365a04f5d4c8afbf53465a27ffb53272df894","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-08-09 09:21:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d171e761f097b50882e796e7eda029217f818437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-08-15 21:33:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"810d736d665578798c16fc2b8671749b84d6e8c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-08-15 21:40:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0c69de0a8f7f2322f3d96f963cb62bdb44e7d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 22:05:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c03093ec5a4a4ea78408abcff233af\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ed84e2f343b947679c8f1be19b989d9c : SUCCESS in 17m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/97d989f5b2a4443d8dec2840e2762a63 : SUCCESS in 17m 05s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5e32db7f8a4b409db1d9579c4e347c32 : SUCCESS in 18m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7a45566320fa459cb64fb1ede3865c16 : SUCCESS in 17m 09s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9884c20843674164b6d8cba7ffb81fec : SUCCESS in 16m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5614b95c03644c448ccfc54e018ed175 : SUCCESS in 15m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/137c0a4fd9584a318b534321d1f384d4 : SUCCESS in 16m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/27ce6ef1f8d440668c0e9ad0ec15a92d : SUCCESS in 15m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f784489c391e44268f0fbd5743d3f07b : SUCCESS in 24m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3cd9fd19c7ce4918a79e355181e6445c : SUCCESS in 16m 26s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b398fd51826e4e41b7cfc754d74d0d54 : SUCCESS in 17m 38s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5239c9cbaff546e5a06fbcaf2a44ff81 : SUCCESS in 17m 11s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d494b458a4804364bdb9c041b708d768 : SUCCESS in 12m 28s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/ced77fc39c6041cebcae16afae0ce81f : SUCCESS in 17m 53s","accounts_in_message":[],"_revision_number":5},{"id":"f8ace6ab54791ffee0073d652eecb6c315f0c95a","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-08-15 22:11:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"aab9a5ec30bd63c622019b5661d686928f185ecc","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2023-08-15 22:21:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"941bc9689d188a3d64144a1bc5a455d7a6ad2e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 22:21:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d152889b57bbbcb540012ed2d6f9ed80690c4525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 22:49:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4e258928f2409ba1918896fecdd0f1\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/932e6cf461574876990e8523526df55d : SUCCESS in 15m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1fa181bd1f1f4cff8f41f401bd101122 : SUCCESS in 16m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b420bdfbed84458384554bba4386b6d1 : SUCCESS in 17m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7c42a2784952499c9c377c59773a874a : SUCCESS in 19m 28s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b9b0fbddb63649f9aac804f937871666 : SUCCESS in 14m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/90055cff5e634f29bc5078999f9702d1 : SUCCESS in 14m 31s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f577ccb863e045c68be8f98647fb1538 : SUCCESS in 19m 00s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/ef4c3dedc0464d5abbadc2fe8a31df35 : SUCCESS in 12m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/81b39384cc00456f941418532e2c4632 : SUCCESS in 14m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/42c12f5c0a354ab585b14457d42d93b4 : SUCCESS in 19m 23s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9e989d3e7ce34486a02b6c011cd67a57 : SUCCESS in 18m 55s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b92d1227b40d4e52ac4d006a07e82b52 : SUCCESS in 12m 24s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/010202bba69b470b8f8560aad46fdd29 : SUCCESS in 9m 55s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8f5ab21208f3424db8fcdb88589c8e39 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":5},{"id":"591b79d873f01c70c87eeabc703af8f4e93621a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 22:49:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"82082bf82de53846b228e62fa154bc011c4f3497","revisions":{"76459caf38f451c483aefc208a95162cdb2f0c0c":{"kind":"REWORK","_number":1,"created":"2023-08-04 14:54:33.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/31/890531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/890531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/890531/1"}}},"commit":{"parents":[{"commit":"2f88cf46441bae413787b6c478c13f478e7f8cce","subject":"Fix build of those wheels which depend on python-nss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2f88cf46441bae413787b6c478c13f478e7f8cce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 14:53:47.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 14:53:47.000000000","tz":600},"subject":"Copy default configuration files to /etc/","message":"Copy default configuration files to /etc/\n\nChange-Id: I244c29844d3ceb29b73e9a60b284d900bbeeb4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76459caf38f451c483aefc208a95162cdb2f0c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76459caf38f451c483aefc208a95162cdb2f0c0c"}]},"branch":"refs/heads/master"},"6d2f601cfc734fe2f2d379393e5d0cfca007bfeb":{"kind":"REWORK","_number":2,"created":"2023-08-04 16:17:16.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/31/890531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/890531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/890531/2"}}},"commit":{"parents":[{"commit":"2f88cf46441bae413787b6c478c13f478e7f8cce","subject":"Fix build of those wheels which depend on python-nss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2f88cf46441bae413787b6c478c13f478e7f8cce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 14:53:47.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 16:17:00.000000000","tz":600},"subject":"Copy default configuration files to /etc/","message":"Copy default configuration files to /etc/\n\nChange-Id: I244c29844d3ceb29b73e9a60b284d900bbeeb4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6d2f601cfc734fe2f2d379393e5d0cfca007bfeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6d2f601cfc734fe2f2d379393e5d0cfca007bfeb"}]},"branch":"refs/heads/master"},"5fc14ac7b230bd694f21c68ae5501e00c423dd89":{"kind":"REWORK","_number":3,"created":"2023-08-07 14:48:41.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/31/890531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/890531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/890531/3"}}},"commit":{"parents":[{"commit":"2f88cf46441bae413787b6c478c13f478e7f8cce","subject":"Fix build of those wheels which depend on python-nss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2f88cf46441bae413787b6c478c13f478e7f8cce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 14:53:47.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-07 14:48:20.000000000","tz":600},"subject":"Copy default configuration files to /etc/","message":"Copy default configuration files to /etc/\n\nChange-Id: I244c29844d3ceb29b73e9a60b284d900bbeeb4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5fc14ac7b230bd694f21c68ae5501e00c423dd89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5fc14ac7b230bd694f21c68ae5501e00c423dd89"}]},"branch":"refs/heads/master"},"4ca79352f770cb8e40efd55243a3b64e7e036bcf":{"kind":"REWORK","_number":4,"created":"2023-08-15 21:33:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/890531/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/890531/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/890531/4"}}},"commit":{"parents":[{"commit":"2f88cf46441bae413787b6c478c13f478e7f8cce","subject":"Fix build of those wheels which depend on python-nss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2f88cf46441bae413787b6c478c13f478e7f8cce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 14:53:47.000000000","tz":600},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-08-15 21:28:57.000000000","tz":180},"subject":"Copy default configuration files to /etc/","message":"Copy default configuration files to /etc/\n\nThis will allow us to shorten the values.yaml\nin many cases.\n\nThis is disabled by default. Set COPY_DEFAULT_CONFIG_FILES\u003dyes\nto enable it.\n\nChange-Id: I244c29844d3ceb29b73e9a60b284d900bbeeb4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4ca79352f770cb8e40efd55243a3b64e7e036bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4ca79352f770cb8e40efd55243a3b64e7e036bcf"}]},"branch":"refs/heads/master"},"82082bf82de53846b228e62fa154bc011c4f3497":{"kind":"REWORK","_number":5,"created":"2023-08-15 21:40:04.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/890531/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/890531/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/890531/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/890531/5"}}},"commit":{"parents":[{"commit":"2f88cf46441bae413787b6c478c13f478e7f8cce","subject":"Fix build of those wheels which depend on python-nss","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2f88cf46441bae413787b6c478c13f478e7f8cce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2023-08-04 14:53:47.000000000","tz":600},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-08-15 21:39:53.000000000","tz":180},"subject":"Copy default configuration files to /etc/","message":"Copy default configuration files to /etc/\n\nThis will allow us to shorten the values.yaml\nin many cases.\n\nThis is disabled by default. Set COPY_DEFAULT_CONFIG_FILES\u003dyes\nto enable it.\n\nChange-Id: I244c29844d3ceb29b73e9a60b284d900bbeeb4ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/82082bf82de53846b228e62fa154bc011c4f3497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/82082bf82de53846b228e62fa154bc011c4f3497"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
