)]}'
{"id":"openstack%2Floci~902733","triplet_id":"openstack%2Floci~master~I6fa07a45f491e7101f85cb90731611409a279719","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-12-06 05:02:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fa07a45f491e7101f85cb90731611409a279719","subject":"Add nodesets.yaml and stop using tmpfs for docker","status":"MERGED","created":"2023-12-05 22:28:06.000000000","updated":"2023-12-06 05:02:23.000000000","submitted":"2023-12-06 05:02:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902733","meta_rev_id":"e1235540984d624fe613fc486399a5c4d98bbaeb","_number":902733,"virtual_id_number":902733,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 05:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2023-12-06 04:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2023-12-06 04:30:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 23:40:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-06 04:30:37.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"c44de7c9f8861164074f86e4d5aba8182b879c9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-12-05 22:28:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d766a3e8d223b774494d0a9d6604676bc946bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-12-05 22:29:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fd9587ecf7622395368f19ac920f065f0e637ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-12-05 22:58:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"34e067a36e88f95844473ab7164feee0e1395bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-12-05 23:07:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"57fc3d74966d7a44fbbe384b544f4d525738237d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 23:40:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76910367215d42f5aacf0938b50076ab\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1c3fb0c232544c4ba221d5e337074566 : SUCCESS in 17m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9def4d7b05b04aeaa1f62b3617b15d5c : SUCCESS in 15m 13s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d8a1ae68779c499db78d821cb8b2225b : SUCCESS in 15m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6e7350e71d924a64a0a07398c628c17c : SUCCESS in 19m 37s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c6d79e7e4dfa4f54b9329ddfafaf883c : SUCCESS in 18m 55s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a1195d65b86343869a95d385de710fa8 : SUCCESS in 19m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0c54ccf1d480468a8a8e0de7a8f9e471 : SUCCESS in 20m 51s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/c032ae960a7e4ed890b7b7e03dbfc1c1 : SUCCESS in 21m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9ee0be94a8424df9a26256cd5f426746 : SUCCESS in 24m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c7f694348bed497a8335bcbb230919d5 : SUCCESS in 24m 44s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a8a9ae64c0f843908fad0a305efe6a0d : SUCCESS in 25m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f2b1f55e50d44188bc7b9781e792d928 : SUCCESS in 19m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e55b6ad837b144428d8a2a147f2c4b5e : SUCCESS in 14m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/aa9f547ddec94941ad12f3d15740e072 : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":4},{"id":"9347abec533126262f21df0bee7a3ed52d97ef25","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2023-12-06 04:30:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"27e4cd49e112ad9d02adffa188e89dcb0527fe84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 04:31:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b8dadb0bbf9e518be584913b8aaa22667fa00436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 05:02:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9051e8fff73b449fb912e7ee85696d09\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/eb1c9622bbcb459e86b8567653be284c : SUCCESS in 21m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6bc193d6ec454753b5111777cf68d315 : SUCCESS in 25m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8b9e990e753d4a02a8e5721e26338d36 : SUCCESS in 18m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a61dbdaf6e364200a70d0939c8cbca59 : SUCCESS in 19m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0b9a5ea0d0724bb7ab8703530d376121 : SUCCESS in 18m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ae341d5617d047d68a2753aaaa784da8 : SUCCESS in 15m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a5b91bb647684de2a4ddfc2741c28f6e : SUCCESS in 19m 45s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/7fcf2c5643ea45b39ed2a2d0325fcd04 : SUCCESS in 18m 16s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/bcf8578abe69462184e09288ace0682b : SUCCESS in 20m 56s\n- loci-nova https://zuul.opendev.org/t/openstack/build/650717ea127c4b8698b1a10865c17f3c : SUCCESS in 16m 23s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/63262c1326844b8b869421cf48e6bc19 : SUCCESS in 18m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/554dbb744d93488888d3cfd1df646c51 : SUCCESS in 14m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/726a0dff90014209bc5209f540dbf02b : SUCCESS in 13m 38s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/65ae3029bb61466eb1eb64513ba650e3 : SUCCESS in 21m 18s","accounts_in_message":[],"_revision_number":4},{"id":"e1235540984d624fe613fc486399a5c4d98bbaeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 05:02:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6537fe480e058c62271b1f9ff09508ca42dc4885","revisions":{"4a2f786640c13228192dcebcda7d2921620f6e9f":{"kind":"REWORK","_number":1,"created":"2023-12-05 22:28:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/33/902733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/33/902733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/33/902733/1"}}},"commit":{"parents":[{"commit":"235b8e9d65d52964a6094ce867a3eed4752dbbe4","subject":"Merge \"Remove pyldap from pydep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/235b8e9d65d52964a6094ce867a3eed4752dbbe4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:27:40.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:27:40.000000000","tz":-360},"subject":"Add nodesets.yaml","message":"Add nodesets.yaml\n\nChange-Id: I6fa07a45f491e7101f85cb90731611409a279719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4a2f786640c13228192dcebcda7d2921620f6e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4a2f786640c13228192dcebcda7d2921620f6e9f"}]},"branch":"refs/heads/master"},"49c60fe41ab33b96b53ffd2aee93f60ffcca8f43":{"kind":"REWORK","_number":2,"created":"2023-12-05 22:29:27.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/33/902733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/33/902733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/33/902733/2"}}},"commit":{"parents":[{"commit":"235b8e9d65d52964a6094ce867a3eed4752dbbe4","subject":"Merge \"Remove pyldap from pydep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/235b8e9d65d52964a6094ce867a3eed4752dbbe4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:27:40.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:29:21.000000000","tz":-360},"subject":"Add nodesets.yaml","message":"Add nodesets.yaml\n\nChange-Id: I6fa07a45f491e7101f85cb90731611409a279719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/49c60fe41ab33b96b53ffd2aee93f60ffcca8f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/49c60fe41ab33b96b53ffd2aee93f60ffcca8f43"}]},"branch":"refs/heads/master"},"875d2b569d68a7db9a7e6da1cd3963a102b0bc05":{"kind":"REWORK","_number":3,"created":"2023-12-05 22:58:35.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/33/902733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/33/902733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/33/902733/3"}}},"commit":{"parents":[{"commit":"235b8e9d65d52964a6094ce867a3eed4752dbbe4","subject":"Merge \"Remove pyldap from pydep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/235b8e9d65d52964a6094ce867a3eed4752dbbe4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:27:40.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:58:05.000000000","tz":-360},"subject":"Add nodesets.yaml and set smaller swap","message":"Add nodesets.yaml and set smaller swap\n\nChange-Id: I6fa07a45f491e7101f85cb90731611409a279719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/875d2b569d68a7db9a7e6da1cd3963a102b0bc05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/875d2b569d68a7db9a7e6da1cd3963a102b0bc05"}]},"branch":"refs/heads/master"},"6537fe480e058c62271b1f9ff09508ca42dc4885":{"kind":"REWORK","_number":4,"created":"2023-12-05 23:07:09.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/33/902733/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/33/902733/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/902733/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/33/902733/4"}}},"commit":{"parents":[{"commit":"235b8e9d65d52964a6094ce867a3eed4752dbbe4","subject":"Merge \"Remove pyldap from pydep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/235b8e9d65d52964a6094ce867a3eed4752dbbe4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 22:27:40.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-12-05 23:06:42.000000000","tz":-360},"subject":"Add nodesets.yaml and stop using tmpfs for docker","message":"Add nodesets.yaml and stop using tmpfs for docker\n\nChange-Id: I6fa07a45f491e7101f85cb90731611409a279719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6537fe480e058c62271b1f9ff09508ca42dc4885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6537fe480e058c62271b1f9ff09508ca42dc4885"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
