)]}'
{"id":"openstack%2Floci~917271","triplet_id":"openstack%2Floci~master~Icfe563f1a0f780cc2315c3107c62744141100a00","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Icfe563f1a0f780cc2315c3107c62744141100a00","subject":"[WIP] Debug image building","status":"ABANDONED","created":"2024-04-28 01:04:11.000000000","updated":"2024-04-28 06:37:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"d9a988db8d6b5cd3ad706d15693daf2062f4b5e3","_number":917271,"virtual_id_number":917271,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-28 06:31:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-04-28 01:32:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b57fc3643ba425950048c59c5477772498b3ca55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-28 01:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e9f4f392016b30b35c6746a5de233b6ae456244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 01:32:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01f3831c1dd24743866b34dd3c399a24\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/cdcd62df395f4061a5a85c996a135760 : SUCCESS in 16m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1b2db3d6d31349e1b823964f727e6143 : SUCCESS in 21m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/034c9334b3b6492590cf814460cdf707 : SUCCESS in 16m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ad62ca68466e4524b4f9454fe9c26ed9 : SUCCESS in 20m 37s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a69fcd99a2c54a71b4191696b511c25a : SUCCESS in 15m 21s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bd13ed9ca0f3440dad595bb29640ee31 : SUCCESS in 19m 44s\n- loci-manila https://zuul.opendev.org/t/openstack/build/aa83bc686929497996c4d6c88fd8f11d : SUCCESS in 15m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/aaf68fe1773c4490af103cfaf69963b9 : SUCCESS in 20m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/548edacde0b6470a80f78b0cb069b2db : SUCCESS in 17m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/14f596a55d2645ce8a18317caa28f8d1 : SUCCESS in 22m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e9a8900f4ece443f916c12a264464276 : SUCCESS in 20m 51s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1100e1215a7c47f0a4049b8e701a2752 : SUCCESS in 16m 20s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/205451ec489a4ffdb6696ca1cfd9033b : SUCCESS in 16m 50s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/edcbf95a29c7425db47164b66d1293ac : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"584e46756bd0dbb5398d8df941ded1996cf98683","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-28 02:10:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"caa3a636cfd37877d57a5e8639f3b16d1e231006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 02:58:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83249857a44a4fdebd5bafa85d413ba7\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0d476b3594ca41b784a2e9a3ab78cc8e : SUCCESS in 21m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e74c0748fe2e42c1a5444a18a4adbdee : SUCCESS in 17m 48s\n- loci-heat https://zuul.opendev.org/t/openstack/build/93807acec97d4e7c9a1596de3d88c5f9 : SUCCESS in 20m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/256518ebbcdb41758031cf22c537fc37 : SUCCESS in 18m 05s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3eeec828aac540c59c37186e8d29877e : SUCCESS in 19m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e6b67788651640a9a46eabdc80d00c43 : SUCCESS in 17m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/261971e56ffa4eb9bbf25a673f5bb304 : SUCCESS in 15m 27s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/c2de1076b46a429c8003a420c44e5c4b : SUCCESS in 17m 36s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/131625d073f84a009238dd20954d6b5e : SUCCESS in 23m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d1ba8b8e39994ec5a48516158c986f03 : SUCCESS in 25m 45s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4c79e7038251425590b0e2772c853e87 : SUCCESS in 20m 57s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c93d3f3d7f2d4e80a5841913090fbbcc : SUCCESS in 18m 26s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/511970a46cb143ac913c6ebe6d91261b : SUCCESS in 14m 58s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/01491b75793c4d68861d5fc3c69cdb40 : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":2},{"id":"11d48f9b7e84d170b35ffbbc27f228a7cc2f55c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-28 04:33:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4a29d24582e6a351f3d2af1e1ae67a8b52d8009b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 05:02:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a83e95eb4c84c499d566ff1241c4c23\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/670b96d5909c4ab3a9099875d50d3f0b : SUCCESS in 21m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/78c1dd83d8834b61be2406d88bde83b9 : SUCCESS in 22m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/28d62c743aaa4faeaabd95897f8eb78e : SUCCESS in 22m 52s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7c982f3d5f1b475cac48cbd4b65e9ccb : SUCCESS in 21m 11s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/14807087aa984bd1b438d61fb5e66a9b : SUCCESS in 18m 51s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9a70d6d90f3e47c8a840e392eac21541 : SUCCESS in 20m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6b1b2947cc49442c97cbe579d3d0cbe6 : SUCCESS in 14m 46s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/58a5103ab02147fc84e24cc755bfa2bc : SUCCESS in 18m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b90e901325d4496e8c9af3c4880b36a7 : SUCCESS in 15m 35s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2642674373cc41e6b0f6a35a91de7538 : SUCCESS in 15m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/dea25547ceae41188cdea6a095f90e95 : SUCCESS in 14m 45s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c0f5dde20eaf41e08ecc2f8eb79a7a73 : SUCCESS in 13m 53s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6848037f046642b9b04b9ac5607553c2 : SUCCESS in 10m 43s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/adcb6796408b44a09523163303c90075 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"c761ead1eea9df068e9dafd2d34c454e1a58422e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-28 05:06:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bdc069b65d642d44b4b49cda9694184ae0bae069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 05:48:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9bf9101e5144acb9d3f9140a387deb5\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dc7455f9c9fc4ca79866885f354c8cbd : SUCCESS in 17m 52s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4aa2a389f1ef400faac79610f4f12138 : SUCCESS in 20m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/033a350d94a4425ab352734549edce86 : SUCCESS in 20m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c250852058444d0c9500a89526b4c43b : SUCCESS in 15m 07s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/09b70b1450614402a4ed6727c7a44cb8 : SUCCESS in 16m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a72b1a3a83f04579ba366fe984204f67 : SUCCESS in 17m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6bf7197e0893447986c4912c9c84bec6 : SUCCESS in 17m 37s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/eea3f0aa4238405c8caee8c1a489956e : SUCCESS in 15m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/69d6c0537a2e4f67bcde2d9127c7cbf0 : SUCCESS in 15m 05s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a0a80662317345489a8a02d854d6cff2 : SUCCESS in 26m 32s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fc88e35225684193b326a70b438ea32b : SUCCESS in 17m 28s\n- loci-placement https://zuul.opendev.org/t/openstack/build/89f364b82be64f5b9784bf4513c938a9 : SUCCESS in 19m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f5631c00df1e4c5085e239756dc12992 : SUCCESS in 14m 06s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d19d4ec4e7664326b96a52adf1706b55 : SUCCESS in 19m 33s","accounts_in_message":[],"_revision_number":4},{"id":"4cc0749a3943c318f973c5624b33593335c2537c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-28 06:01:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d1e70654084e18381c80ea67fa59e533c442126a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-28 06:31:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/488989e9a5cf4949b090d10e46af95cd\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/64313b3a4d3c4b589de055303a3abe2a : FAILURE in 12m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2188ad0d746746609ba180d1cbf9b98e : FAILURE in 14m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/17faca3c51724266ace96489222280b5 : FAILURE in 12m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6adcf41a1dc149a1aa195357a2b1318e : FAILURE in 12m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/131f0f0db91340369a385ab6fd90e0a9 : FAILURE in 18m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a7b5ac0d2a8d4d41a4a30367bba5d171 : FAILURE in 12m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/386546fa6906411c82bac3e26e849c8c : FAILURE in 16m 09s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/bc9c9c2577cf4797a92b74e710424785 : FAILURE in 12m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a769dc69d8bb423bb5550f691584d571 : FAILURE in 17m 37s\n- loci-nova https://zuul.opendev.org/t/openstack/build/91386ce7939146d1b4b6d7cecd95e3dd : FAILURE in 14m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2ca5b42b0bd5419b98f8d6c163a60ebc : FAILURE in 16m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a87876e130434abc95fa5df72cb95178 : FAILURE in 16m 42s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0caa35643a8641efb2d8fd9756c15581 : SUCCESS in 13m 40s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/8b07c87f18fe4fd0b8968a697759d4e6 : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":5},{"id":"d9a988db8d6b5cd3ad706d15693daf2062f4b5e3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-04-28 06:37:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f1c7b55c515adbf8ce51d0d69b7b9ea0887c537c","revisions":{"8f91aee0298edb26c88dd86acdae3a12249d413f":{"kind":"REWORK","_number":1,"created":"2024-04-28 01:04:11.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/917271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/71/917271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/71/917271/1"}}},"commit":{"parents":[{"commit":"6aa92e029da939d9df55abf3de6cb4628aa77f76","subject":"Stop building Ubuntu Focal images for master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6aa92e029da939d9df55abf3de6cb4628aa77f76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 01:03:02.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 01:03:02.000000000","tz":-300},"subject":"[WIP] Debug image building","message":"[WIP] Debug image building\n\nChange-Id: Icfe563f1a0f780cc2315c3107c62744141100a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8f91aee0298edb26c88dd86acdae3a12249d413f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8f91aee0298edb26c88dd86acdae3a12249d413f"}]},"branch":"refs/heads/master"},"6b51273f1e38886a7f282af78da5972b7fff983b":{"kind":"REWORK","_number":2,"created":"2024-04-28 02:10:10.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/917271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/71/917271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/71/917271/2"}}},"commit":{"parents":[{"commit":"6aa92e029da939d9df55abf3de6cb4628aa77f76","subject":"Stop building Ubuntu Focal images for master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6aa92e029da939d9df55abf3de6cb4628aa77f76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 01:03:02.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 02:10:04.000000000","tz":-300},"subject":"[WIP] Debug image building","message":"[WIP] Debug image building\n\nChange-Id: Icfe563f1a0f780cc2315c3107c62744141100a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b51273f1e38886a7f282af78da5972b7fff983b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b51273f1e38886a7f282af78da5972b7fff983b"}]},"branch":"refs/heads/master"},"5c59f877803af74087ea4a6b919323b7d61daa43":{"kind":"REWORK","_number":3,"created":"2024-04-28 04:33:29.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/917271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/71/917271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/71/917271/3"}}},"commit":{"parents":[{"commit":"6aa92e029da939d9df55abf3de6cb4628aa77f76","subject":"Stop building Ubuntu Focal images for master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6aa92e029da939d9df55abf3de6cb4628aa77f76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 01:03:02.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 04:33:23.000000000","tz":-300},"subject":"[WIP] Debug image building","message":"[WIP] Debug image building\n\nChange-Id: Icfe563f1a0f780cc2315c3107c62744141100a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c59f877803af74087ea4a6b919323b7d61daa43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c59f877803af74087ea4a6b919323b7d61daa43"}]},"branch":"refs/heads/master"},"5df99b4bbb1f1c2cf0acfc9d2991759a6fb85c52":{"kind":"REWORK","_number":4,"created":"2024-04-28 05:06:35.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/917271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/71/917271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/71/917271/4"}}},"commit":{"parents":[{"commit":"6aa92e029da939d9df55abf3de6cb4628aa77f76","subject":"Stop building Ubuntu Focal images for master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6aa92e029da939d9df55abf3de6cb4628aa77f76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 01:03:02.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 05:06:30.000000000","tz":-300},"subject":"[WIP] Debug image building","message":"[WIP] Debug image building\n\nChange-Id: Icfe563f1a0f780cc2315c3107c62744141100a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5df99b4bbb1f1c2cf0acfc9d2991759a6fb85c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5df99b4bbb1f1c2cf0acfc9d2991759a6fb85c52"}]},"branch":"refs/heads/master"},"f1c7b55c515adbf8ce51d0d69b7b9ea0887c537c":{"kind":"REWORK","_number":5,"created":"2024-04-28 06:01:50.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/71/917271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/71/917271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/71/917271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/71/917271/5"}}},"commit":{"parents":[{"commit":"6aa92e029da939d9df55abf3de6cb4628aa77f76","subject":"Stop building Ubuntu Focal images for master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6aa92e029da939d9df55abf3de6cb4628aa77f76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 01:03:02.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-04-28 06:01:44.000000000","tz":-300},"subject":"[WIP] Debug image building","message":"[WIP] Debug image building\n\nChange-Id: Icfe563f1a0f780cc2315c3107c62744141100a00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f1c7b55c515adbf8ce51d0d69b7b9ea0887c537c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f1c7b55c515adbf8ce51d0d69b7b9ea0887c537c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
