)]}'
{"id":"openstack%2Floci~922253","triplet_id":"openstack%2Floci~master~I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2024-06-20 15:03:57.000000000","reason":"Change was submitted"},"8898":{"account":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"last_update":"2024-06-20 15:03:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b","subject":"Add networking-baremetal to pydep.txt","status":"MERGED","created":"2024-06-18 20:08:06.000000000","updated":"2024-06-20 15:03:57.000000000","submitted":"2024-06-20 15:03:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922253","meta_rev_id":"c2db58b411ceaa8b63957254f6102e856712ce93","_number":922253,"virtual_id_number":922253,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-20 15:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-06-20 10:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":2,"date":"2024-06-20 10:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-20 14:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2024-06-20 14:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-18 20:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 17:49:21.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2024-06-19 21:05:30.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2024-06-20 14:38:33.000000000","updated_by":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"reviewer":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"state":"REVIEWER"}],"messages":[{"id":"70e9795b8a4bbc48c03ad6c69126b9fdccea0b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-18 20:08:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f97ad8ea1feab9ce29b9dd9b94f1c7d15c0f479d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 20:31:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d5394bd8fea4f3ca6e5632a38a54f18\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6645928e68c745718a4d7208d5e66050 : RETRY_LIMIT in 2m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c33da32c6d274528804c10f910ae9553 : RETRY_LIMIT in 3m 33s\n- loci-heat https://zuul.opendev.org/t/openstack/build/32c31b12bb594f8998b9fedbbde5cd48 : RETRY_LIMIT in 2m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b4ce1adfd589418e955984fcdb1f8cc5 : RETRY_LIMIT in 2m 39s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a883164222814605a5260bed6846b838 : RETRY_LIMIT in 2m 11s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/066e9b49dd7a43ca87ba564041452d64 : RETRY_LIMIT in 2m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/261823ead82344d8b5d5e861a9907989 : RETRY_LIMIT in 2m 30s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/c34c1a751bbd4176827f80e26856efee : RETRY_LIMIT in 2m 20s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a5101286a4fc497a89be92112a41d455 : RETRY_LIMIT in 2m 10s\n- loci-nova https://zuul.opendev.org/t/openstack/build/217aa955d2344767bb8b77e3e1a41de9 : RETRY_LIMIT in 2m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4017c531a8164870839e70dfb499f32d : RETRY_LIMIT in 2m 25s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ba7ac285e6dc494bababd1a62ce9a1ce : RETRY_LIMIT in 2m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a536781867c547e99f9304f6f06b1dd6 : RETRY_LIMIT in 2m 10s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6bdd30bc5fe74fd18e3ae7dfd146e523 : RETRY_LIMIT in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bce955b75280c416b159331ba626916dc6f1e06c","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-18 20:42:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c026096d810b615bc2a0dd0482c73252678c537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 21:23:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/040dd919dda44c38adc9e0e3f02e9014\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ec66ab24d7774ea891b5e9595497f0b5 : RETRY_LIMIT in 2m 39s\n- loci-glance https://zuul.opendev.org/t/openstack/build/14ca5ccaf4034eb2bde836027fe7f81f : RETRY_LIMIT in 2m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9adf03e498964cde8ed1794aafd53cd6 : RETRY_LIMIT in 3m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/913581450d424382b530319e99b0d512 : RETRY_LIMIT in 2m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/86f7554dc1a8482db93a7bf8726f8daa : RETRY_LIMIT in 3m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/757d485622884d268185d8f902f4e8ba : RETRY_LIMIT in 2m 23s\n- loci-manila https://zuul.opendev.org/t/openstack/build/783ac9271f5d4490b698f66e20aeba33 : RETRY_LIMIT in 2m 27s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/13a62073aaa14c2e8df20a89b59df610 : RETRY_LIMIT in 2m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/836e4b5bc9af404aa8170b6e39c418b5 : RETRY_LIMIT in 2m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2a16e60bc4c14826b11b342b3f726d81 : RETRY_LIMIT in 2m 37s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8e3b4aec652d498494ccd12964beda0f : RETRY_LIMIT in 2m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/16633582392344388448390723822dc5 : RETRY_LIMIT in 2m 30s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/afa45a037c0a499bb3413d4b4aa7cf69 : RETRY_LIMIT in 2m 26s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/369ecc7c7f4f4ce8b93302be480608f9 : RETRY_LIMIT in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"397c1f66d5b5aac8c20dd8db02fe200ae52fcf15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-18 21:42:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbaa1101e25b7baa4cd8d2aba7f6f2e24679ffd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-18 21:51:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"324afbb1fc574e551932395faeb7bbed16e77a32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 22:00:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04ad4166a0334152a3c13840b6efe353\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/51b2eab7c5ed47ebb3324771e5430346 : FAILURE in 2m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ba5fe166aec24112bcdb5f3572528c74 : FAILURE in 4m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/59129583c1cf44d586e383337a4b80d0 : FAILURE in 3m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3f00c5971a3d47509f36d1dbfd674f1c : FAILURE in 3m 59s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/407c64de7561419b83f9028ee3ec1f1f : FAILURE in 4m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fd161f907c5e49b9973f8144fc66cb18 : FAILURE in 2m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8a2c958ca8b04cc4bc370e201e96d5b1 : FAILURE in 2m 56s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/d784d1cce0f24bb5985002eaae377962 : FAILURE in 2m 54s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/bbfcfe6e5c4f42828f2c51d65c3880a3 : FAILURE in 3m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2021478019ea428aa933cd5dabec9dfd : FAILURE in 2m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ce855a52ef0541cbabfcb1ec4a282d10 : FAILURE in 2m 46s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e8ccb77b1375403096ba5730cb309c69 : FAILURE in 2m 37s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/72e023901a124c0c93a281695d9715af : FAILURE in 3m 34s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b1531bf2b5b24fd49ef437e7cbb703c8 : FAILURE in 2m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f4d33b6a9ed4407510b0a78b87e71c5eb7a10e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-18 22:03:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"43ac44a1fb31016b969058df8e56bbf8333f82c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-18 22:12:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e8859aa843b3f163b6486c7c286b8f5c69ee28e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 22:49:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccea890844c74859891bc68f014c8b7c\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/37f3b52a06ba41b7bcd3e70b4a92f97e : SUCCESS in 20m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0d7b2d4319ed40b0af2392d8e4f9a788 : SUCCESS in 19m 43s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a78e743cecaa4b63b8ee2fbf53122500 : SUCCESS in 20m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/253773f822194642a29d02f4e8f0c7eb : SUCCESS in 19m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/40a95fc47d7940fc92f162a1737e9afa : SUCCESS in 19m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1bd58ec716024f669bc43e0f637904a8 : SUCCESS in 18m 38s\n- loci-manila https://zuul.opendev.org/t/openstack/build/19f4d954d82f4c20a4d980f40e25eee2 : SUCCESS in 19m 11s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4df3008a98504ea7ae6cb098930ab183 : SUCCESS in 18m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3ff7382a38464d739fe1657bb7881c71 : SUCCESS in 24m 26s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bf2d612616d047978b51369a04b37174 : SUCCESS in 22m 27s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/65c0b6b0d7cd45fb82b5bb9d707200de : SUCCESS in 20m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/26208b9134c74f608e7914e7daeea2ea : SUCCESS in 18m 48s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a1548ed8851644708b5799e1dbc37201 : SUCCESS in 14m 24s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fc9bb7e637cd487f805df8c9e229d416 : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":5},{"id":"eef09b4af64a721e7bde98e390335de023e603db","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-06-19 17:49:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e07bc723c5485f7ba53a060fbec6fd194e51ba3c","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2024-06-19 21:05:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"74b243046aa99bcad9eb7966309e6645f36655c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 21:05:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9aeeb34143879cfe1564182a02ca06b3f2475636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 21:22:29.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66b27be329894706a592f3b0481a4554\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d8613500a6e6463e941c01f76c94d35d : FAILURE in 10m 32s\n- loci-glance https://zuul.opendev.org/t/openstack/build/30f1a269de6d4c58bc2f8b7c881b7891 : FAILURE in 10m 25s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b9015bed9241423c8e75acd55058fc7e : FAILURE in 10m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b7c5d5dc258a4a94888a80db6196ad05 : FAILURE in 14m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e5488c8b1fd04ebcaba9fb19a454144c : FAILURE in 13m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/54710d8e490843a096cdf38b1bff0436 : FAILURE in 14m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/17a46fcd138b43a3b41c629b122b752f : FAILURE in 8m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f41bf039069a4eb389b7ee899a4f7aa9 : FAILURE in 8m 36s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6a8314ed52734fc293e1458c28f6d51a : FAILURE in 10m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9d2b25d125b4467a9a3743aa028daa3c : FAILURE in 9m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7a5f3733369c41b49bdf21e091f3777f : FAILURE in 11m 22s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b93b57094ea64faca910928cd2978145 : FAILURE in 10m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/64d92959c6b3467fb5d4841deda99160 : FAILURE in 10m 39s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/ea276b8555b4464babbd44f9c60be89e : FAILURE in 9m 49s","accounts_in_message":[],"_revision_number":5},{"id":"25ea50fdcb6d9b38393420c22e74c85f27b08610","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-20 08:11:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f6b82d1a9c87f310125011a15f224b1d14b9b3df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 08:36:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f94d0b99c1140f6b0c1ee92ac063a85\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dfc3f3b1ab134f159ed1e97b93daad90 : FAILURE in 9m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0e5e7b785b3b409c8b40d697ba0c923e : FAILURE in 9m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a1ce37cbfa3a4911a62ad813b000d079 : FAILURE in 8m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/947d15595e3c4d28a84e3273824ce1ae : FAILURE in 9m 41s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ec0cd91941d140098b26f05e589b5850 : FAILURE in 9m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bac5b2ebc825409e9cb972056c10789a : FAILURE in 10m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7b0505f40d0d412697f0c99954ea59db : FAILURE in 10m 02s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2834495e6d23482ea8dc6b3c4cf6d4af : FAILURE in 9m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/efe9431b6d734db2a92b484a37f03862 : FAILURE in 9m 35s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f8ac276bb7f3412199f802c0b1c3cd1c : FAILURE in 10m 08s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7df1fe12099c42ba85da70015684df06 : FAILURE in 10m 15s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7513faf4e9a84ccba25c9a40e2932fcb : FAILURE in 10m 06s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/81ddac1c95bf4be3a3d1213248752b8e : FAILURE in 9m 10s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5655af36d57e40208ee69610a666d2bb : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":5},{"id":"5cef45c3946e4c05d9989a35a28fa87e06bb935b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-20 10:35:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"6301e82e713f37fe9f471659e12535d317defc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 11:13:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cecfaaf3e0a4e889ddc4932f29eb81a\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2a1d62d5c7fc43389c879e11e76f2dfb : SUCCESS in 19m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9bcaf8e6da5e43d0ae44f4d5eee5f705 : SUCCESS in 20m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c1f7971d5e564f35b7098acf20b5222c : SUCCESS in 22m 30s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6957da901b4b4543be4249e3de6241c4 : SUCCESS in 22m 28s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d2870fb00d7a4d7699de6b30adf6e447 : SUCCESS in 19m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c9da9cbb1fd34643844173f10938c46b : SUCCESS in 19m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/961650dc1b814e65b829e80e07a5d855 : SUCCESS in 17m 31s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/52b2470d2275475d90c42e6468782f3a : SUCCESS in 19m 17s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/986a52bb99854c3c80808fe4dd9f0016 : SUCCESS in 19m 45s\n- loci-nova https://zuul.opendev.org/t/openstack/build/75ccf002f1f54646a175be2fafb22863 : SUCCESS in 25m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/76afd3e7a17e407f8b8eb42d7767ea66 : SUCCESS in 19m 34s\n- loci-placement https://zuul.opendev.org/t/openstack/build/74b23e08a8cc4f86ba3ee76ad096452c : SUCCESS in 17m 47s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/cadb541b1dcd4752b5456a8404534b34 : SUCCESS in 13m 36s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/e7a737caa4ff4e7f8300182eee269093 : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":6},{"id":"490bdaaf4571a08e318e3964c71703eee16e7d62","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-06-20 14:33:34.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"fccb502d9a2f4c31064cb44894599489ce358f01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 14:34:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"15044b0e8a8398054d6236b0739a9e2cc9f70831","author":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"date":"2024-06-20 14:38:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1efc99d177b0768592a208fe33320d1923e6317d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 15:03:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818e4b1603b14923b8376db04a2d0dfc\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ff168bd5c8cf4b129068bb370c05b42c : SUCCESS in 16m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/55cb8c31867144088a26eebdaa868b8d : SUCCESS in 17m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/74f6b893a50c48be9c6e7434b1511c71 : SUCCESS in 15m 46s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b3f901253dbc4142bb42376cd141df3d : SUCCESS in 17m 54s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d1d8da05e0fc4c98b8987b320f0fc7f5 : SUCCESS in 14m 33s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/04360fd7921d4100921028ab37b258e6 : SUCCESS in 15m 08s\n- loci-manila https://zuul.opendev.org/t/openstack/build/90f1aaf27cc54c31b75aef4e9bbca08c : SUCCESS in 20m 30s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/0f355c6acf3a46deab4edb3234bfbc5a : SUCCESS in 17m 30s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4abe3853c623416796917d0794d8670b : SUCCESS in 18m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8507b53ac4e5443bacb981473e1a1459 : SUCCESS in 21m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fb9d861137a3401fb63cf49553c4c9d1 : SUCCESS in 23m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c74768bff70043a2b5d2f20810451f87 : SUCCESS in 17m 11s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/722bbd155b274c8baa578a7c976ddacf : SUCCESS in 12m 42s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/15f88ff469e847baaa6a2723a697ed4f : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":6},{"id":"c2db58b411ceaa8b63957254f6102e856712ce93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 15:03:57.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4226e755ee700e384ab2c6a76cd7b24fa4cf49fc","revisions":{"40bd6bb665c4e2ccda271210a2fe38ea8085e71c":{"kind":"REWORK","_number":1,"created":"2024-06-18 20:08:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/922253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/922253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/922253/1"}}},"commit":{"parents":[{"commit":"0842d310b33a51d2b1d90930c62b8ac451867d58","subject":"Merge \"Fix building python-nss\u003d\u003d1.0.1 for Focal images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0842d310b33a51d2b1d90930c62b8ac451867d58"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"subject":"Add networking-baremetal to pydep.txt","message":"Add networking-baremetal to pydep.txt\n\nNetworking-baremetal package provides ironic-neutron-agent\nand is not among the requirements of neither Neutron itself\nnor global upper-constraints. So when we try to install it\nusing PIP_PACKAGES build env variable pip can not find it\namong pre-downloaded wheels. So let\u0027s add it to pydep.txt\nand when necessary to install it together with neutron\nwe can pass the baremetal profile.\n\nChange-Id: I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/40bd6bb665c4e2ccda271210a2fe38ea8085e71c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/40bd6bb665c4e2ccda271210a2fe38ea8085e71c"}]},"branch":"refs/heads/master"},"90edfa102c47d46340e1c0b5d0b80a0ee398d7b8":{"kind":"REWORK","_number":2,"created":"2024-06-18 21:42:11.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/922253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/922253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/922253/2"}}},"commit":{"parents":[{"commit":"0842d310b33a51d2b1d90930c62b8ac451867d58","subject":"Merge \"Fix building python-nss\u003d\u003d1.0.1 for Focal images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0842d310b33a51d2b1d90930c62b8ac451867d58"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 21:42:01.000000000","tz":-300},"subject":"Add networking-baremetal to pydep.txt","message":"Add networking-baremetal to pydep.txt\n\nNetworking-baremetal package provides ironic-neutron-agent\nand is not among the requirements of neither Neutron itself\nnor global upper-constraints. So when we try to install it\nusing PIP_PACKAGES build env variable pip can not find it\namong pre-downloaded wheels. So let\u0027s add it to pydep.txt\nand when necessary to install it together with neutron\nwe can pass the baremetal profile.\n\nChange-Id: I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/90edfa102c47d46340e1c0b5d0b80a0ee398d7b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/90edfa102c47d46340e1c0b5d0b80a0ee398d7b8"}]},"branch":"refs/heads/master"},"67d587780b4a6dfd89d1fe5e8c86884b24173687":{"kind":"REWORK","_number":3,"created":"2024-06-18 21:51:47.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/922253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/922253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/922253/3"}}},"commit":{"parents":[{"commit":"0842d310b33a51d2b1d90930c62b8ac451867d58","subject":"Merge \"Fix building python-nss\u003d\u003d1.0.1 for Focal images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0842d310b33a51d2b1d90930c62b8ac451867d58"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 21:51:37.000000000","tz":-300},"subject":"Add networking-baremetal to pydep.txt","message":"Add networking-baremetal to pydep.txt\n\nNetworking-baremetal package provides ironic-neutron-agent\nand is not among the requirements of neither Neutron itself\nnor global upper-constraints. So when we try to install it\nusing PIP_PACKAGES build env variable pip can not find it\namong pre-downloaded wheels. So let\u0027s add it to pydep.txt\nand when necessary to install it together with neutron\nwe can pass the baremetal profile.\n\nChange-Id: I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67d587780b4a6dfd89d1fe5e8c86884b24173687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67d587780b4a6dfd89d1fe5e8c86884b24173687"}]},"branch":"refs/heads/master"},"550ecc8ec5d9a7088bbfbf0e82e7964263d1d324":{"kind":"REWORK","_number":4,"created":"2024-06-18 22:03:43.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/922253/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/922253/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/922253/4"}}},"commit":{"parents":[{"commit":"0842d310b33a51d2b1d90930c62b8ac451867d58","subject":"Merge \"Fix building python-nss\u003d\u003d1.0.1 for Focal images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0842d310b33a51d2b1d90930c62b8ac451867d58"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 22:03:34.000000000","tz":-300},"subject":"Add networking-baremetal to pydep.txt","message":"Add networking-baremetal to pydep.txt\n\nNetworking-baremetal package provides ironic-neutron-agent\nand is not among the requirements of neither Neutron itself\nnor global upper-constraints. So when we try to install it\nusing PIP_PACKAGES build env variable pip can not find it\namong pre-downloaded wheels. So let\u0027s add it to pydep.txt\nand when necessary to install it together with neutron\nwe can pass the baremetal profile.\n\nChange-Id: I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/550ecc8ec5d9a7088bbfbf0e82e7964263d1d324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/550ecc8ec5d9a7088bbfbf0e82e7964263d1d324"}]},"branch":"refs/heads/master"},"279b8bbffa049002c8779701ede404de5762b063":{"kind":"REWORK","_number":5,"created":"2024-06-18 22:12:34.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/922253/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/922253/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/922253/5"}}},"commit":{"parents":[{"commit":"0842d310b33a51d2b1d90930c62b8ac451867d58","subject":"Merge \"Fix building python-nss\u003d\u003d1.0.1 for Focal images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0842d310b33a51d2b1d90930c62b8ac451867d58"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 22:12:25.000000000","tz":-300},"subject":"Add networking-baremetal to pydep.txt","message":"Add networking-baremetal to pydep.txt\n\nNetworking-baremetal package provides ironic-neutron-agent\nand is not among the requirements of neither Neutron itself\nnor global upper-constraints. So when we try to install it\nusing PIP_PACKAGES build env variable pip can not find it\namong pre-downloaded wheels. So let\u0027s add it to pydep.txt\nand when necessary to install it together with neutron\nwe can pass the baremetal profile.\n\nChange-Id: I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/279b8bbffa049002c8779701ede404de5762b063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/279b8bbffa049002c8779701ede404de5762b063"}]},"branch":"refs/heads/master"},"4226e755ee700e384ab2c6a76cd7b24fa4cf49fc":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-06-20 10:35:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/922253/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/922253/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/922253/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/922253/6"}}},"commit":{"parents":[{"commit":"73d2ed26577c3bf3ec658d578895033569ae471c","subject":"Fix requirements build for master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/73d2ed26577c3bf3ec658d578895033569ae471c"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-18 20:01:09.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-06-20 10:35:59.000000000","tz":0},"subject":"Add networking-baremetal to pydep.txt","message":"Add networking-baremetal to pydep.txt\n\nNetworking-baremetal package provides ironic-neutron-agent\nand is not among the requirements of neither Neutron itself\nnor global upper-constraints. So when we try to install it\nusing PIP_PACKAGES build env variable pip can not find it\namong pre-downloaded wheels. So let\u0027s add it to pydep.txt\nand when necessary to install it together with neutron\nwe can pass the baremetal profile.\n\nChange-Id: I0f3eed5e1babd7fafcfbe2bae18a8b6b3b93397b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4226e755ee700e384ab2c6a76cd7b24fa4cf49fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4226e755ee700e384ab2c6a76cd7b24fa4cf49fc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
