)]}'
{"id":"openstack%2Floci~931896","triplet_id":"openstack%2Floci~master~Iadf9a14aa0aa9082abb4ba015b08d2e37a20f521","project":"openstack/loci","branch":"master","topic":"upstream","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-10-10 23:01:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadf9a14aa0aa9082abb4ba015b08d2e37a20f521","subject":"Allow to install extra projects from source","status":"MERGED","created":"2024-10-09 10:29:24.000000000","updated":"2024-10-10 23:04:47.000000000","submitted":"2024-10-10 23:01:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931896-upstream","meta_rev_id":"99af2d7dc44644c1a8aa0568cbaaa7f1e8e31503","_number":931896,"virtual_id_number":931896,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 23:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-10-10 15:58:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-10-10 21:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-10-10 21:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 11:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 15:58:01.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-10-10 21:48:34.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"672ef2a6c7ed2c268cd562f3393811477666c33e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-09 10:29:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b31e1dce3370b4d81c18be9d953de95225256c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 11:23:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d53d995f03840d4916907221b02de66\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/00853be2b71e41eab1e2f867ba1cedd3 : SUCCESS in 46m 29s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/4aa7b9406a4b4d3eab9d1ef71b457d2e : SUCCESS in 41m 45s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9a9574af78a84ca48f68ec3c3a076849 : SUCCESS in 40m 25s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8f9063438314454cb81a11c68ff8f520 : SUCCESS in 40m 44s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6460701cfd854f978a97e3a3bbc0893a : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bbd3933118f5078119d1cdb414555a688637fe1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-09 18:43:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fadfff0d346a5d404ac81df4fd7c1f555d620ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 19:32:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c743f6a237d49e3a881f135b840c8d2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6b4728604ef44a89b6aa58decb412826 : SUCCESS in 47m 37s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/879937cb878c4acdaf40ed90f3c0606a : SUCCESS in 40m 31s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/70a419a7fef346aa9ff00cde497c0e9f : SUCCESS in 40m 04s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fee1f8f4bb0449b290b8bcd0b76ea51d : SUCCESS in 40m 42s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f68cfd00573e45e48e37f7c7f2d39309 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":2},{"id":"231bea3eecbfc599e432164e5be6dc9bc0937b8b","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-10-10 15:58:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"715fe00c44fe581c58fc268e001caa231dc7df36","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-10-10 21:48:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f1abdfd6b606f8aa9cec50a3654c6ecb1613adce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 21:48:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f7778a4964744cdb143a30027e28eccbab06b461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 23:01:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce6c1b96faa4f128277c827e47cb510\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/070e64a5659e40658a10f7078e1771d0 : SUCCESS in 1h 13m 00s (non-voting)\n- loci-upload-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/c345d03db3434bce8a6ca2bcc8131109 : SUCCESS in 1h 09m 31s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/48cca64606df42e8ac248de9ececac2e : SUCCESS in 1h 08m 12s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d90b82e8b1dc4670b2648dc28f61535d : SUCCESS in 1h 04m 13s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3cbf01a47b324814a27b47a3657f50b0 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e7a2dba597dc56363a6ee0ce8ea0a151427112e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 23:01:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"99af2d7dc44644c1a8aa0568cbaaa7f1e8e31503","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 23:04:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/745f43575fe8410e9c3c25c0c5d7d063\n\n- loci-promote-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/f4ad5801b6f2413582e953864733af8a : SUCCESS in 2m 33s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d04edfc2a5f749dbb632f3815b48b966 : SUCCESS in 2m 21s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/68f6e4bd89ed459593c193cab02d52ec : SUCCESS in 2m 21s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8d2222d9e5304d98a840a2a5d1194fe4 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"977cd4ba10371cc237822b36394c19405d70945f","revisions":{"715653b1d5f7a8c8d7f5308ecae292f13f0729b1":{"kind":"REWORK","_number":1,"created":"2024-10-09 10:29:24.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/96/931896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/931896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/931896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/931896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/931896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/931896/1"}}},"commit":{"parents":[{"commit":"b95d8e31ecf7dded0637ff5a8e78c72b5cbd8495","subject":"Allow to pass project from external directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b95d8e31ecf7dded0637ff5a8e78c72b5cbd8495"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:28:49.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:28:49.000000000","tz":0},"subject":"Allow to install extra projects from source","message":"Allow to install extra projects from source\n\nIt may be required to install additional python packages from source.\nTo do this download them into loci/data directory and specify names of\nprojects via EXTRA_PROJECTS argument during build.\n\nChange-Id: Iadf9a14aa0aa9082abb4ba015b08d2e37a20f521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/715653b1d5f7a8c8d7f5308ecae292f13f0729b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/715653b1d5f7a8c8d7f5308ecae292f13f0729b1"}]},"branch":"refs/heads/master"},"977cd4ba10371cc237822b36394c19405d70945f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-09 18:43:39.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/96/931896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/931896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/931896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/931896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/931896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/931896/2"}}},"commit":{"parents":[{"commit":"447845c0f79425d500f2b6ad1a660dec42681f9f","subject":"Allow to pass project from external directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/447845c0f79425d500f2b6ad1a660dec42681f9f"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:28:49.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 18:43:39.000000000","tz":0},"subject":"Allow to install extra projects from source","message":"Allow to install extra projects from source\n\nIt may be required to install additional python packages from source.\nTo do this download them into loci/data directory and specify names of\nprojects via EXTRA_PROJECTS argument during build.\n\nChange-Id: Iadf9a14aa0aa9082abb4ba015b08d2e37a20f521\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/977cd4ba10371cc237822b36394c19405d70945f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/977cd4ba10371cc237822b36394c19405d70945f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
