)]}'
{"id":"openstack%2Floci~931901","triplet_id":"openstack%2Floci~master~Ic574ff530d23cc04301949a0bcdb637737055da4","project":"openstack/loci","branch":"master","topic":"upstream","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-12-03 19:36:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic574ff530d23cc04301949a0bcdb637737055da4","subject":"Allow to specify python extras during project install","status":"MERGED","created":"2024-10-09 10:47:52.000000000","updated":"2024-12-03 19:39:52.000000000","submitted":"2024-12-03 19:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931901-upstream","meta_rev_id":"93867f8ee89a9002ce8c23c1b5e55949313d67ca","_number":931901,"virtual_id_number":931901,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-03 19:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-12-02 16:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-12-03 18:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-12-03 18:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 11:47:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 16:37:26.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-12-03 18:01:48.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"96d6889748078b52c8367d1b067bd5b5e62f4588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-09 10:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1847f1928a53f1c6d16271a00c109d19775670e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 11:47:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28bf73048d3f46aa96f89bf21d196daf\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b1547e97340f4353a58f22f59783e3fd : SUCCESS in 55m 26s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/422f2a50019a450f9dea98cff3bdf0ec : SUCCESS in 41m 50s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7068d01bfae64908afb7368508c8a60f : SUCCESS in 42m 38s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8d3bc4efe5ff43e28e9634fe12dfa2d9 : SUCCESS in 42m 18s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cec08a7e99964fb0b126458b5c1d36cf : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":1},{"id":"19f281a4cae03561409621437bab17b66e9e467c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-09 18:43:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c6e81095a7979ff9791a0a041ad652038f64c517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 19:34:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ba591a16544a99a3872a892d5f994d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1f68d4fb559a452aa1ae6249475e8199 : SUCCESS in 48m 56s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/bd832f1fdd9b4b06932430b02854720e : SUCCESS in 41m 22s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5bb20032cb8240e092cec2078946fc53 : SUCCESS in 40m 22s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1dbd90529c3c4090bce72fc3358396fd : SUCCESS in 39m 55s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/961f1a8b6bc14dc19a363c40fdd9f115 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7105093302aabfc31fe529d8a08346d351c6ff8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-28 14:58:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"93853471fd5d53d12a69f09cf40005639fb47c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 15:43:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79803436a3764ea792c3a7ed8ae39101\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d54556a4e8794c44b7ff98052ec5a04a : SUCCESS in 38m 52s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9a8852c8c4714c66b6ba7d0a3267b73d : FAILURE in 4m 34s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/df8875a963894f6580d48b1ed685e3ed : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e1e1409877cf4a9fbee9de77375bc3b4 : RETRY_LIMIT in 3m 50s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3d319a2134a4485ba2e364dc277b2ad1 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89b7a0302f56412493071ba215efd6ae : FAILURE in 4m 14s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/046ba71ad0be427a84a61b069d7193fc : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/35136f0f5ed441b1a7bb8c6191d241e7 : RETRY_LIMIT in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"75565d1e2576e8976e73938f7fe33d5b4bfe2804","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-28 19:01:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea3da4cf58ca12ca5ec442589c2c1dee45982401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 21:33:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c8626dc19ea4d6a92879e0cfc375646\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c14c68f050624814beaa03a685824eec : SUCCESS in 2h 31m 37s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89d4acfce3ab46e78c6a09337b54814d : SUCCESS in 1h 03m 01s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/465d7887cd744588aca6f139edfbdfda : SUCCESS in 1h 08m 37s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/44f7ee15c0ff446791f5626cb02b2deb : SUCCESS in 59m 40s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/96bfdd04bd1846fb85911ee83464a773 : SUCCESS in 1h 06m 24s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90277aa12ddc434a8fc693fd38898ff0 : SUCCESS in 57m 59s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/86f3c310822241ad905eadd330674bb0 : SUCCESS in 1h 11m 32s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b1bee67dc96549639a0ead50daebc916 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e0ef66c496cc57ddcba8a04e5558d105ae020cd5","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-12-02 16:37:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bf5db25ddb35f741500b60286018d41f30c22b57","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-12-03 18:01:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e6d1be010215c44a36aaf862c6dbceb5d1964cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 18:02:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3e6305c951ff59b6af5b73e785235f018a5c1bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 19:36:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c00c9495ab4e54a9b9bba84c75f2a2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0a02519aeedd4e0b8fe05bccfce05d48 : SUCCESS in 1h 28m 30s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c22c15fb49d448af966c5842d1030f24 : SUCCESS in 1h 09m 14s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/77e3e29d243a43ba9c1263358bc89851 : SUCCESS in 1h 05m 35s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b9584ecc50b44cc2bf87bec482094bfd : SUCCESS in 1h 13m 49s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e472757bee29414f9d0ce6772aced5ab : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":3},{"id":"03c76145e02b570b180c02210e547c6c513a3715","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 19:36:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"93867f8ee89a9002ce8c23c1b5e55949313d67ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 19:39:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17f592af32334c329f5c3ef078dc021c\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b0435e9456554f018ed8b7db570ff9d7 : SUCCESS in 3m 32s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eec10ed1f34a46d59f9d5ef4e177eea7 : SUCCESS in 2m 44s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/133586fc681d4f9caa2bfbb258aa45f7 : SUCCESS in 3m 24s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f5d4c5831cfa4a8480ed80e9317e721c : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0a81087440cd04f7ddff05110684ebe108cad49","revisions":{"51a120b7e910c5372e1298e1d06032dbe12c2d2b":{"kind":"REWORK","_number":1,"created":"2024-10-09 10:47:52.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/01/931901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/01/931901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/01/931901/1"}}},"commit":{"parents":[{"commit":"420a8425c5667702fcbbca1797546061fba0d080","subject":"Build horizon static during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/420a8425c5667702fcbbca1797546061fba0d080"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:47:37.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:47:37.000000000","tz":0},"subject":"Allow to specify python extras during project install","message":"Allow to specify python extras during project install\n\nAdd PROJECT_PIP_EXTRAS that allows to specify which python extras\nto use during project install.\n\nChange-Id: Ic574ff530d23cc04301949a0bcdb637737055da4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/51a120b7e910c5372e1298e1d06032dbe12c2d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/51a120b7e910c5372e1298e1d06032dbe12c2d2b"}]},"branch":"refs/heads/master"},"5d2b2b4f89f412b9e45ccdfe6eff54b5f36ef7c5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-09 18:43:47.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/01/931901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/01/931901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/01/931901/2"}}},"commit":{"parents":[{"commit":"dc8797d448e1bc725fc2eafaa5d2e1a4ffd31a2e","subject":"Build horizon static during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dc8797d448e1bc725fc2eafaa5d2e1a4ffd31a2e"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:47:37.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 18:43:47.000000000","tz":0},"subject":"Allow to specify python extras during project install","message":"Allow to specify python extras during project install\n\nAdd PROJECT_PIP_EXTRAS that allows to specify which python extras\nto use during project install.\n\nChange-Id: Ic574ff530d23cc04301949a0bcdb637737055da4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d2b2b4f89f412b9e45ccdfe6eff54b5f36ef7c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d2b2b4f89f412b9e45ccdfe6eff54b5f36ef7c5"}]},"branch":"refs/heads/master","description":"Rebase"},"d0a81087440cd04f7ddff05110684ebe108cad49":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-28 14:58:04.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/01/931901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/01/931901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/01/931901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/01/931901/3"}}},"commit":{"parents":[{"commit":"185487e8c8d61de4568e83d55d59ee38023d2eef","subject":"Build horizon static during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/185487e8c8d61de4568e83d55d59ee38023d2eef"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-09 10:47:37.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-28 14:57:37.000000000","tz":0},"subject":"Allow to specify python extras during project install","message":"Allow to specify python extras during project install\n\nAdd PROJECT_PIP_EXTRAS that allows to specify which python extras\nto use during project install.\n\nChange-Id: Ic574ff530d23cc04301949a0bcdb637737055da4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d0a81087440cd04f7ddff05110684ebe108cad49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d0a81087440cd04f7ddff05110684ebe108cad49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
