)]}'
{"id":"openstack%2Floci~932167","triplet_id":"openstack%2Floci~master~If1b558c1fe85137716c0836b7049274affeef2c4","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"8898":{"account":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"last_update":"2024-10-28 19:49:33.000000000","reason":"Change was submitted"},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-10-28 19:49:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1b558c1fe85137716c0836b7049274affeef2c4","subject":"Set default locale","status":"MERGED","created":"2024-10-11 08:21:32.000000000","updated":"2024-10-28 19:52:55.000000000","submitted":"2024-10-28 19:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932167","meta_rev_id":"4f6b0ec44108e4c2d8b86644fb4a2e69db9a492e","_number":932167,"virtual_id_number":932167,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-28 19:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-10-28 18:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-10-23 06:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2024-10-28 18:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16771,"name":"mpolenchuk","email":"mpolenchuk@mirantis.com","username":"mpolenchuk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-11 08:21:32.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":16771,"name":"mpolenchuk","email":"mpolenchuk@mirantis.com","username":"mpolenchuk"},"state":"CC"},{"updated":"2024-10-11 09:32:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-11 23:59:18.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-10-23 05:16:21.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"0090723dad0e27e045d543a7be1f7f880a3f7b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-11 08:21:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52ebff48a466706b3bceff1c175f5359bede4e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 09:32:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89675114894c41e18e6d87cdc0fd424a\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9f3ae193d4c642148d7293e0b16f1f14 : SUCCESS in 1h 07m 21s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/23623c9fb95247e49da42052d4794e5f : SUCCESS in 1h 00m 58s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4c744e51d825455f819a4273e9cd994f : SUCCESS in 1h 00m 22s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e9d59405d72d47ae9222e50afeb3d91b : SUCCESS in 58m 05s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/634696bcfecf4d94a6434c3ce3854ac9 : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":1},{"id":"688e462963f44e56e865310b7aa89ebab6f7d4f8","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-10-11 23:59:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0cca881c82e00a43e5809653a2a2d88eb62e3709","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2024-10-23 05:16:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c1df4aeabf856fced324bae0b92679cf9fcb244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-10-23 06:23:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ffc7c01f3d40ee674ac7cb4825a25f551c9a51e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 07:14:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ead394d3fbd48a79e387c723c184487\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/fcfcc600d5e9471fb9255086343a1447 : SUCCESS in 50m 06s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/027c609ba0e4424bb646ad68dfe808a0 : SUCCESS in 47m 03s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9d5c9253090f4dfe9104ba2751d797cc : SUCCESS in 44m 32s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/770e66b60a914bafb0666d98b8efcc14 : SUCCESS in 45m 51s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/36a711db8f364e7aaa2371171c6bccf3 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":2},{"id":"7294159471c4838aca094eee03d7985f9ff80233","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-10-28 18:13:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99b7b80b4d4fe92f3e7aff8db27eafbbadb7cfc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 18:13:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a1c8223048852a3994bbf816c2f81ccb7cdca69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:49:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b679d1d476649f989043c601b5bb7af\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/51d75562259040029e57fe950efa5881 : SUCCESS in 1h 30m 45s (non-voting)\n- loci-upload-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/b61d1e41903442ab9d6b341d02d8eefe : SUCCESS in 1h 22m 20s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f07d2c96d17a4acfad7dea552c8606c1 : SUCCESS in 1h 09m 29s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/083feaf59d2641418795ca763f4f1a21 : SUCCESS in 1h 13m 41s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1978fa3833bc4ee48173e574b7d57c2e : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"52ad2c43c8773fea8bc78823e13222bd320190cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:49:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f6b0ec44108e4c2d8b86644fb4a2e69db9a492e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 19:52:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59e5dd2aa23a4488895e169a92cc7f37\n\n- loci-promote-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/0ce87628576745e38aad08780b79d88b : SUCCESS in 3m 05s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5a29ea4df0644163bbdd14faa2e22997 : SUCCESS in 2m 36s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e8ae2f11b86945f69a8711b5b64dbdbb : SUCCESS in 2m 18s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/99aa3d2540484979a7656bc8e61181ee : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","revisions":{"eef1dd1390c67b6ac9b62122106071ebf072c0ad":{"kind":"REWORK","_number":1,"created":"2024-10-11 08:21:32.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/932167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/932167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/932167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/932167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/932167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/932167/1"}}},"commit":{"parents":[{"commit":"5d2b2b4f89f412b9e45ccdfe6eff54b5f36ef7c5","subject":"Allow to specify python extras during project install","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5d2b2b4f89f412b9e45ccdfe6eff54b5f36ef7c5"}]}],"author":{"name":"Michael Polenchuk","email":"mpolenchuk@mirantis.com","date":"2020-06-22 12:43:26.000000000","tz":240},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-11 08:21:16.000000000","tz":0},"subject":"Set default locale","message":"Set default locale\n\nNeutron tempest plugin (common/ssh.py) gets locale for\nexecute scripts on remote machines, so set it to default\nvalue to avoid an empty one.\n\nCloses-Bug: PRODX-5583\nChange-Id: If1b558c1fe85137716c0836b7049274affeef2c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eef1dd1390c67b6ac9b62122106071ebf072c0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eef1dd1390c67b6ac9b62122106071ebf072c0ad"}]},"branch":"refs/heads/master"},"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-23 06:23:55.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/932167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/932167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/932167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/932167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/932167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/932167/2"}}},"commit":{"parents":[{"commit":"4251c1c4e88677edecdc845a5c99bf078db1a88f","subject":"Temporary customization for 2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4251c1c4e88677edecdc845a5c99bf078db1a88f"}]}],"author":{"name":"Michael Polenchuk","email":"mpolenchuk@mirantis.com","date":"2020-06-22 12:43:26.000000000","tz":240},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-10-23 06:23:55.000000000","tz":0},"subject":"Set default locale","message":"Set default locale\n\nNeutron tempest plugin (common/ssh.py) gets locale for\nexecute scripts on remote machines, so set it to default\nvalue to avoid an empty one.\n\nCloses-Bug: PRODX-5583\nChange-Id: If1b558c1fe85137716c0836b7049274affeef2c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
