)]}'
{"id":"openstack%2Floci~934263","triplet_id":"openstack%2Floci~master~If391d17f14f054afca061036487574805c393634","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2024-11-15 18:24:42.000000000","reason":"Change was submitted"},"34520":{"account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"last_update":"2024-11-15 15:42:56.000000000","reason":"\u003cGERRIT_ACCOUNT_34520\u003e replied on the change","reason_account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}},"hashtags":[],"change_id":"If391d17f14f054afca061036487574805c393634","subject":"Test images before publishing them","status":"MERGED","created":"2024-11-06 18:39:31.000000000","updated":"2024-11-15 18:34:17.000000000","submitted":"2024-11-15 18:24:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934263","meta_rev_id":"4d1d3d3eb197c951ef223c71e402995a99b07321","_number":934263,"virtual_id_number":934263,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 18:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-11-15 16:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2024-11-15 15:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-11-15 16:50:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 18:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-06 22:19:58.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2024-11-15 16:50:35.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"34573cd08c33ab5ec21001c65f1bc887b46332c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-06 18:39:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"079a484b2f70f407803210503cd821b4646f6b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:39:52.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Job loci-compute-kit-2023-1-ubuntu_focal not defined\n\nThe problem appears in the the \"opendev.org/openstack/loci\" project stanza:\n\n  project:\n      check:\n        jobs:\n          - loci-buildset-registry\n          # 2023.1 Ubuntu Focal\n          - loci-build-2023-1-ubuntu_focal\n          - loci-compute-kit-2023-1-ubuntu_focal\n          - loci-cinder-2023-1-ubuntu_focal\n          # 2023.1 Ubuntu Jammy\n  ...\n\n  in \"openstack/loci/.zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"359e31af2fc5ec5e65f24c6d35c3df6c0fc179de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-06 19:17:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bbb3c595344263f8a58ecdf09c1763263bbe33e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 22:14:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b8177e7ec14c9f891b81eb89531e26\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/309c7b142ee0423c9988b8d8199f9fcd : SUCCESS in 2h 50m 42s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/fd47806c0f664af4bfe9cb62fafc80c5 : SUCCESS in 1h 05m 52s\n- loci-compute-kit-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/c76a7593ec2f465b87aebc48612816ea : SUCCESS in 1h 09m 06s\n- loci-cinder-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/9adad7e5c7614073b4bb6cdcf25b44e9 : SUCCESS in 34m 16s\n- loci-build-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c7d8b675a44045b9bdc33e062a4f0746 : SUCCESS in 1h 02m 14s\n- loci-compute-kit-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1d7a49f03a4f45019ab93d57c29e813c : SUCCESS in 1h 06m 58s\n- loci-cinder-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bbed6d8ad72b4aafa5ca4a36b6a747d7 : SUCCESS in 46m 33s\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/94919ca8fc344ad6a0a8a766ac3c5a4d : SUCCESS in 1h 02m 12s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ead96765c37b48d5aa83e1908c79ce53 : SUCCESS in 1h 12m 09s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/46537fdde1314cd98aefe3d7717de57e : SUCCESS in 47m 20s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/11e1193c88b84afcb9216c36ef4c8b7f : SUCCESS in 1h 02m 11s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/973b787f832745a99f2509c0e8a843e0 : SUCCESS in 1h 09m 47s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d882b0459a514fde9d620171decfbc0f : SUCCESS in 43m 14s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/80c31e1a3fe44b4e8b16b91ae6050cd3 : SUCCESS in 1h 04m 57s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0ffa06e967ae432391758540b1517233 : SUCCESS in 1h 06m 48s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4e611bd613c141f79c287ec96b33dd26 : SUCCESS in 45m 13s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/36c11be8b9e5497eb1b3078023afe1c1 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1fa77bb317e3940e717a5ebbd0026260666b2620","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-11-06 22:19:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4719bca3f179ac89bc50decc78124d928fff645e","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-06 22:31:44.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"34517e9c3d657d4905d84916ddd3155e5b030797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-07 02:08:44.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7b7321c3fcf4a6a8f2253511237a937e7be7f6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 05:28:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645e72d7273542f2813fdeeb45c348ea\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2d752560cba540238c03c50502b8f08f : SUCCESS in 3h 01m 57s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/c87e19bc9a784602bb2cf5b9f2aef9a8 : SUCCESS in 1h 09m 28s\n- loci-compute-kit-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/afdafd84524b4579a077ab5dfd9a24b8 : SUCCESS in 1h 17m 35s\n- loci-cinder-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/5fbec19fc28742569ed0a1a4bf424c99 : SUCCESS in 49m 30s\n- loci-build-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8f16b7060dd2485ba059ebd79b891b23 : SUCCESS in 1h 11m 32s\n- loci-compute-kit-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/72cf3ad1250e481c940ab2184e40c271 : SUCCESS in 1h 20m 51s\n- loci-cinder-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/76e3221497814bba94cb95711b2d7ec8 : SUCCESS in 59m 38s\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3b435f90a5694189a7c7ffb07fb6749b : SUCCESS in 58m 50s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b097febe274148d4a2ae2a30ba5930ff : SUCCESS in 1h 14m 16s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2b1c48137b0c4e11bc1ff502cc96f0f4 : SUCCESS in 55m 40s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7c626ec8ca73437883aa294ddc779bae : SUCCESS in 1h 05m 38s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9e917c428658417fa7a9e8506b916293 : SUCCESS in 1h 16m 00s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5e6b05c513744d83a757cd0e1d4de188 : SUCCESS in 57m 45s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/310ee1e4624640298ff02281fd0a34e5 : SUCCESS in 1h 05m 45s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5a24d1234439482184339b949dab2a76 : SUCCESS in 1h 15m 17s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d27b02022ae9478fae2baebdf94bc314 : SUCCESS in 53m 58s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/97d88d23366748949e692287b7e7a65f : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":3},{"id":"5f79ec0953633a0504f9372a35b937e78240c907","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-11-07 13:59:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c26c910679c541f0852dc902fb30242d39f8d4eb","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-07 16:25:17.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"6a1b863a6c86ce903177314f94371b3e08e9f9b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-07 16:51:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a30c2f0da59847fee624c75c9d32a6c522941326","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-07 16:56:42.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"86b8dc827cad81cd4d75b2b9a0a98b2d8bd81e84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-07 17:07:37.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"879ba8568e600ebadb8eef9b9d6227cdb1785eb2","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-07 18:57:11.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"b08561d893d0593e475a143555322b53eb2db28e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 19:33:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0300b0c12ee42cab5bbc0f41c8c44d8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bf59660c777a43678d4ede6e13f8c76f : SUCCESS in 2h 23m 28s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/ba361c2e84ad4ddcb17af1fc699c20ab : SUCCESS in 59m 47s\n- loci-compute-kit-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/ec49b46f6e7743ddb44136f6e123f30c : SUCCESS in 1h 00m 47s\n- loci-cinder-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/ebbf43e78c134fa0910f67d9c2297bb0 : SUCCESS in 34m 23s\n- loci-build-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/73486c70928a4f5a8c31540c966529bd : SUCCESS in 52m 13s\n- loci-compute-kit-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/597136c4fe804e008a970c3b27d1dac7 : SUCCESS in 1h 03m 09s\n- loci-cinder-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a05b4ed399a94e9d8a250b0a96c77367 : SUCCESS in 40m 53s\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7352877d4484460a957c50a4a27ed0f2 : SUCCESS in 57m 48s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/98a237a9e88e49a89c21dd7d8d8d8791 : SUCCESS in 59m 01s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/82b4d27566c74e5e89f82ccab164726d : SUCCESS in 40m 56s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/312b8d62f9584d7fa09d6f400c06429e : SUCCESS in 55m 26s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7b0661240096485694a7ef812c9fc7c9 : SUCCESS in 1h 00m 25s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d1db4971f1814dcb88b0999767a511a5 : SUCCESS in 42m 34s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/136e3f24eee64ab7bc83dc251b4f0ace : SUCCESS in 56m 33s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0a76488861eb42a79380d23f55a16ed8 : SUCCESS in 1h 01m 14s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0f4b3907f7cd43349c5b428530a8413b : SUCCESS in 43m 00s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f233160b01874437809bcda3eae707ae : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":5},{"id":"d2123ec6eb86c154f7c79bc13d52f951c896b0a0","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-08 22:18:17.000000000","message":"Patch Set 5: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5bab7198feef4826d83b0916f62e382e04e2d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 00:27:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48760e4f90e6468786a199bdf8cbab36\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c97d48e32f8b472baa74bd05a984817c : SUCCESS in 2h 08m 56s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/393c47f4e505435a89099f22fa5e4a93 : SUCCESS in 58m 34s\n- loci-compute-kit-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/7eb97e9bd89749ab82f65e16f7bd7409 : SUCCESS in 58m 46s\n- loci-cinder-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/e0114b617f5a442b8247ad06e3cc1807 : SUCCESS in 32m 35s\n- loci-build-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eaca57f14f744f1eb1c3854c9c912d6c : SUCCESS in 1h 07m 44s\n- loci-compute-kit-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/453144bf029c4f8c8ed33fe63ddb16e2 : SUCCESS in 56m 25s\n- loci-cinder-2023-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e98636784c3e436eac824a9f0ce3f8c2 : SUCCESS in 37m 21s\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4004950b61f24271aee136a2c92886c4 : SUCCESS in 1h 05m 29s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/745e84b993c0490798ae1fb56c71e3b5 : SUCCESS in 56m 54s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0a503c9e26aa4460b8b0f327e08f164f : SUCCESS in 38m 46s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/19255ec010a74a83aa1f9248f26df5d3 : SUCCESS in 1h 04m 15s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/133ea67b6ece40749d86bc7829f59e34 : SUCCESS in 59m 01s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/55927cacc982479eb5b10e1f1adba02b : SUCCESS in 37m 41s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e530c7c01d1346718b5087adab815179 : SUCCESS in 55m 36s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/27200568c54c4f568be26634f8870f9a : SUCCESS in 1h 00m 04s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/105a1266c45c4530ac941633a7f27cb9 : SUCCESS in 38m 49s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/625c7a77135b4934945de18d2b56a6fa : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":5},{"id":"99d9a8325af2eddcc25edc59fdb504241602eb13","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-09 03:33:20.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"7622b5c1e0f97f46045cb17f46f6636c04c6a5bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-13 23:39:24.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a159d521976763c11369ca23c798bb88d831fdcb","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-13 23:39:37.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"070bf068c0479393c517928c18cdd14cf75cb00d","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-13 23:39:44.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"713a3ce8e629e6a331a10aac51abc5b1102e5f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 00:31:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17c52d62112b4579adc00237efbf9a9a\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/146af86e0bd540a597be97f27ecb7733 : SUCCESS in 51m 45s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c91e962a813e4fdc987be9959a6f1865 : FAILURE in 47m 25s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/958e50af9cf6425386a7fd0bc73407ba : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy","accounts_in_message":[],"_revision_number":6},{"id":"7b411c229ef185c9fe6a210c29cbdc1eb3540a25","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-14 00:41:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"754048c3047221dd5bbd032ed0885e01cd4a314e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 02:58:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a6061840ffd4a09b60f962aac25c53d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/8cd7b6617a8d4ec9841f47eabd68f20e : SUCCESS in 2h 16m 35s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/002445f76c2046b5af4d09bc2ab8c92c : SUCCESS in 1h 04m 59s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bb0024bb8ba64584b79f3b99bc87a96f : FAILURE in 58m 35s","accounts_in_message":[],"_revision_number":6},{"id":"ccd2338f418ddd2b3e6a354ea8fe54c4358b9788","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-14 03:02:46.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7562bc11cd2a57f6d3d87ae08c9a295b3a79f8fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 03:05:33.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/935032 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"4e16d51be402ba18e04ef8e290b518bd804e4ffa","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-14 03:09:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fabe3a978c60cf5749f4cd3fd00ee439edc3c6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 04:41:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32db2a6781ba4d30a3ba7111fe19a038\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a8c1f7b0ab714a1286e45cf1223c796d : SUCCESS in 59m 58s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d909d802f2544b3ca68a11ec179eb15f : SUCCESS in 44m 58s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/174626b17555410daa8a6cf7d9fc04ea : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":7},{"id":"2da71fbec10326299e684f341f913e6b96de8e2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-15 02:06:51.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e9e3cc85ede0a748bd0f1219204cd3d5ec45df9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 02:15:12.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/loci for 934263,8","accounts_in_message":[],"_revision_number":8},{"id":"c0ca8a7a1da22a87327eea4c0160a83abfd053f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-15 03:06:38.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"24e72217351d991047e41a02b2727f3d7e7735fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 05:09:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee0f4365a1b41a08fd7b8a0f4cdab66\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/adc8bcfdd2d149139f73fb16ea02f6cd : SUCCESS in 1h 51m 06s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0c1670f38d5d4c9fbd0894b672980f8a : SUCCESS in 39m 13s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b969b0feb87d4890b3b670227f9a14c8 : SUCCESS in 45m 44s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e1b6d29e0edc42f7a4d3a49f3e01790f : SUCCESS in 34m 42s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9884d05811004b89907cd259119c12d5 : SUCCESS in 42m 18s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/53dbcc47cdac470caf3d1f4b4b39b4f2 : SUCCESS in 48m 56s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8e5b54eae7974b0e850c7ba8a6a61a69 : SUCCESS in 34m 02s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/43b0eeab522648089590c1ef7ab20984 : SUCCESS in 42m 22s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e00491f8e2224576b2952c1a0cfa1d0e : SUCCESS in 49m 29s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/41b583e4a5a04f94a0da71f526b868ce : SUCCESS in 33m 41s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e967a95056d24b72b432074ff4eee8ab : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":9},{"id":"0e911114177cd9f7146804bb566b414fe20c13b5","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-15 15:41:49.000000000","message":"Patch Set 9:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"8edec63a5f2242463fe920fe5e78b3e8379e89eb","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-11-15 15:42:56.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a9dbc5da1e33b62ed7d3fd552e14c0ccb4b107a0","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2024-11-15 16:50:35.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1882b5c4879d06bfa2a2f6bf9e1b7d8ecb7e7311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:51:10.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4aefd7601cd3a8c257cfe7cd17a5e95d154e4562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:24:42.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5381163379246558f6293ff0c4c09c2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c4bedc37ed9148a4b07cc56b6161fdf2 : SUCCESS in 1h 27m 10s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4730833531f94c599bc9c6c74b6578d7 : SUCCESS in 1h 11m 19s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4c3a0cc9212b49a0965a8a1106cbe72e : SUCCESS in 1h 13m 24s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e9ff911c193c45d890d8fb9c51f8df9a : SUCCESS in 1h 18m 23s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/be76527c369a41aca883788e5fa2ad4c : SUCCESS in 1h 12m 00s","accounts_in_message":[],"_revision_number":9},{"id":"0358513cd797ca60dd7d92b9641994434f8930a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:24:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"4d1d3d3eb197c951ef223c71e402995a99b07321","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:34:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9541060724ef498795821306ae96ec4b\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a0cc929c7cec40788f43f2d13afcee0a : SUCCESS in 2m 32s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/361593fd2e5541f0b1b52f8e95e60247 : SUCCESS in 2m 32s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0c0ec8c87d454dcd89072e1c9f837af5 : SUCCESS in 2m 34s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/93397cc86cda462d8dcd69653b8894cb : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b1bd2bf27bb34382e94c47556622046425e23e20","revisions":{"64b4bf21573f499a5ec15b0e3fbff7d63b7d52a5":{"kind":"REWORK","_number":1,"created":"2024-11-06 18:39:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/1"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/64b4bf21573f499a5ec15b0e3fbff7d63b7d52a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/64b4bf21573f499a5ec15b0e3fbff7d63b7d52a5"}]},"branch":"refs/heads/master"},"1fc5500ed634ef52b9eaf12d8dd768622d4e18d5":{"kind":"REWORK","_number":2,"created":"2024-11-06 19:17:55.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/2"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 19:17:50.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1fc5500ed634ef52b9eaf12d8dd768622d4e18d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1fc5500ed634ef52b9eaf12d8dd768622d4e18d5"}]},"branch":"refs/heads/master"},"5c4cb82a4ec056ffa6f715709274e5997fda09d2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-07 02:08:44.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/3"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-07 02:03:29.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nDepends-On: Ie33ae8dd4f29746ff99cfc31962d924e12b62e1b\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c4cb82a4ec056ffa6f715709274e5997fda09d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c4cb82a4ec056ffa6f715709274e5997fda09d2"}]},"branch":"refs/heads/master"},"90453a8e2d8389b73b27b1422e6bc55875454ff5":{"kind":"REWORK","_number":4,"created":"2024-11-07 16:51:33.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/4"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-07 16:51:24.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nDepends-On: Ie33ae8dd4f29746ff99cfc31962d924e12b62e1b\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/90453a8e2d8389b73b27b1422e6bc55875454ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/90453a8e2d8389b73b27b1422e6bc55875454ff5"}]},"branch":"refs/heads/master"},"725d4bb6f91c40ca4d159d861bc4bbe04d9abed6":{"kind":"REWORK","_number":5,"created":"2024-11-07 17:07:37.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/5"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-07 17:07:30.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nDepends-On: Ie33ae8dd4f29746ff99cfc31962d924e12b62e1b\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/725d4bb6f91c40ca4d159d861bc4bbe04d9abed6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/725d4bb6f91c40ca4d159d861bc4bbe04d9abed6"}]},"branch":"refs/heads/master"},"eeefecccb5f3ee154a71fe2002d7f2bb4679e08e":{"kind":"REWORK","_number":6,"created":"2024-11-13 23:39:24.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/6"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-13 23:39:12.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eeefecccb5f3ee154a71fe2002d7f2bb4679e08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eeefecccb5f3ee154a71fe2002d7f2bb4679e08e"}]},"branch":"refs/heads/master"},"dd3d7b8aa54871ef95b6363596e7f6f5e368d9f4":{"kind":"REWORK","_number":7,"created":"2024-11-14 03:02:46.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/7"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-14 03:01:02.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build jobs for 2023.1 Ubuntu Jammy\nand 2023.2 Ubuntu Jammy\n\nDepends-On: I0eb69b36ab70095b732c34f88fbe6ef665627533\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dd3d7b8aa54871ef95b6363596e7f6f5e368d9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dd3d7b8aa54871ef95b6363596e7f6f5e368d9f4"}]},"branch":"refs/heads/master"},"2664a57609894607a38369195930a785ccea117b":{"kind":"REWORK","_number":8,"created":"2024-11-15 02:06:51.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/8"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-15 02:06:08.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build job for 2023.2 Ubuntu Jammy\n\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2664a57609894607a38369195930a785ccea117b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2664a57609894607a38369195930a785ccea117b"}]},"branch":"refs/heads/master"},"b1bd2bf27bb34382e94c47556622046425e23e20":{"kind":"REWORK","_number":9,"created":"2024-11-15 03:06:38.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/63/934263/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/63/934263/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/63/934263/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/63/934263/9"}}},"commit":{"parents":[{"commit":"3669c19c08d8e56e143490004e37785fe72b96d4","subject":"Drop 2023.1 release is unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3669c19c08d8e56e143490004e37785fe72b96d4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-06 18:38:12.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-11-15 03:05:58.000000000","tz":-360},"subject":"Test images before publishing them","message":"Test images before publishing them\n\nAlso build job for 2023.2 Ubuntu Jammy\n\nChange-Id: If391d17f14f054afca061036487574805c393634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b1bd2bf27bb34382e94c47556622046425e23e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b1bd2bf27bb34382e94c47556622046425e23e20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
