)]}'
{"id":"openstack%2Floci~934967","triplet_id":"openstack%2Floci~master~Ie50485859c0ebea90e26e4c98ea71d66b6323f2f","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-11-18 18:19:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie50485859c0ebea90e26e4c98ea71d66b6323f2f","subject":"Honor global-requirements when buiding wheels","status":"MERGED","created":"2024-11-13 13:44:31.000000000","updated":"2024-11-18 18:22:51.000000000","submitted":"2024-11-18 18:19:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934967","meta_rev_id":"c31b0439ead96ba44c8b3371620c071988711d51","_number":934967,"virtual_id_number":934967,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-18 18:19:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-11-18 14:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-11-18 17:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-11-18 17:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 16:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-18 14:59:09.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-11-18 17:25:08.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"e408729b4bc37534b523ef543fc4ddb7c342a782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 13:44:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc1c2f5dc77d8f5a50f4d0985d4c8e76cdbf1b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 14:34:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"075072b2247feb96806768bd885ce7cc9f79a399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 14:55:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"97ff5576693876483e2c40a28e900fbdd47a504f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:09:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d15be72ae2042dfba3afa6ecf331541\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/546acf815a0548b395b1434247d538ec : SUCCESS in 1h 08m 22s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/6da928b7ad834bf89e1c39757b5df436 : FAILURE in 48m 54s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a8ebbe7fa8d14c939f902946755216a3 : SUCCESS in 56m 21s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b075481551704c5683465fec919fc32d : SUCCESS in 1h 00m 08s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6e71401a058f45338c174fb16f3a988d : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":3},{"id":"39f962be396681ab8a9686768aa3a62cc1e4cd23","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 16:13:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d17fc002a55aae3377d4bcce0d9167f60a9de62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:55:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0299f4617ba847d5afb5f39b56c21225\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2bca230790b449ee8e773552c63cbeb0 : SUCCESS in 40m 54s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/53824eb82823470f9dfaa281d6dfd403 : FAILURE in 16m 05s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90c1d71b5c544b1e8c385a5edb7eab77 : SUCCESS in 37m 12s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/708e528f3f124449af42bf7ecac88e52 : SUCCESS in 36m 21s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b447639eb7254f7ba63849bc3925e5f8 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":3},{"id":"52ce5bd23c36f06868f6ae8e8b1e4c5e21c6f7e0","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 17:39:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"336dcbfcb9bc98b986910f82946674c062fc987d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 17:42:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"317acad9c0311be971449d02794f4c5d340c653d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 18:12:38.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6bb398961c6a3d255606a6c496a03735886ca048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 19:12:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e74bb6b5185443f85f5224d4a110a32\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/be7935698c974781a62833e3ccb61152 : SUCCESS in 58m 27s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/eb941ec12fa941e39646178d39788b4f : FAILURE in 3m 00s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c4baf9d6413b46bcafdbb9c90e40ba89 : SUCCESS in 52m 32s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6c03d01a675d4d3d9dde93842777b8e7 : SUCCESS in 51m 34s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/72ec1ea0fff2481f95a32b14ec39a8ba : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":5},{"id":"2124c369fce258d1b8dd3b856c65444929a9c3cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-14 07:07:01.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2a546986761a672be50bbdf0423d50e8dbee33e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 07:53:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8cb41abfd14e6ebfa0b66b08febbf4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0a2d20ca1c10425cb6a69d80eb3329f5 : SUCCESS in 45m 17s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5c4bf4b74e33414aaaeb99f6dc2bd8ec : SUCCESS in 41m 47s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ee01f874b824446cacfe74a7de509808 : SUCCESS in 41m 38s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/24b7216fc5384be5b30529df5a5d8c60 : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":6},{"id":"70eb5a6cf135ee79f0abf09453273435b5124781","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-18 14:59:09.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"614a1ee64340b32a7fac46d29b0bb8818be8f815","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2024-11-18 17:25:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"026bd4a6d01cad3e6d18aa52febd4ffc1617fa51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 17:25:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9c30cffde7afe4ad4ebd5ed8dc9a611d846f79a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 18:19:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61615c751b644475ae4b354e642469af\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ea3ebb142fbb4374b41475a69734def0 : SUCCESS in 53m 29s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b80cb7e150b44adca3b7f88587cafaee : SUCCESS in 50m 18s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c5a1dd6dd1a14b47b82bea658564c6b8 : SUCCESS in 50m 03s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6a723559cf1f4b92a4fb7d62bf6e6cec : SUCCESS in 49m 11s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/12197ffa5cc54ee38c1a6600a5bef51f : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":6},{"id":"90ba94a1277a9048484ab836664b92ea38852832","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 18:19:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c31b0439ead96ba44c8b3371620c071988711d51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 18:22:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5370175b46234b58b7a0d0ad6c00f06b\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/823fda9bf6f64077896e958cb9b88a34 : SUCCESS in 2m 35s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/035913ca618147ee8e4c0b2e29fe3287 : SUCCESS in 3m 09s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90e4ffec547e49eb9636338432c8c56a : SUCCESS in 2m 46s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f0fcf3ef07ac4f30bedbed4c01000b68 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8fa7edab4219c97fa20b1ce4d04a3734e740aeae","revisions":{"016f88ac5bbd3516598fbf61313d95d29025ea9e":{"kind":"REWORK","_number":1,"created":"2024-11-13 13:44:31.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/934967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/934967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/934967/1"}}},"commit":{"parents":[{"commit":"dc8797d448e1bc725fc2eafaa5d2e1a4ffd31a2e","subject":"Build horizon static during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dc8797d448e1bc725fc2eafaa5d2e1a4ffd31a2e"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:42.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:48.000000000","tz":0},"subject":"Honor global-requirements when buiding wheels","message":"Honor global-requirements when buiding wheels\n\nNot all dependencies are present in upper-constraints\none of examples in setuptools. Since global-requirements\nmay limit allowed versions add it here explicitly\nAdd global-requirements when building wheels explicitly.\n\nChange-Id: Ie50485859c0ebea90e26e4c98ea71d66b6323f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/016f88ac5bbd3516598fbf61313d95d29025ea9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/016f88ac5bbd3516598fbf61313d95d29025ea9e"}]},"branch":"refs/heads/master"},"2ba0e202c4f7bf4d0d12fa090169f36caf5153f6":{"kind":"REWORK","_number":2,"created":"2024-11-13 14:34:02.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/934967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/934967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/934967/2"}}},"commit":{"parents":[{"commit":"dc8797d448e1bc725fc2eafaa5d2e1a4ffd31a2e","subject":"Build horizon static during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dc8797d448e1bc725fc2eafaa5d2e1a4ffd31a2e"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:42.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 14:34:02.000000000","tz":0},"subject":"Honor global-requirements when buiding wheels","message":"Honor global-requirements when buiding wheels\n\nNot all dependencies are present in upper-constraints\none of examples in setuptools. Since global-requirements\nmay limit allowed versions add it here explicitly\nAdd global-requirements when building wheels explicitly.\n\nChange-Id: Ie50485859c0ebea90e26e4c98ea71d66b6323f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2ba0e202c4f7bf4d0d12fa090169f36caf5153f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2ba0e202c4f7bf4d0d12fa090169f36caf5153f6"}]},"branch":"refs/heads/master"},"06afd43deb5e6cb2ef4d54ee4883832136037832":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-13 14:55:26.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/934967/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/934967/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/934967/3"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:42.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 14:55:26.000000000","tz":0},"subject":"Honor global-requirements when buiding wheels","message":"Honor global-requirements when buiding wheels\n\nNot all dependencies are present in upper-constraints\none of examples in setuptools. Since global-requirements\nmay limit allowed versions add it here explicitly\nAdd global-requirements when building wheels explicitly.\n\nChange-Id: Ie50485859c0ebea90e26e4c98ea71d66b6323f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06afd43deb5e6cb2ef4d54ee4883832136037832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06afd43deb5e6cb2ef4d54ee4883832136037832"}]},"branch":"refs/heads/master","description":"Rebase"},"a52abf6b8d83755eec7e2914dfcaf2148484e464":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-13 17:42:16.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/934967/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/934967/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/934967/4"}}},"commit":{"parents":[{"commit":"2bb270b0ee54fa0349ba8603b73ab7eab0e5bd17","subject":"Drop cyborg image build for 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2bb270b0ee54fa0349ba8603b73ab7eab0e5bd17"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:42.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:42:16.000000000","tz":0},"subject":"Honor global-requirements when buiding wheels","message":"Honor global-requirements when buiding wheels\n\nNot all dependencies are present in upper-constraints\none of examples in setuptools. Since global-requirements\nmay limit allowed versions add it here explicitly\nAdd global-requirements when building wheels explicitly.\n\nChange-Id: Ie50485859c0ebea90e26e4c98ea71d66b6323f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a52abf6b8d83755eec7e2914dfcaf2148484e464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a52abf6b8d83755eec7e2914dfcaf2148484e464"}]},"branch":"refs/heads/master","description":"Rebase"},"7629281f1a4751d29b232e20837558c910f45bc7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-13 18:12:38.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/934967/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/934967/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/934967/5"}}},"commit":{"parents":[{"commit":"497c8813d46b11e828c10bec6fd23fd012bd264a","subject":"Drop 2023.1 release is unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/497c8813d46b11e828c10bec6fd23fd012bd264a"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:42.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 18:12:38.000000000","tz":0},"subject":"Honor global-requirements when buiding wheels","message":"Honor global-requirements when buiding wheels\n\nNot all dependencies are present in upper-constraints\none of examples in setuptools. Since global-requirements\nmay limit allowed versions add it here explicitly\nAdd global-requirements when building wheels explicitly.\n\nChange-Id: Ie50485859c0ebea90e26e4c98ea71d66b6323f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7629281f1a4751d29b232e20837558c910f45bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7629281f1a4751d29b232e20837558c910f45bc7"}]},"branch":"refs/heads/master","description":"Rebase"},"8fa7edab4219c97fa20b1ce4d04a3734e740aeae":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-14 07:07:01.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/67/934967/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/67/934967/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/67/934967/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/67/934967/6"}}},"commit":{"parents":[{"commit":"3669c19c08d8e56e143490004e37785fe72b96d4","subject":"Drop 2023.1 release is unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3669c19c08d8e56e143490004e37785fe72b96d4"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 13:38:42.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-14 07:07:01.000000000","tz":0},"subject":"Honor global-requirements when buiding wheels","message":"Honor global-requirements when buiding wheels\n\nNot all dependencies are present in upper-constraints\none of examples in setuptools. Since global-requirements\nmay limit allowed versions add it here explicitly\nAdd global-requirements when building wheels explicitly.\n\nChange-Id: Ie50485859c0ebea90e26e4c98ea71d66b6323f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8fa7edab4219c97fa20b1ce4d04a3734e740aeae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8fa7edab4219c97fa20b1ce4d04a3734e740aeae"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
