)]}'
{"id":"openstack%2Floci~934993","triplet_id":"openstack%2Floci~master~I182a944c6634a381767b64e40e5aaad3be7aa748","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-11-15 02:08:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I182a944c6634a381767b64e40e5aaad3be7aa748","subject":"Drop 2023.1 release is unmaintained","status":"MERGED","created":"2024-11-13 17:41:54.000000000","updated":"2024-11-15 02:22:49.000000000","submitted":"2024-11-15 02:08:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934993","meta_rev_id":"534d6685164efb16a5a0d1cbcf509d2f68a361cb","_number":934993,"virtual_id_number":934993,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 02:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-11-15 00:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-11-15 00:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-11-15 00:40:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 18:46:18.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"CC"},{"updated":"2024-11-13 19:59:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 00:39:51.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-11-15 00:40:53.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"607ea892f80ea82abacd47b8b598f2940ac57c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 17:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfefee3db7a9b5c0f72538b1dfe3e375f441dbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 17:42:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d84c449eb4a88e1fe784d65741fd2284b6f55a2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 18:12:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9316d18a1de3ea8b532b59d90db3c4776b56f658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-13 18:12:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c3950cac91d097894087c0a96835b4107f5c9dc3","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-13 18:46:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d91222dce8903b1ada0707175d290e8e0be05484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 19:59:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7926fde20be345eeb039d159491cc625\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1913e24994194156816fbf849ecf7c62 : SUCCESS in 1h 45m 26s (non-voting)\n- loci-build-2023-1-ubuntu_focal https://zuul.opendev.org/t/openstack/build/eeb84897d326478eb3731456a10b546b : FAILURE in 2m 39s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b483ec3066694f89b4d72e3a42d72a98 : SUCCESS in 49m 34s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/843e0ae01adb4cf88d6fd53ad0e2eef4 : SUCCESS in 1h 06m 21s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1e0db75071f1486a867fa9936dbd6cfd : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b1afda612c9354f7f044f1bb99dbd25a7a8f5e22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-14 07:05:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8e88c50897e861396a9f0079ab4e3dbf88d6f3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-14 07:06:54.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3f9b3ddb5854af01b7657380297f9e72c06bf182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 08:17:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/526ed7ca907843bdbdb801cb7ad154c6\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f7b75e8baf164ca89ffd6ac8d7ae63e7 : SUCCESS in 1h 02m 43s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/064b68b76a4d47aab931545a51a55de4 : SUCCESS in 59m 24s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/79fcf30a4a4743fdb08f1a4f0134fffe : SUCCESS in 55m 53s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/76b529dc99c14d50a3e131fcf6cb2960 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":6},{"id":"7e40bcb2149f406717bdeb2bd54c1f145292d1c8","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-15 00:39:51.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"849df2c7208acab33638b957d6ab84b401b05d8c","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-11-15 00:40:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0ba71493288a48581754dedf93f28848d81ffa59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 00:41:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"08357ab0cdabc60f16a3ad05221d4e925dc5596b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 02:08:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c103072ec851469cbb145c61a2e8a322\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/46e7a92112f24588aeac36134ca2cb4e : SUCCESS in 1h 21m 38s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/258d3b2cf9e6428daba01387758ffc0a : SUCCESS in 1h 15m 11s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d9062469d22049678b85c95522e9960b : SUCCESS in 1h 12m 20s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/726a6cded8bc479f9a4d8c5c53f74d9c : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":6},{"id":"0b3d49d9e4d06cdb5c77140c10bd5879ada2dfc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 02:08:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"534d6685164efb16a5a0d1cbcf509d2f68a361cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 02:22:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6c98ca5a9940f89c60ee8f9c4c79d8\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8197b0f0168c489bb6bee5d0863fcdc4 : SUCCESS in 4m 23s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c7b5d94033a940fabde927f3cde2379b : SUCCESS in 3m 48s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a0b959cda4a844db8afea444b2f29593 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3669c19c08d8e56e143490004e37785fe72b96d4","revisions":{"ec2b3e02e5bc33eb24b3ac1d0268bd05c673c328":{"kind":"REWORK","_number":1,"created":"2024-11-13 17:41:54.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/93/934993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/934993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/934993/1"}}},"commit":{"parents":[{"commit":"016f88ac5bbd3516598fbf61313d95d29025ea9e","subject":"Honor global-requirements when buiding wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/016f88ac5bbd3516598fbf61313d95d29025ea9e"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"subject":"Drop cyborg image build for 2023.1","message":"Drop cyborg image build for 2023.1\n\nThe release is unmaintained and don\u0027t have stable branch\n\nChange-Id: I182a944c6634a381767b64e40e5aaad3be7aa748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec2b3e02e5bc33eb24b3ac1d0268bd05c673c328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec2b3e02e5bc33eb24b3ac1d0268bd05c673c328"}]},"branch":"refs/heads/master"},"2bb270b0ee54fa0349ba8603b73ab7eab0e5bd17":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-13 17:42:05.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/93/934993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/934993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/934993/2"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:42:05.000000000","tz":0},"subject":"Drop cyborg image build for 2023.1","message":"Drop cyborg image build for 2023.1\n\nThe release is unmaintained and don\u0027t have stable branch\n\nChange-Id: I182a944c6634a381767b64e40e5aaad3be7aa748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2bb270b0ee54fa0349ba8603b73ab7eab0e5bd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2bb270b0ee54fa0349ba8603b73ab7eab0e5bd17"}]},"branch":"refs/heads/master","description":"Rebase"},"8fed165f3931d46b9b3d3e1834cdbc316ca5c421":{"kind":"REWORK","_number":3,"created":"2024-11-13 18:12:11.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/93/934993/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/934993/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/934993/3"}}},"commit":{"parents":[{"commit":"016f88ac5bbd3516598fbf61313d95d29025ea9e","subject":"Honor global-requirements when buiding wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/016f88ac5bbd3516598fbf61313d95d29025ea9e"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 18:11:45.000000000","tz":0},"subject":"Drop 2023.1 release is unmaintained","message":"Drop 2023.1 release is unmaintained\n\nThe release is unmaintained and don\u0027t have stable branch\n\nChange-Id: I182a944c6634a381767b64e40e5aaad3be7aa748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8fed165f3931d46b9b3d3e1834cdbc316ca5c421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8fed165f3931d46b9b3d3e1834cdbc316ca5c421"}]},"branch":"refs/heads/master"},"497c8813d46b11e828c10bec6fd23fd012bd264a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-13 18:12:26.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/93/934993/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/934993/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/934993/4"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 18:12:26.000000000","tz":0},"subject":"Drop 2023.1 release is unmaintained","message":"Drop 2023.1 release is unmaintained\n\nThe release is unmaintained and don\u0027t have stable branch\n\nChange-Id: I182a944c6634a381767b64e40e5aaad3be7aa748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/497c8813d46b11e828c10bec6fd23fd012bd264a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/497c8813d46b11e828c10bec6fd23fd012bd264a"}]},"branch":"refs/heads/master","description":"Rebase"},"26b34c86cc0bd97ae3230d6135476a35a77beb2f":{"kind":"REWORK","_number":5,"created":"2024-11-14 07:05:34.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/93/934993/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/934993/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/934993/5"}}},"commit":{"parents":[{"commit":"016f88ac5bbd3516598fbf61313d95d29025ea9e","subject":"Honor global-requirements when buiding wheels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/016f88ac5bbd3516598fbf61313d95d29025ea9e"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-14 07:05:14.000000000","tz":0},"subject":"Drop 2023.1 release is unmaintained","message":"Drop 2023.1 release is unmaintained\n\nThe release is unmaintained and don\u0027t have stable branch\n\nChange-Id: I182a944c6634a381767b64e40e5aaad3be7aa748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26b34c86cc0bd97ae3230d6135476a35a77beb2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26b34c86cc0bd97ae3230d6135476a35a77beb2f"}]},"branch":"refs/heads/master"},"3669c19c08d8e56e143490004e37785fe72b96d4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-14 07:06:54.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/93/934993/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/934993/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/934993/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/934993/6"}}},"commit":{"parents":[{"commit":"8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579","subject":"Set default locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8488ce67a1f3d7c2bbc58ec7ae8d7fc2284ae579"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-13 17:41:45.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-14 07:06:54.000000000","tz":0},"subject":"Drop 2023.1 release is unmaintained","message":"Drop 2023.1 release is unmaintained\n\nThe release is unmaintained and don\u0027t have stable branch\n\nChange-Id: I182a944c6634a381767b64e40e5aaad3be7aa748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3669c19c08d8e56e143490004e37785fe72b96d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3669c19c08d8e56e143490004e37785fe72b96d4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
