)]}'
{"id":"openstack%2Floci~935787","triplet_id":"openstack%2Floci~master~I8b04e06e5151010b2061440dcd609999c6c829a9","project":"openstack/loci","branch":"master","topic":"ustream","hashtags":[],"change_id":"I8b04e06e5151010b2061440dcd609999c6c829a9","subject":"Add builder python script","status":"ABANDONED","created":"2024-11-20 13:39:42.000000000","updated":"2025-01-10 00:51:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eef208041cd388fda33569a6fc5c72886b45c726","_number":935787,"virtual_id_number":935787,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-20 15:22:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-11-21 19:22:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 15:22:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8723cc1c0e2012a0ecd72d723f8cdce6e9c87ffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-20 13:39:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a5dd19d514ae5fcb853443c14f9b44c9ea3bc8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 15:22:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa035ff55c4947a5af3f2dc41a4f22ed\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/272b4b959c0b41ceb17971462ecfc913 : SUCCESS in 1h 31m 06s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/81ffd2a849744dc3833d5420f51f4a75 : SUCCESS in 35m 30s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b067a983c2d045f89848922b60bf02e8 : SUCCESS in 51m 38s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/387c4707c923476b9ecb7af334dd09e9 : SUCCESS in 35m 37s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8f4bc1c7dd3b45beaa75a31d58f52b1b : SUCCESS in 44m 50s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7fb4b25911544ba384640183eef3f99d : SUCCESS in 36m 58s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/02cf13a4aa60487987c1e6a98f2e489e : SUCCESS in 46m 02s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b3cc991bf8c944ac801b6e6cbc958902 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b96538941b41fa1acfb65ef3e32a59d9fcc68528","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-21 19:22:50.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"eef208041cd388fda33569a6fc5c72886b45c726","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2025-01-10 00:51:20.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 49 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"379beea7186ae01cb4ea86c64f1465bcc6dc61e0","revisions":{"379beea7186ae01cb4ea86c64f1465bcc6dc61e0":{"kind":"REWORK","_number":1,"created":"2024-11-20 13:39:42.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/87/935787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/87/935787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/935787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/935787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/935787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/87/935787/1"}}},"commit":{"parents":[{"commit":"0d8f1366ce2181a44b48ac81a9957daf883908cc","subject":"Allow to build libraries from sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0d8f1366ce2181a44b48ac81a9957daf883908cc"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-20 13:37:16.000000000","tz":0},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-20 13:38:31.000000000","tz":0},"subject":"Add builder python script","message":"Add builder python script\n\nThe patch adds a loci.py builder script that allows to define\nbuild configuration in machine readable format - yaml file\n\n  python3 loci.py  --config build_profiles/caracal-minimal.yaml \\\n     build --image-path registry.it.just.works:32500/openstack \\\n     --image-tag m4 --push\n\nAnd build all images with one command.\n\nChange-Id: I8b04e06e5151010b2061440dcd609999c6c829a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/379beea7186ae01cb4ea86c64f1465bcc6dc61e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/379beea7186ae01cb4ea86c64f1465bcc6dc61e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
