)]}'
{"id":"openstack%2Floci~935884","triplet_id":"openstack%2Floci~master~Ife7b0000a66936f3581f968d8825294b71ac3c49","project":"openstack/loci","branch":"master","topic":"ustream","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-11-27 07:58:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14525\u003e replied on the change","reason_account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}}},"hashtags":[],"change_id":"Ife7b0000a66936f3581f968d8825294b71ac3c49","subject":"Add tss user/group for nova image","status":"MERGED","created":"2024-11-21 12:33:56.000000000","updated":"2024-11-27 12:26:33.000000000","submitted":"2024-11-27 12:23:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935884-ustream","meta_rev_id":"d8ab12921406cdcefcd85d2958498900cf6e2d6f","_number":935884,"virtual_id_number":935884,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-27 12:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-11-26 21:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-11-26 23:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-11-26 23:48:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 14:40:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-26 21:55:43.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-11-26 23:48:52.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"189aee3a79f377c23bfe2a615c17a827ef44e24e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-21 12:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64cbc52f322895d5e234c2c95e7a78f87f1e6796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 14:40:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8c655f7eaa41b7b469bb8f7806ef62\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a8cd3bccbefa476a9458e46b368a10cf : SUCCESS in 1h 50m 13s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/96aa244c23144ad1b6617081ea926235 : SUCCESS in 47m 02s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8393c8373f454a75b8e5b1a17300dbc1 : SUCCESS in 48m 21s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5badbc29d14240cdbd8973015788274a : SUCCESS in 46m 43s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6c5549c765e1469aae9b65052cc95dc2 : SUCCESS in 46m 06s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b82af9cc22514777850970b470c490ad : SUCCESS in 46m 24s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fd69f00cb9dd4903b077ec21ae4c579d : SUCCESS in 47m 57s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b97d8e36e9564655bf6734cb1e7c660b : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":1},{"id":"24d723f8cb0daeb743b6aee1cde5651ff76abc35","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-25 15:45:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3274aaf957dfcba4323735ad3d4c68eca216a55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-26 07:50:56.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eeea4cc1fa292e01bc7b27fbae40143635a39e74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-26 07:51:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"532d3d7ed5aabd4f9c2a98b85c8472f3b5c1e077","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-26 07:51:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17a58241486df1078a851d71e7578eff2e71711b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 09:21:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb7165b4c4ee4aeca0852fbfafb293f7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1d569e8b31814ceb9f14592b830d782e : SUCCESS in 1h 29m 45s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d6788ca10bdb4f82be3c5d987b91c085 : SUCCESS in 38m 30s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ec59ab564abe447cb601f7141d4617ec : SUCCESS in 44m 45s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/461b1e4e760845f8a9992abdef682a48 : SUCCESS in 39m 08s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1992e279cff043cb8bf0d0bc49769ce1 : SUCCESS in 45m 53s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/febb26cf778b4f29a33df7a7ac0e7209 : SUCCESS in 39m 26s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b19f77b2ec33483b883dc17496960bfa : SUCCESS in 46m 58s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2b1771d931124313a817e28421e844c3 : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f820ba43b9643f95f690573fbd686b3eea480187","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-11-26 21:55:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"65318c684ba6fcb519ca5a94c49b5fe6eddb0366","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-11-26 23:48:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"124f29a761d4b63352a4ada0f9a0d9c132b5c11d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 23:49:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"09c28fedc7fcae1147f12447db0a5837e49495da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 00:11:47.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8c2b6d6156a45a7bf4ce719857df924\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/63729fcf20c44790a0019fa527aeef17 : SUCCESS in 22m 15s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8179bf3d397c43a19ef8052422ff70e0 : POST_FAILURE in 14m 17s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f2273df1d7a84096a80d53a6af75ab03 : POST_FAILURE in 16m 59s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c14a51c3a888492d8ca9c7a6b633fe41 : POST_FAILURE in 17m 55s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/48e11a889de34264bf391b2759afab1d : POST_FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":3},{"id":"b14d3ddbac79c50c121b8371306d9b51311ac332","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-27 07:58:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"382907eef0bae0fb7c6e115d5938aaa9ee47c485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 10:40:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380e96c98e42482aa04b87f604388848\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0d5fa6c4f2834074aaac8e33974c986c : SUCCESS in 2h 41m 47s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b38408268b6a499ab8e8170b677dd620 : SUCCESS in 1h 08m 08s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/96a06d277b014511b4641e1509414d3d : SUCCESS in 1h 12m 50s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89c6cd6584204c3c9fd3ca7238d96a3d : SUCCESS in 1h 15m 38s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8db941c4f2e64148997d406b7f765e8c : SUCCESS in 1h 17m 06s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ec538cbfa70b49808897dfe509227893 : SUCCESS in 1h 10m 58s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bf6d524c73ab43839e0e6a445a663c26 : SUCCESS in 1h 13m 53s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f03f5dee094c47d2b69a7977c4e6c8c5 : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"4a4013bf2332978bc31e40a5e1c2c1e800b00b37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 10:41:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"542276fb5cb3ed2b5d7ce00a4767e5383b99169f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:23:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30dadee99f1144e8bf074e5b111df309\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/cd5a1c8fb0834625a9c2458909d42a8f : SUCCESS in 1h 36m 55s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/199a819f492c49aa9bfe5eaa39877711 : SUCCESS in 1h 16m 40s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/69c7b7af9a7947e1b020a515da52adf2 : SUCCESS in 1h 09m 31s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0bb1aa732b8f4021a6d5e02fefe0661c : SUCCESS in 51m 39s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eabbd3959ba7412ab5ef37e50422120f : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":3},{"id":"399baa0982451a7567a9504f7395586bbc0e6c4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:23:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d8ab12921406cdcefcd85d2958498900cf6e2d6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:26:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db27461d869f488ba5a2554a1608d38e\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d62383e99785459f97b46b48ec716f23 : SUCCESS in 2m 26s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7c082ded0cdb48caaadc0da967ce78fe : SUCCESS in 2m 41s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/75ec6cdf9e0449c9944c0ed7ebea0b96 : SUCCESS in 2m 35s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fc3f20048f3b4761aa1039151e4503db : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"97263c21d144032cdcf12432990cffa308829956","revisions":{"c9b14bc7f37145a60f25d69f063efa9af93c0062":{"kind":"REWORK","_number":1,"created":"2024-11-21 12:33:56.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/84/935884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/84/935884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/84/935884/1"}}},"commit":{"parents":[{"commit":"808ff6759439b98c400ea7ea09f6ceabea5d78b3","subject":"Revert \"Fix broken pydep for older releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/808ff6759439b98c400ea7ea09f6ceabea5d78b3"}]}],"author":{"name":"dbiletskiy","email":"dbiletskiy@mirantis.com","date":"2023-12-08 13:01:27.000000000","tz":60},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-21 09:04:31.000000000","tz":0},"subject":"Add tss user/group for nova image","message":"Add tss user/group for nova image\n\nWhen using emulated TPM, needs to set the user/group\nthat swtpm binary runs as.\n\nRelated-Prod: PRODX-37352\nChange-Id: Ife7b0000a66936f3581f968d8825294b71ac3c49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9b14bc7f37145a60f25d69f063efa9af93c0062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9b14bc7f37145a60f25d69f063efa9af93c0062"}]},"branch":"refs/heads/master"},"9ffb63e907fd46b87783650704928bd7002aa5c9":{"kind":"REWORK","_number":2,"created":"2024-11-26 07:50:56.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/84/935884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/84/935884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/84/935884/2"}}},"commit":{"parents":[{"commit":"808ff6759439b98c400ea7ea09f6ceabea5d78b3","subject":"Revert \"Fix broken pydep for older releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/808ff6759439b98c400ea7ea09f6ceabea5d78b3"}]}],"author":{"name":"dbiletskiy","email":"dbiletskiy@mirantis.com","date":"2023-12-08 13:01:27.000000000","tz":60},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-26 07:50:53.000000000","tz":0},"subject":"Add tss user/group for nova image","message":"Add tss user/group for nova image\n\nWhen using emulated TPM, needs to set the user/group\nthat swtpm binary runs as.\n\nRelated-Prod: PRODX-37352\nChange-Id: Ife7b0000a66936f3581f968d8825294b71ac3c49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9ffb63e907fd46b87783650704928bd7002aa5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9ffb63e907fd46b87783650704928bd7002aa5c9"}]},"branch":"refs/heads/master"},"97263c21d144032cdcf12432990cffa308829956":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-26 07:51:03.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/84/935884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/84/935884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/935884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/84/935884/3"}}},"commit":{"parents":[{"commit":"1b578c3d515bdf8cadacbafa6136685f01e7fee4","subject":"Merge \"Honor global-requirements when buiding wheels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1b578c3d515bdf8cadacbafa6136685f01e7fee4"}]}],"author":{"name":"dbiletskiy","email":"dbiletskiy@mirantis.com","date":"2023-12-08 13:01:27.000000000","tz":60},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-26 07:51:03.000000000","tz":0},"subject":"Add tss user/group for nova image","message":"Add tss user/group for nova image\n\nWhen using emulated TPM, needs to set the user/group\nthat swtpm binary runs as.\n\nRelated-Prod: PRODX-37352\nChange-Id: Ife7b0000a66936f3581f968d8825294b71ac3c49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97263c21d144032cdcf12432990cffa308829956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97263c21d144032cdcf12432990cffa308829956"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
